)]}'
{"id":"openstack%2Fdesignate~657177","triplet_id":"openstack%2Fdesignate~master~Ia3d2951f4e90817782bf959a6b2402eac26dbc7f","project":"openstack/designate","branch":"master","topic":"ksv2_n_region","hashtags":[],"change_id":"Ia3d2951f4e90817782bf959a6b2402eac26dbc7f","subject":"Remove KS v2 support and add region support","status":"MERGED","created":"2019-05-05 01:04:56.000000000","updated":"2019-06-09 09:30:21.000000000","submitted":"2019-06-09 09:30:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657177-1560072621868-b5d80b2d","meta_rev_id":"ca49fad245b54863868111489e067a860b2041ba","_number":657177,"virtual_id_number":657177,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-06-09 09:30:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-05-27 09:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-09 08:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-27 09:52:51.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-09 09:30:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87014fefbeaa3a5c743e5e289e45d82caf879049","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-05 01:04:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a396a6582c2a731a49ac174d4e6354425b63eb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 02:12:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/77/657177/1/check/designate-bind9/d2aaed8/ : SUCCESS in 59m 02s\n- designate-pdns4 http://logs.openstack.org/77/657177/1/check/designate-pdns4/e725908/ : SUCCESS in 37m 27s\n- designate-bind9-manager-model http://logs.openstack.org/77/657177/1/check/designate-bind9-manager-model/23837cf/ : SUCCESS in 38m 25s\n- designate-bind9-py27 http://logs.openstack.org/77/657177/1/check/designate-bind9-py27/168585a/ : SUCCESS in 42m 25s\n- designate-pdns4-py27 http://logs.openstack.org/77/657177/1/check/designate-pdns4-py27/1766db5/ : SUCCESS in 37m 13s\n- designate-pdns4-postgres http://logs.openstack.org/77/657177/1/check/designate-pdns4-postgres/e1270d1/ : SUCCESS in 38m 16s\n- designate-grenade-pdns4 http://logs.openstack.org/77/657177/1/check/designate-grenade-pdns4/79a4498/ : SUCCESS in 57m 14s\n- openstack-tox-cover http://logs.openstack.org/77/657177/1/check/openstack-tox-cover/ec0d49a/cover/ : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/657177/1/check/openstack-tox-lower-constraints/ffb4e61/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/77/657177/1/check/openstack-tox-pep8/ae662e7/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/77/657177/1/check/openstack-tox-py27/631284a/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/77/657177/1/check/openstack-tox-py36/4f7d393/ : SUCCESS in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/77/657177/1/check/openstack-tox-py37/78a8423/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/77/657177/1/check/openstack-tox-docs/826999a/html/ : SUCCESS in 4m 48s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/77/657177/1/check/neutron-tempest-plugin-designate-scenario/8f7080d/ : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c5e8a6ec5c2a5eabec3b2042242ef08cdb4bdcde","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-05 02:37:20.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"9e40edf4c1f90f39ae5bd8e3044b9d25cb9b95f7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-05 02:41:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"471ea6d468f979891fe5310d6bd6231ac6fcb05b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-05 03:15:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a8d22195d6c63f4bb8bbcbdf5fb8aefccbd25e81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 04:28:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/77/657177/3/check/designate-bind9/7605d40/ : SUCCESS in 41m 05s\n- designate-pdns4 http://logs.openstack.org/77/657177/3/check/designate-pdns4/a44ebeb/ : SUCCESS in 40m 36s\n- designate-bind9-manager-model http://logs.openstack.org/77/657177/3/check/designate-bind9-manager-model/4eb5972/ : SUCCESS in 49m 55s\n- designate-bind9-py27 http://logs.openstack.org/77/657177/3/check/designate-bind9-py27/ddfaf74/ : SUCCESS in 45m 24s\n- designate-pdns4-py27 http://logs.openstack.org/77/657177/3/check/designate-pdns4-py27/d2be716/ : SUCCESS in 41m 06s\n- designate-pdns4-postgres http://logs.openstack.org/77/657177/3/check/designate-pdns4-postgres/6f574e6/ : SUCCESS in 37m 01s\n- designate-grenade-pdns4 http://logs.openstack.org/77/657177/3/check/designate-grenade-pdns4/d8efc1f/ : SUCCESS in 54m 52s\n- openstack-tox-cover http://logs.openstack.org/77/657177/3/check/openstack-tox-cover/b8c55fa/cover/ : SUCCESS in 7m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/657177/3/check/openstack-tox-lower-constraints/05ae36d/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/77/657177/3/check/openstack-tox-pep8/9aa7cee/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/77/657177/3/check/openstack-tox-py27/60f3cfa/ : SUCCESS in 6m 52s\n- openstack-tox-py36 http://logs.openstack.org/77/657177/3/check/openstack-tox-py36/43e6db3/ : SUCCESS in 6m 08s\n- openstack-tox-py37 http://logs.openstack.org/77/657177/3/check/openstack-tox-py37/306fd37/ : SUCCESS in 9m 16s\n- openstack-tox-docs http://logs.openstack.org/77/657177/3/check/openstack-tox-docs/d37267f/html/ : SUCCESS in 5m 28s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/77/657177/3/check/neutron-tempest-plugin-designate-scenario/7595c5b/ : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":3},{"id":"305592c528eeb47b0f888b98110beb5deeec7244","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 00:09:49.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"47c46d0763ea88d264b6f9ab122e51b4f07da74e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 01:18:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/77/657177/4/check/designate-bind9/691ca76/ : SUCCESS in 37m 02s\n- designate-pdns4 http://logs.openstack.org/77/657177/4/check/designate-pdns4/2430d8f/ : SUCCESS in 37m 30s\n- designate-bind9-manager-model http://logs.openstack.org/77/657177/4/check/designate-bind9-manager-model/962fe7f/ : SUCCESS in 35m 58s\n- designate-bind9-py27 http://logs.openstack.org/77/657177/4/check/designate-bind9-py27/44fbd75/ : SUCCESS in 38m 29s\n- designate-pdns4-py27 http://logs.openstack.org/77/657177/4/check/designate-pdns4-py27/67325f7/ : SUCCESS in 37m 50s\n- designate-pdns4-postgres http://logs.openstack.org/77/657177/4/check/designate-pdns4-postgres/60ab616/ : SUCCESS in 36m 14s\n- designate-grenade-pdns4 http://logs.openstack.org/77/657177/4/check/designate-grenade-pdns4/53bb1be/ : SUCCESS in 1h 03m 03s\n- openstack-tox-cover http://logs.openstack.org/77/657177/4/check/openstack-tox-cover/52fbd1b/cover/ : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/657177/4/check/openstack-tox-lower-constraints/006e488/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/77/657177/4/check/openstack-tox-pep8/f32a751/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/77/657177/4/check/openstack-tox-py27/d9bbfb6/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/77/657177/4/check/openstack-tox-py36/5e21d1a/ : SUCCESS in 5m 09s\n- openstack-tox-py37 http://logs.openstack.org/77/657177/4/check/openstack-tox-py37/f6ea3e0/ : SUCCESS in 8m 50s\n- openstack-tox-docs http://logs.openstack.org/77/657177/4/check/openstack-tox-docs/140b4c7/html/ : SUCCESS in 4m 56s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/77/657177/4/check/neutron-tempest-plugin-designate-scenario/42063f4/ : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":4},{"id":"033f0a3198058d444142f94490160f64cf9b92f0","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-20 10:44:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"afac01c7442e1ff5e8d391cb768a3c898994268d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-26 04:05:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b8e2172ae178c422ffe86dfcff2782d3e44bb583","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-26 04:05:38.000000000","message":"Patch Set 5:\n\nRebased to fix merge conflict","accounts_in_message":[],"_revision_number":5},{"id":"f02fd062d0a322625c9dddf80501baf58386b734","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-26 05:06:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/77/657177/5/check/designate-bind9/95f87db/ : SUCCESS in 42m 39s\n- designate-pdns4 http://logs.openstack.org/77/657177/5/check/designate-pdns4/6d49060/ : SUCCESS in 34m 13s\n- designate-bind9-manager-model http://logs.openstack.org/77/657177/5/check/designate-bind9-manager-model/71bd65a/ : SUCCESS in 32m 16s\n- designate-bind9-py27 http://logs.openstack.org/77/657177/5/check/designate-bind9-py27/be8a69f/ : SUCCESS in 31m 50s\n- designate-pdns4-py27 http://logs.openstack.org/77/657177/5/check/designate-pdns4-py27/23b2f28/ : SUCCESS in 29m 15s\n- designate-pdns4-postgres http://logs.openstack.org/77/657177/5/check/designate-pdns4-postgres/3233aba/ : SUCCESS in 32m 57s\n- designate-grenade-pdns4 http://logs.openstack.org/77/657177/5/check/designate-grenade-pdns4/ab0f2bc/ : SUCCESS in 50m 22s\n- openstack-tox-cover http://logs.openstack.org/77/657177/5/check/openstack-tox-cover/2ee881b/cover/ : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/657177/5/check/openstack-tox-lower-constraints/e65e33d/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/77/657177/5/check/openstack-tox-pep8/a9025c5/ : SUCCESS in 5m 07s\n- openstack-tox-py27 http://logs.openstack.org/77/657177/5/check/openstack-tox-py27/3155ef1/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/77/657177/5/check/openstack-tox-py36/53ea9d1/ : SUCCESS in 5m 00s\n- openstack-tox-py37 http://logs.openstack.org/77/657177/5/check/openstack-tox-py37/95cda0b/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/77/657177/5/check/openstack-tox-docs/01ef4f5/html/ : SUCCESS in 4m 51s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/77/657177/5/check/neutron-tempest-plugin-designate-scenario/dba9c7e/ : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":5},{"id":"d08290429ebab475af444790c9da909bf495ad90","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-27 09:52:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e0b9ba67db864c4fb9b404c9c58dff15c39d7e32","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-09 08:34:45.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2ad10e790749ef7d6a8206050f9efe069c381e8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 08:34:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b6c233511acc211cd25ec1285b028d21cb3f6e33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 09:30:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"2930a6cef5e1e13cf3e1b5d5692f8e3c5af241be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 09:30:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/77/657177/5/gate/designate-bind9/18b93c4/ : SUCCESS in 34m 38s\n- designate-pdns4 http://logs.openstack.org/77/657177/5/gate/designate-pdns4/4bd3550/ : SUCCESS in 37m 09s\n- designate-bind9-manager-model http://logs.openstack.org/77/657177/5/gate/designate-bind9-manager-model/21f193c/ : SUCCESS in 34m 16s\n- designate-bind9-py27 http://logs.openstack.org/77/657177/5/gate/designate-bind9-py27/95a768d/ : SUCCESS in 33m 57s\n- designate-pdns4-py27 http://logs.openstack.org/77/657177/5/gate/designate-pdns4-py27/7916a31/ : SUCCESS in 32m 08s\n- designate-pdns4-postgres http://logs.openstack.org/77/657177/5/gate/designate-pdns4-postgres/d58a446/ : SUCCESS in 35m 28s\n- designate-grenade-pdns4 http://logs.openstack.org/77/657177/5/gate/designate-grenade-pdns4/679382e/ : SUCCESS in 54m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/657177/5/gate/openstack-tox-lower-constraints/bc4b0ca/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/77/657177/5/gate/openstack-tox-pep8/6dfa4df/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/77/657177/5/gate/openstack-tox-py27/c80ffbc/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/77/657177/5/gate/openstack-tox-py36/d9ea124/ : SUCCESS in 5m 18s\n- openstack-tox-py37 http://logs.openstack.org/77/657177/5/gate/openstack-tox-py37/5207b63/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/77/657177/5/gate/openstack-tox-docs/b274112/html/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4eed31a99fa257bf0dd57d16c4a712188821ab31","revisions":{"37c1f2c1a08f528e750fbfccb9dd9ec9d3663bf5":{"kind":"REWORK","_number":1,"created":"2019-05-05 01:04:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/657177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/657177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/657177/1"}}},"commit":{"parents":[{"commit":"c117412ad45d7bb9a592281a3681dc189ca0dd82","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c117412ad45d7bb9a592281a3681dc189ca0dd82"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-05 01:02:58.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-05 01:02:58.000000000","tz":-420},"subject":"Remove KS v2 support and add region support","message":"Remove KS v2 support and add region support\n\nKeystone V2 is no longer supported and is removed\nfrom the D2D driver in this patch. In addition\nwe also add support to pass a region_name on to the\ndriver.\n\nChange-Id: Ia3d2951f4e90817782bf959a6b2402eac26dbc7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/37c1f2c1a08f528e750fbfccb9dd9ec9d3663bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/37c1f2c1a08f528e750fbfccb9dd9ec9d3663bf5"}]},"branch":"refs/heads/master"},"3a2c3763aca99470e1a46991de39164c566eef1a":{"kind":"REWORK","_number":2,"created":"2019-05-05 02:41:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/657177/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/657177/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/657177/2"}}},"commit":{"parents":[{"commit":"c117412ad45d7bb9a592281a3681dc189ca0dd82","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c117412ad45d7bb9a592281a3681dc189ca0dd82"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-05 01:02:58.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-05 02:41:44.000000000","tz":-420},"subject":"Remove KS v2 support and add region support","message":"Remove KS v2 support and add region support\n\nKeystone V2 is no longer supported and is removed\nfrom the D2D driver in this patch. In addition\nwe also add support to pass a region_name on to the\ndriver.\n\nChange-Id: Ia3d2951f4e90817782bf959a6b2402eac26dbc7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3a2c3763aca99470e1a46991de39164c566eef1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3a2c3763aca99470e1a46991de39164c566eef1a"}]},"branch":"refs/heads/master"},"179bffdd1314be11bf9d0bba4a0ad07e5dc7a4d7":{"kind":"REWORK","_number":3,"created":"2019-05-05 03:15:32.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/657177/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/657177/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/657177/3"}}},"commit":{"parents":[{"commit":"c117412ad45d7bb9a592281a3681dc189ca0dd82","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c117412ad45d7bb9a592281a3681dc189ca0dd82"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-05 01:02:58.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-05 03:15:06.000000000","tz":-420},"subject":"Remove KS v2 support and add region support","message":"Remove KS v2 support and add region support\n\nKeystone V2 is no longer supported and is removed\nfrom the D2D driver in this patch. In addition\nwe also add support to pass a region_name on to the\ndriver.\n\nChange-Id: Ia3d2951f4e90817782bf959a6b2402eac26dbc7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/179bffdd1314be11bf9d0bba4a0ad07e5dc7a4d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/179bffdd1314be11bf9d0bba4a0ad07e5dc7a4d7"}]},"branch":"refs/heads/master"},"36b5711de0342d06dc9fa8da933c4c49d9a99859":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-20 00:09:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/657177/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/657177/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/657177/4"}}},"commit":{"parents":[{"commit":"178328f0a8d2485214814275f7b7fdaefe799aca","subject":"Merge \"Fixing stevedore warning when running devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/178328f0a8d2485214814275f7b7fdaefe799aca"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-05 01:02:58.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-20 00:09:49.000000000","tz":0},"subject":"Remove KS v2 support and add region support","message":"Remove KS v2 support and add region support\n\nKeystone V2 is no longer supported and is removed\nfrom the D2D driver in this patch. In addition\nwe also add support to pass a region_name on to the\ndriver.\n\nChange-Id: Ia3d2951f4e90817782bf959a6b2402eac26dbc7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/36b5711de0342d06dc9fa8da933c4c49d9a99859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/36b5711de0342d06dc9fa8da933c4c49d9a99859"}]},"branch":"refs/heads/master"},"4eed31a99fa257bf0dd57d16c4a712188821ab31":{"kind":"REWORK","_number":5,"created":"2019-05-26 04:05:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/77/657177/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/657177/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/657177/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/657177/5"}}},"commit":{"parents":[{"commit":"4e49b5581fca2bda0d1638efafa88efa8925f05e","subject":"Merge \"Modernized backend tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e49b5581fca2bda0d1638efafa88efa8925f05e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-05 01:02:58.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-26 04:04:55.000000000","tz":-420},"subject":"Remove KS v2 support and add region support","message":"Remove KS v2 support and add region support\n\nKeystone V2 is no longer supported and is removed\nfrom the D2D driver in this patch. In addition\nwe also add support to pass a region_name on to the\ndriver.\n\nChange-Id: Ia3d2951f4e90817782bf959a6b2402eac26dbc7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4eed31a99fa257bf0dd57d16c4a712188821ab31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4eed31a99fa257bf0dd57d16c4a712188821ab31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
