)]}'
{"id":"openstack%2Fdesignate~657382","triplet_id":"openstack%2Fdesignate~master~I1338d2ca180274ae84d4b358ff467e6655eb8022","project":"openstack/designate","branch":"master","topic":"bug/1827070","hashtags":[],"change_id":"I1338d2ca180274ae84d4b358ff467e6655eb8022","subject":"Designate service_statuses is not working","status":"MERGED","created":"2019-05-06 13:25:24.000000000","updated":"2019-05-08 13:16:09.000000000","submitted":"2019-05-08 13:16:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657382-1557321369360-b00d5b3e","meta_rev_id":"87ac98000872817033b505285cee04407b4fe656","_number":657382,"virtual_id_number":657382,"owner":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2019-05-06 14:37:57.000000000","_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":2,"date":"2019-05-08 13:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-05-08 12:15:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-07 02:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-05-08 12:15:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"date":"2019-05-07 02:56:41.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 02:21:59.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-05-07 02:56:41.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-05-08 12:15:23.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-08 13:16:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21485762940086aeb33adea622ba912cebd3014d","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-05-06 13:25:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d597d0bbc0b8fc9f656f37e1ebdcc20e7369535","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 14:33:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/82/657382/1/check/designate-bind9/07829ac/ : SUCCESS in 39m 51s\n- designate-pdns4 http://logs.openstack.org/82/657382/1/check/designate-pdns4/64a0c17/ : SUCCESS in 40m 18s\n- designate-bind9-manager-model http://logs.openstack.org/82/657382/1/check/designate-bind9-manager-model/4935c03/ : SUCCESS in 43m 14s\n- designate-bind9-py27 http://logs.openstack.org/82/657382/1/check/designate-bind9-py27/4f333f1/ : SUCCESS in 41m 41s\n- designate-pdns4-py27 http://logs.openstack.org/82/657382/1/check/designate-pdns4-py27/e47e3b6/ : SUCCESS in 37m 17s\n- designate-pdns4-postgres http://logs.openstack.org/82/657382/1/check/designate-pdns4-postgres/fafebbb/ : SUCCESS in 36m 50s\n- designate-grenade-pdns4 http://logs.openstack.org/82/657382/1/check/designate-grenade-pdns4/3285d17/ : SUCCESS in 1h 02m 00s\n- openstack-tox-cover http://logs.openstack.org/82/657382/1/check/openstack-tox-cover/5de2c3a/ : FAILURE in 3m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/657382/1/check/openstack-tox-lower-constraints/4a48c13/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/82/657382/1/check/openstack-tox-pep8/d86219a/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/82/657382/1/check/openstack-tox-py27/0c8cb4d/ : SUCCESS in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/82/657382/1/check/openstack-tox-py36/ad6a990/ : SUCCESS in 5m 22s\n- openstack-tox-py37 http://logs.openstack.org/82/657382/1/check/openstack-tox-py37/005f4ff/ : SUCCESS in 7m 48s\n- openstack-tox-docs http://logs.openstack.org/82/657382/1/check/openstack-tox-docs/778ba9b/html/ : SUCCESS in 4m 46s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/82/657382/1/check/neutron-tempest-plugin-designate-scenario/370059a/ : SUCCESS in 37m 56s","accounts_in_message":[],"_revision_number":1},{"id":"dbae5ca7d1595b693458885bab54faa3755b3802","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-05-06 14:37:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"92433b67854e1b256b0d9f742a845759924949ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 15:40:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/82/657382/1/check/designate-bind9/d64c7cf/ : SUCCESS in 45m 22s\n- designate-pdns4 http://logs.openstack.org/82/657382/1/check/designate-pdns4/a097947/ : SUCCESS in 41m 14s\n- designate-bind9-manager-model http://logs.openstack.org/82/657382/1/check/designate-bind9-manager-model/2ef5ce8/ : SUCCESS in 38m 42s\n- designate-bind9-py27 http://logs.openstack.org/82/657382/1/check/designate-bind9-py27/297c63f/ : SUCCESS in 42m 02s\n- designate-pdns4-py27 http://logs.openstack.org/82/657382/1/check/designate-pdns4-py27/87280e5/ : SUCCESS in 36m 51s\n- designate-pdns4-postgres http://logs.openstack.org/82/657382/1/check/designate-pdns4-postgres/188e596/ : SUCCESS in 40m 12s\n- designate-grenade-pdns4 http://logs.openstack.org/82/657382/1/check/designate-grenade-pdns4/598d747/ : SUCCESS in 58m 10s\n- openstack-tox-cover http://logs.openstack.org/82/657382/1/check/openstack-tox-cover/bf14dda/cover/ : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/657382/1/check/openstack-tox-lower-constraints/4dd0562/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/82/657382/1/check/openstack-tox-pep8/85e98ed/ : SUCCESS in 9m 24s\n- openstack-tox-py27 http://logs.openstack.org/82/657382/1/check/openstack-tox-py27/fa595d6/ : SUCCESS in 6m 12s\n- openstack-tox-py36 http://logs.openstack.org/82/657382/1/check/openstack-tox-py36/4a78032/ : SUCCESS in 6m 40s\n- openstack-tox-py37 http://logs.openstack.org/82/657382/1/check/openstack-tox-py37/44a595e/ : SUCCESS in 6m 19s\n- openstack-tox-docs http://logs.openstack.org/82/657382/1/check/openstack-tox-docs/dc167fd/html/ : SUCCESS in 5m 09s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/82/657382/1/check/neutron-tempest-plugin-designate-scenario/72a9df6/ : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":1},{"id":"52ebb22a5f4160d067aa488b62d46828f358e133","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-06 17:42:24.000000000","message":"Patch Set 1: Code-Review+1\n\nWould it be possible to add some sort of test for this?","accounts_in_message":[],"_revision_number":1},{"id":"547f0eac2dfe9a5540550ceda77bead27d1e628a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-07 02:21:59.000000000","message":"Patch Set 1: Code-Review+2\n\nWe should probably look at re-designing the heartbeats at some point, but for now this looks good.","accounts_in_message":[],"_revision_number":1},{"id":"8c4280b542327915af38d85dc45f27826a8c7222","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-08 12:15:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"48d72db2a878d3c923afaa18546d705e3336d874","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 12:15:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10bb53c3c3d56283e64fab565940cddbeedeb0e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 13:16:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/82/657382/1/gate/designate-bind9/69814d6/ : SUCCESS in 35m 49s\n- designate-pdns4 http://logs.openstack.org/82/657382/1/gate/designate-pdns4/c6e4daf/ : SUCCESS in 37m 54s\n- designate-bind9-manager-model http://logs.openstack.org/82/657382/1/gate/designate-bind9-manager-model/7e4db9b/ : SUCCESS in 39m 41s\n- designate-bind9-py27 http://logs.openstack.org/82/657382/1/gate/designate-bind9-py27/e6cd7d9/ : SUCCESS in 41m 59s\n- designate-pdns4-py27 http://logs.openstack.org/82/657382/1/gate/designate-pdns4-py27/0821f1f/ : SUCCESS in 35m 21s\n- designate-pdns4-postgres http://logs.openstack.org/82/657382/1/gate/designate-pdns4-postgres/c07b3fe/ : SUCCESS in 38m 46s\n- designate-grenade-pdns4 http://logs.openstack.org/82/657382/1/gate/designate-grenade-pdns4/1684e03/ : SUCCESS in 55m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/657382/1/gate/openstack-tox-lower-constraints/ae47f57/ : SUCCESS in 6m 06s\n- openstack-tox-pep8 http://logs.openstack.org/82/657382/1/gate/openstack-tox-pep8/e892eb2/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/82/657382/1/gate/openstack-tox-py27/a172450/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/82/657382/1/gate/openstack-tox-py36/3188875/ : SUCCESS in 5m 32s\n- openstack-tox-py37 http://logs.openstack.org/82/657382/1/gate/openstack-tox-py37/05afab3/ : SUCCESS in 7m 44s\n- openstack-tox-docs http://logs.openstack.org/82/657382/1/gate/openstack-tox-docs/f560640/html/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d07e396dd66927416cb62f0c9ac5ad527f80bec5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 13:16:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9214a42a8e502b135224d5f7d0f3871ac7daa00f","revisions":{"9214a42a8e502b135224d5f7d0f3871ac7daa00f":{"kind":"REWORK","_number":1,"created":"2019-05-06 13:25:24.000000000","uploader":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"ref":"refs/changes/82/657382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/657382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/657382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/657382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/657382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/657382/1"}}},"commit":{"parents":[{"commit":"c117412ad45d7bb9a592281a3681dc189ca0dd82","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c117412ad45d7bb9a592281a3681dc189ca0dd82"}]}],"author":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-05-06 12:47:33.000000000","tz":540},"committer":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-05-06 12:47:33.000000000","tz":540},"subject":"Designate service_statuses is not working","message":"Designate service_statuses is not working\n\nWhen I set the 2 or more each designate service workers,\nthose service_status was not working.\nIn this case, The \u0027HeartBeatEmitter.start\u0027 is called\nby child processes.\nAnd The \u0027HeartBeatEmitter._emit_heartbeat\u0027 is called\nby Parent process.\nParent process\u0027 \u0027_running\u0027 value never change to TRUE\non parent process.\nSo in this case, designate isn\u0027t emitted the heartbeat.\n\nChange-Id: I1338d2ca180274ae84d4b358ff467e6655eb8022\nCloses-Bug: #1827070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9214a42a8e502b135224d5f7d0f3871ac7daa00f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9214a42a8e502b135224d5f7d0f3871ac7daa00f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
