)]}'
{"id":"openstack%2Fdesignate~657929","triplet_id":"openstack%2Fdesignate~master~I27ad92ae4f17e56e5879b7338db3543e5a67d8a5","project":"openstack/designate","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I27ad92ae4f17e56e5879b7338db3543e5a67d8a5","subject":"Update Python 3 test runtimes for Train","status":"MERGED","created":"2019-05-09 06:26:43.000000000","updated":"2019-05-20 17:58:55.000000000","submitted":"2019-05-20 17:58:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657929-1558375135369-33725622","meta_rev_id":"08013e0e0227d995c62f82a05a79a003e9057fa7","_number":657929,"virtual_id_number":657929,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"date":"2019-05-10 00:27:57.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-20 17:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-05-20 16:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-20 16:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-05-20 16:51:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 16:44:53.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-05-20 16:51:01.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-20 17:58:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"702cd8dd9c71b245413ba030ac46f8f6828f6918","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-09 06:26:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48839cc2b7a83790f22763bfbec1617de4138a05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 07:32:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/29/657929/1/check/test-release-openstack/5f6f353/ : SUCCESS in 7m 06s\n- designate-bind9 http://logs.openstack.org/29/657929/1/check/designate-bind9/7e35692/ : SUCCESS in 39m 22s\n- designate-pdns4 http://logs.openstack.org/29/657929/1/check/designate-pdns4/e1a3077/ : SUCCESS in 39m 33s\n- designate-bind9-manager-model http://logs.openstack.org/29/657929/1/check/designate-bind9-manager-model/15bf38f/ : SUCCESS in 1h 01m 39s\n- designate-bind9-py27 http://logs.openstack.org/29/657929/1/check/designate-bind9-py27/fa8b887/ : SUCCESS in 43m 39s\n- designate-pdns4-py27 http://logs.openstack.org/29/657929/1/check/designate-pdns4-py27/e1342a6/ : SUCCESS in 37m 09s\n- designate-pdns4-postgres http://logs.openstack.org/29/657929/1/check/designate-pdns4-postgres/40f55c8/ : SUCCESS in 30m 24s\n- designate-grenade-pdns4 http://logs.openstack.org/29/657929/1/check/designate-grenade-pdns4/af293f3/ : SUCCESS in 58m 36s\n- openstack-tox-cover http://logs.openstack.org/29/657929/1/check/openstack-tox-cover/b9cfe63/cover/ : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/657929/1/check/openstack-tox-lower-constraints/95b9d2e/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/29/657929/1/check/openstack-tox-pep8/c73ab61/ : SUCCESS in 8m 38s\n- openstack-tox-py27 http://logs.openstack.org/29/657929/1/check/openstack-tox-py27/9c730bb/ : SUCCESS in 5m 49s\n- openstack-tox-py36 http://logs.openstack.org/29/657929/1/check/openstack-tox-py36/2c95b76/ : SUCCESS in 5m 31s\n- openstack-tox-py37 http://logs.openstack.org/29/657929/1/check/openstack-tox-py37/4143b4e/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/29/657929/1/check/openstack-tox-docs/104e0cf/html/ : SUCCESS in 4m 55s\n- build-openstack-releasenotes http://logs.openstack.org/29/657929/1/check/build-openstack-releasenotes/7f5867c/html/ : SUCCESS in 5m 58s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/29/657929/1/check/neutron-tempest-plugin-designate-scenario/c6855dd/ : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":1},{"id":"abb3a59c1e2d37cb09c12e0cd17f4c50748730f8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-09 09:36:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2edc2094d96153d7f464806a788f71edb44434a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 10:42:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/29/657929/2/check/test-release-openstack/200f8db/ : SUCCESS in 3m 23s\n- designate-bind9 http://logs.openstack.org/29/657929/2/check/designate-bind9/75a3240/ : SUCCESS in 42m 56s\n- designate-pdns4 http://logs.openstack.org/29/657929/2/check/designate-pdns4/c66a5d8/ : SUCCESS in 38m 30s\n- designate-bind9-manager-model http://logs.openstack.org/29/657929/2/check/designate-bind9-manager-model/e713b38/ : SUCCESS in 40m 03s\n- designate-bind9-py27 http://logs.openstack.org/29/657929/2/check/designate-bind9-py27/927130b/ : SUCCESS in 39m 44s\n- designate-pdns4-py27 http://logs.openstack.org/29/657929/2/check/designate-pdns4-py27/ad29fb6/ : SUCCESS in 39m 43s\n- designate-pdns4-postgres http://logs.openstack.org/29/657929/2/check/designate-pdns4-postgres/79e45ef/ : SUCCESS in 36m 56s\n- designate-grenade-pdns4 http://logs.openstack.org/29/657929/2/check/designate-grenade-pdns4/3e481c5/ : SUCCESS in 1h 00m 13s\n- openstack-tox-cover http://logs.openstack.org/29/657929/2/check/openstack-tox-cover/03b06f2/cover/ : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/657929/2/check/openstack-tox-lower-constraints/46a1c77/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/29/657929/2/check/openstack-tox-pep8/a2f5e9a/ : FAILURE in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/29/657929/2/check/openstack-tox-py27/65d9d06/ : SUCCESS in 5m 34s\n- openstack-tox-py36 http://logs.openstack.org/29/657929/2/check/openstack-tox-py36/900d9dd/ : SUCCESS in 5m 50s\n- openstack-tox-py37 http://logs.openstack.org/29/657929/2/check/openstack-tox-py37/f79fb97/ : SUCCESS in 6m 30s\n- openstack-tox-docs http://logs.openstack.org/29/657929/2/check/openstack-tox-docs/1cf0dc1/html/ : SUCCESS in 5m 21s\n- build-openstack-releasenotes http://logs.openstack.org/29/657929/2/check/build-openstack-releasenotes/f68ea6d/html/ : SUCCESS in 9m 16s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/29/657929/2/check/neutron-tempest-plugin-designate-scenario/c8e675c/ : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":2},{"id":"87723f3487c94b4316c51a4b5a8ee1c93b5d3d6b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-10 00:27:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"822cda815760c64af2718ef27c7ed2b94fe61921","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 01:44:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/29/657929/2/check/test-release-openstack/324eeeb/ : SUCCESS in 3m 35s\n- designate-bind9 http://logs.openstack.org/29/657929/2/check/designate-bind9/502e4d1/ : SUCCESS in 34m 44s\n- designate-pdns4 http://logs.openstack.org/29/657929/2/check/designate-pdns4/b544fe3/ : SUCCESS in 39m 06s\n- designate-bind9-manager-model http://logs.openstack.org/29/657929/2/check/designate-bind9-manager-model/7457829/ : SUCCESS in 39m 24s\n- designate-bind9-py27 http://logs.openstack.org/29/657929/2/check/designate-bind9-py27/f1c31ea/ : SUCCESS in 41m 31s\n- designate-pdns4-py27 http://logs.openstack.org/29/657929/2/check/designate-pdns4-py27/96aa069/ : SUCCESS in 37m 21s\n- designate-pdns4-postgres http://logs.openstack.org/29/657929/2/check/designate-pdns4-postgres/990757d/ : SUCCESS in 33m 54s\n- designate-grenade-pdns4 http://logs.openstack.org/29/657929/2/check/designate-grenade-pdns4/278b840/ : SUCCESS in 59m 51s\n- openstack-tox-cover http://logs.openstack.org/29/657929/2/check/openstack-tox-cover/85cfe05/cover/ : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/657929/2/check/openstack-tox-lower-constraints/d05161c/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/29/657929/2/check/openstack-tox-pep8/5275ec6/ : FAILURE in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/29/657929/2/check/openstack-tox-py27/d986b4d/ : SUCCESS in 5m 49s\n- openstack-tox-py36 http://logs.openstack.org/29/657929/2/check/openstack-tox-py36/6562275/ : SUCCESS in 6m 03s\n- openstack-tox-py37 http://logs.openstack.org/29/657929/2/check/openstack-tox-py37/3e17778/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/29/657929/2/check/openstack-tox-docs/6272e28/html/ : SUCCESS in 5m 03s\n- build-openstack-releasenotes http://logs.openstack.org/29/657929/2/check/build-openstack-releasenotes/f84fae8/html/ : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/29/657929/2/check/neutron-tempest-plugin-designate-scenario/408d499/ : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4813c0b5b26fd5843316ea4d09a85e8e394fe9de","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-13 03:11:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"632a66fa25c964f014733211e3bb2947069a613c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 03:59:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/29/657929/2/check/test-release-openstack/080d717/ : SUCCESS in 3m 21s\n- designate-bind9 http://logs.openstack.org/29/657929/2/check/designate-bind9/246ddc7/ : SUCCESS in 35m 44s\n- designate-pdns4 http://logs.openstack.org/29/657929/2/check/designate-pdns4/04f7bf5/ : FAILURE in 27m 01s\n- designate-bind9-manager-model http://logs.openstack.org/29/657929/2/check/designate-bind9-manager-model/ee305f8/ : FAILURE in 27m 14s\n- designate-bind9-py27 http://logs.openstack.org/29/657929/2/check/designate-bind9-py27/9be739b/ : SUCCESS in 38m 08s\n- designate-pdns4-py27 http://logs.openstack.org/29/657929/2/check/designate-pdns4-py27/a2449a5/ : FAILURE in 41m 05s\n- designate-pdns4-postgres http://logs.openstack.org/29/657929/2/check/designate-pdns4-postgres/16a9a44/ : FAILURE in 26m 59s\n- designate-grenade-pdns4 http://logs.openstack.org/29/657929/2/check/designate-grenade-pdns4/caa9604/ : SUCCESS in 45m 21s\n- openstack-tox-cover http://logs.openstack.org/29/657929/2/check/openstack-tox-cover/45d29fd/cover/ : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/657929/2/check/openstack-tox-lower-constraints/dee8c57/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/29/657929/2/check/openstack-tox-pep8/9c90c60/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/29/657929/2/check/openstack-tox-py27/a0e069b/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/29/657929/2/check/openstack-tox-py36/59a2b57/ : SUCCESS in 5m 06s\n- openstack-tox-py37 http://logs.openstack.org/29/657929/2/check/openstack-tox-py37/8879ac3/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/29/657929/2/check/openstack-tox-docs/56dfcf3/html/ : SUCCESS in 4m 51s\n- build-openstack-releasenotes http://logs.openstack.org/29/657929/2/check/build-openstack-releasenotes/8607141/html/ : SUCCESS in 6m 13s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/29/657929/2/check/neutron-tempest-plugin-designate-scenario/ca60a8e/ : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0365913108def09bca1181c9d1d80a4239149df3","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-20 10:43:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"547ac9a90c30a2635abbc84ba5d46c5a9cc39c19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:49:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/29/657929/2/check/test-release-openstack/5357306/ : SUCCESS in 3m 39s\n- designate-bind9 http://logs.openstack.org/29/657929/2/check/designate-bind9/bd84fd2/ : SUCCESS in 39m 18s\n- designate-pdns4 http://logs.openstack.org/29/657929/2/check/designate-pdns4/f6f3aa9/ : SUCCESS in 38m 51s\n- designate-bind9-manager-model http://logs.openstack.org/29/657929/2/check/designate-bind9-manager-model/5aecbd4/ : SUCCESS in 35m 13s\n- designate-bind9-py27 http://logs.openstack.org/29/657929/2/check/designate-bind9-py27/4ff9c7f/ : SUCCESS in 39m 28s\n- designate-pdns4-py27 http://logs.openstack.org/29/657929/2/check/designate-pdns4-py27/0b90209/ : SUCCESS in 40m 29s\n- designate-pdns4-postgres http://logs.openstack.org/29/657929/2/check/designate-pdns4-postgres/44ab8c6/ : SUCCESS in 36m 59s\n- designate-grenade-pdns4 http://logs.openstack.org/29/657929/2/check/designate-grenade-pdns4/997c0b8/ : SUCCESS in 56m 59s\n- openstack-tox-cover http://logs.openstack.org/29/657929/2/check/openstack-tox-cover/f7bf950/cover/ : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/657929/2/check/openstack-tox-lower-constraints/d863251/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/29/657929/2/check/openstack-tox-pep8/edb60ed/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/29/657929/2/check/openstack-tox-py27/ab6349d/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/29/657929/2/check/openstack-tox-py36/d2b2284/ : SUCCESS in 6m 10s\n- openstack-tox-py37 http://logs.openstack.org/29/657929/2/check/openstack-tox-py37/c4d7e69/ : SUCCESS in 8m 03s\n- openstack-tox-docs http://logs.openstack.org/29/657929/2/check/openstack-tox-docs/2ae7261/html/ : SUCCESS in 5m 00s\n- build-openstack-releasenotes http://logs.openstack.org/29/657929/2/check/build-openstack-releasenotes/b61c6bd/html/ : SUCCESS in 6m 19s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/29/657929/2/check/neutron-tempest-plugin-designate-scenario/6564a14/ : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":2},{"id":"ce844591987b3bd949d0ff8e2b3d354efa77a9b5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 16:44:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a448cbafbf49a300486047c22157cdd412fa6548","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-20 16:51:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ecb48f5e66537bfaf28b7883bae78c1053839f29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 16:51:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"717b421f3c9eb9969ee1ceb4b2f3af7d7d7f74b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:58:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4601898d205730d006b8ebd4a5f38d009219324b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:58:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/29/657929/2/gate/test-release-openstack/2f60710/ : SUCCESS in 3m 30s\n- designate-bind9 http://logs.openstack.org/29/657929/2/gate/designate-bind9/c799354/ : SUCCESS in 40m 18s\n- designate-pdns4 http://logs.openstack.org/29/657929/2/gate/designate-pdns4/12046b5/ : SUCCESS in 34m 54s\n- designate-bind9-manager-model http://logs.openstack.org/29/657929/2/gate/designate-bind9-manager-model/5788105/ : SUCCESS in 39m 56s\n- designate-bind9-py27 http://logs.openstack.org/29/657929/2/gate/designate-bind9-py27/9c0e6b5/ : SUCCESS in 36m 48s\n- designate-pdns4-py27 http://logs.openstack.org/29/657929/2/gate/designate-pdns4-py27/9d96833/ : SUCCESS in 38m 35s\n- designate-pdns4-postgres http://logs.openstack.org/29/657929/2/gate/designate-pdns4-postgres/f12c125/ : SUCCESS in 38m 47s\n- designate-grenade-pdns4 http://logs.openstack.org/29/657929/2/gate/designate-grenade-pdns4/a15d518/ : SUCCESS in 1h 04m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/657929/2/gate/openstack-tox-lower-constraints/751325c/ : SUCCESS in 6m 47s\n- openstack-tox-pep8 http://logs.openstack.org/29/657929/2/gate/openstack-tox-pep8/f7b062f/ : SUCCESS in 6m 59s\n- openstack-tox-py27 http://logs.openstack.org/29/657929/2/gate/openstack-tox-py27/d33a6ef/ : SUCCESS in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/29/657929/2/gate/openstack-tox-py36/36836a8/ : SUCCESS in 4m 59s\n- openstack-tox-py37 http://logs.openstack.org/29/657929/2/gate/openstack-tox-py37/faef4e6/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/29/657929/2/gate/openstack-tox-docs/46b1a6c/html/ : SUCCESS in 5m 28s\n- build-openstack-releasenotes http://logs.openstack.org/29/657929/2/gate/build-openstack-releasenotes/e7c7a86/html/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"27d2c22e8a7799c9d5ed93cf455f22007b262749","revisions":{"2d4b1888cd1d655d2d2e05d0b25db4d4b3965bfc":{"kind":"REWORK","_number":1,"created":"2019-05-09 06:26:43.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/29/657929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/657929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/657929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/657929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/657929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/657929/1"}}},"commit":{"parents":[{"commit":"9214a42a8e502b135224d5f7d0f3871ac7daa00f","subject":"Designate service_statuses is not working","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9214a42a8e502b135224d5f7d0f3871ac7daa00f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:01.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:01.000000000","tz":480},"subject":"Update Python 3 test runtimes for Train","message":"Update Python 3 test runtimes for Train\n\nFollowing the process set down in the TC resolution on updating Python\nversions:\n\nhttps://governance.openstack.org/tc/resolutions/20181024-python-update-process.html\n\nChange-Id: I27ad92ae4f17e56e5879b7338db3543e5a67d8a5\nDepends-On: https://review.opendev.org/#/c/641878/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2d4b1888cd1d655d2d2e05d0b25db4d4b3965bfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2d4b1888cd1d655d2d2e05d0b25db4d4b3965bfc"}]},"branch":"refs/heads/master"},"27d2c22e8a7799c9d5ed93cf455f22007b262749":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-09 09:36:07.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/29/657929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/657929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/657929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/657929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/657929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/657929/2"}}},"commit":{"parents":[{"commit":"9214a42a8e502b135224d5f7d0f3871ac7daa00f","subject":"Designate service_statuses is not working","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9214a42a8e502b135224d5f7d0f3871ac7daa00f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:01.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 09:34:58.000000000","tz":480},"subject":"Update Python 3 test runtimes for Train","message":"Update Python 3 test runtimes for Train\n\nThis goal is to implement the process set out in the 2018-10-24 Python\nUpdate Process TC resolution[1], for the Train cycle to ensure unit\ntesting is in place for all of the Tested Runtimes for Train[2].\nIn practice, this generally means adding unit tests for Python 3.7 and dropping\nunit tests for Python 3.5. Using the Zuul template for Train will ensure that\nall projects that support Python3 will be tested against the agreed runtime\nversions, and make it easier to update them in future.\n\n[1]https://governance.openstack.org/tc/resolutions/20181024-python-update-process.html\n[2]https://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I27ad92ae4f17e56e5879b7338db3543e5a67d8a5\nDepends-On: https://review.opendev.org/#/c/641878/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27d2c22e8a7799c9d5ed93cf455f22007b262749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27d2c22e8a7799c9d5ed93cf455f22007b262749"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
