)]}'
{"id":"openstack%2Fdesignate~658227","triplet_id":"openstack%2Fdesignate~master~Id251376af55b71d7a5b28ef0839baf89c01aac1f","project":"openstack/designate","branch":"master","topic":"improved_tests","hashtags":[],"change_id":"Id251376af55b71d7a5b28ef0839baf89c01aac1f","subject":"Added PowerDNS 4 unit tests","status":"MERGED","created":"2019-05-09 23:25:58.000000000","updated":"2019-05-20 22:28:03.000000000","submitted":"2019-05-20 22:28:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658227-1558391283644-b16e9fe7","meta_rev_id":"7a6a001d6add300c2bc77ca216ebfef994292199","_number":658227,"virtual_id_number":658227,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-05-20 22:28:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-05-20 21:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-05-20 21:17:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-19 21:33:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-05-20 21:17:20.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-20 22:28:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c828c3ab40974ac59640ac0bfa18967c8c7364b7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-09 23:25:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b73341dc82ce1365389baf365a14613ae511ba99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 00:26:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/27/658227/1/check/designate-bind9/a070b40/ : SUCCESS in 34m 45s\n- designate-pdns4 http://logs.openstack.org/27/658227/1/check/designate-pdns4/316369b/ : SUCCESS in 38m 34s\n- designate-bind9-manager-model http://logs.openstack.org/27/658227/1/check/designate-bind9-manager-model/1f81e32/ : SUCCESS in 42m 14s\n- designate-bind9-py27 http://logs.openstack.org/27/658227/1/check/designate-bind9-py27/3695f66/ : SUCCESS in 42m 28s\n- designate-pdns4-py27 http://logs.openstack.org/27/658227/1/check/designate-pdns4-py27/aba360e/ : SUCCESS in 32m 04s\n- designate-pdns4-postgres http://logs.openstack.org/27/658227/1/check/designate-pdns4-postgres/960bace/ : SUCCESS in 32m 53s\n- designate-grenade-pdns4 http://logs.openstack.org/27/658227/1/check/designate-grenade-pdns4/e3e4c99/ : SUCCESS in 55m 52s\n- openstack-tox-cover http://logs.openstack.org/27/658227/1/check/openstack-tox-cover/97df668/cover/ : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/658227/1/check/openstack-tox-lower-constraints/77237c5/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/27/658227/1/check/openstack-tox-pep8/bf1ba91/ : FAILURE in 6m 13s\n- openstack-tox-py27 http://logs.openstack.org/27/658227/1/check/openstack-tox-py27/850ed33/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/27/658227/1/check/openstack-tox-py36/daaf6f9/ : SUCCESS in 4m 53s\n- openstack-tox-py37 http://logs.openstack.org/27/658227/1/check/openstack-tox-py37/c70755e/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/27/658227/1/check/openstack-tox-docs/52e1426/html/ : SUCCESS in 4m 34s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/27/658227/1/check/neutron-tempest-plugin-designate-scenario/02da540/ : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a7ad62e5cd821c4150c37de84b1a783f6105bd4c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-10 02:38:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5985bef684777cb438b3072ec735337559115b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 03:33:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/27/658227/1/check/designate-bind9/bd4f6b4/ : SUCCESS in 33m 14s\n- designate-pdns4 http://logs.openstack.org/27/658227/1/check/designate-pdns4/1fee731/ : SUCCESS in 34m 19s\n- designate-bind9-manager-model http://logs.openstack.org/27/658227/1/check/designate-bind9-manager-model/7b1e161/ : SUCCESS in 36m 00s\n- designate-bind9-py27 http://logs.openstack.org/27/658227/1/check/designate-bind9-py27/e25b45c/ : SUCCESS in 39m 21s\n- designate-pdns4-py27 http://logs.openstack.org/27/658227/1/check/designate-pdns4-py27/2e7c578/ : SUCCESS in 33m 40s\n- designate-pdns4-postgres http://logs.openstack.org/27/658227/1/check/designate-pdns4-postgres/59feab5/ : SUCCESS in 35m 33s\n- designate-grenade-pdns4 http://logs.openstack.org/27/658227/1/check/designate-grenade-pdns4/19f514c/ : SUCCESS in 53m 26s\n- openstack-tox-cover http://logs.openstack.org/27/658227/1/check/openstack-tox-cover/24976f3/cover/ : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/658227/1/check/openstack-tox-lower-constraints/4c449c6/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/27/658227/1/check/openstack-tox-pep8/99244a1/ : FAILURE in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/27/658227/1/check/openstack-tox-py27/76cba53/ : SUCCESS in 5m 51s\n- openstack-tox-py36 http://logs.openstack.org/27/658227/1/check/openstack-tox-py36/dfb72e9/ : SUCCESS in 5m 46s\n- openstack-tox-py37 http://logs.openstack.org/27/658227/1/check/openstack-tox-py37/a7f559e/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/27/658227/1/check/openstack-tox-docs/6a7600e/html/ : SUCCESS in 4m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/27/658227/1/check/neutron-tempest-plugin-designate-scenario/83938a1/ : SUCCESS in 29m 50s","accounts_in_message":[],"_revision_number":1},{"id":"144e46deceb1fb1fe089d6c08b4001f1a44f5020","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-10 18:55:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7ff9f5687be84d720f9c7fb5b20cde991bd3ab73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:54:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/27/658227/1/check/designate-bind9/df1738e/ : SUCCESS in 39m 56s\n- designate-pdns4 http://logs.openstack.org/27/658227/1/check/designate-pdns4/18b84e7/ : SUCCESS in 39m 01s\n- designate-bind9-manager-model http://logs.openstack.org/27/658227/1/check/designate-bind9-manager-model/b05d1ea/ : SUCCESS in 35m 48s\n- designate-bind9-py27 http://logs.openstack.org/27/658227/1/check/designate-bind9-py27/e16cb4a/ : SUCCESS in 43m 57s\n- designate-pdns4-py27 http://logs.openstack.org/27/658227/1/check/designate-pdns4-py27/c2cc23a/ : SUCCESS in 38m 29s\n- designate-pdns4-postgres http://logs.openstack.org/27/658227/1/check/designate-pdns4-postgres/b8a1d70/ : SUCCESS in 37m 58s\n- designate-grenade-pdns4 http://logs.openstack.org/27/658227/1/check/designate-grenade-pdns4/5d5790a/ : SUCCESS in 57m 29s\n- openstack-tox-cover http://logs.openstack.org/27/658227/1/check/openstack-tox-cover/4193cdd/cover/ : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/658227/1/check/openstack-tox-lower-constraints/4149e55/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/27/658227/1/check/openstack-tox-pep8/6505bbe/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/27/658227/1/check/openstack-tox-py27/268a849/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/27/658227/1/check/openstack-tox-py36/9967708/ : SUCCESS in 5m 09s\n- openstack-tox-py37 http://logs.openstack.org/27/658227/1/check/openstack-tox-py37/426c187/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/27/658227/1/check/openstack-tox-docs/cfc09b4/html/ : SUCCESS in 5m 10s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/27/658227/1/check/neutron-tempest-plugin-designate-scenario/6cbc28d/ : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":1},{"id":"24dfe78cb308c4d13acb05a141733cab18920a68","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 06:31:30.000000000","message":"Topic changed from pdns4_tests to improved_tests","accounts_in_message":[],"_revision_number":1},{"id":"21bc16356be8071827d85ac1af4c1fefd4702940","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-20 10:48:31.000000000","message":"Patch Set 1:\n\nwe should probably put this here: https://opendev.org/openstack/designate/src/branch/master/designate/tests/unit/test_backend","accounts_in_message":[],"_revision_number":1},{"id":"45a350a3f45e1608d99a0123acc35c39f1e5f5ba","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 16:35:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17e0d88a100ae3886216d67efedbf2a93956d1aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:32:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/27/658227/2/check/designate-bind9/0a39e9d/ : SUCCESS in 41m 32s\n- designate-pdns4 http://logs.openstack.org/27/658227/2/check/designate-pdns4/5889665/ : SUCCESS in 41m 41s\n- designate-bind9-manager-model http://logs.openstack.org/27/658227/2/check/designate-bind9-manager-model/3faf70c/ : SUCCESS in 41m 01s\n- designate-bind9-py27 http://logs.openstack.org/27/658227/2/check/designate-bind9-py27/b7f5942/ : SUCCESS in 41m 32s\n- designate-pdns4-py27 http://logs.openstack.org/27/658227/2/check/designate-pdns4-py27/2b6ee8e/ : SUCCESS in 33m 27s\n- designate-pdns4-postgres http://logs.openstack.org/27/658227/2/check/designate-pdns4-postgres/8963c54/ : SUCCESS in 38m 43s\n- designate-grenade-pdns4 http://logs.openstack.org/27/658227/2/check/designate-grenade-pdns4/8cbe76d/ : SUCCESS in 53m 49s\n- openstack-tox-cover http://logs.openstack.org/27/658227/2/check/openstack-tox-cover/1760fd6/cover/ : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/658227/2/check/openstack-tox-lower-constraints/e72bcd6/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/27/658227/2/check/openstack-tox-pep8/18f1ebf/ : SUCCESS in 7m 03s\n- openstack-tox-py27 http://logs.openstack.org/27/658227/2/check/openstack-tox-py27/3806546/ : SUCCESS in 9m 12s\n- openstack-tox-py36 http://logs.openstack.org/27/658227/2/check/openstack-tox-py36/49bc0e3/ : SUCCESS in 5m 25s\n- openstack-tox-py37 http://logs.openstack.org/27/658227/2/check/openstack-tox-py37/09dfa87/ : SUCCESS in 6m 08s\n- openstack-tox-docs http://logs.openstack.org/27/658227/2/check/openstack-tox-docs/2d1ef8b/html/ : SUCCESS in 4m 31s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/27/658227/2/check/neutron-tempest-plugin-designate-scenario/b3bdc3c/ : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":2},{"id":"386f24d3ea134954f8506227b51089eb0abd1d49","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-20 21:17:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3fa806a31145375c9d29309e65577c1eeb70b26b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 21:17:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0cf23a2f8c70ac785f2250bcfa2f55113849a4a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 22:28:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7a6a001d6add300c2bc77ca216ebfef994292199","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 22:28:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/27/658227/2/gate/designate-bind9/450e7f8/ : SUCCESS in 35m 13s\n- designate-pdns4 http://logs.openstack.org/27/658227/2/gate/designate-pdns4/036dc99/ : SUCCESS in 45m 59s\n- designate-bind9-manager-model http://logs.openstack.org/27/658227/2/gate/designate-bind9-manager-model/b4b17bd/ : SUCCESS in 38m 44s\n- designate-bind9-py27 http://logs.openstack.org/27/658227/2/gate/designate-bind9-py27/39e092b/ : SUCCESS in 39m 11s\n- designate-pdns4-py27 http://logs.openstack.org/27/658227/2/gate/designate-pdns4-py27/51fec31/ : SUCCESS in 36m 47s\n- designate-pdns4-postgres http://logs.openstack.org/27/658227/2/gate/designate-pdns4-postgres/c4dd421/ : SUCCESS in 38m 14s\n- designate-grenade-pdns4 http://logs.openstack.org/27/658227/2/gate/designate-grenade-pdns4/edf2834/ : SUCCESS in 1h 00m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/658227/2/gate/openstack-tox-lower-constraints/a167043/ : SUCCESS in 6m 43s\n- openstack-tox-pep8 http://logs.openstack.org/27/658227/2/gate/openstack-tox-pep8/1ef6251/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/27/658227/2/gate/openstack-tox-py27/ca85de6/ : SUCCESS in 6m 15s\n- openstack-tox-py36 http://logs.openstack.org/27/658227/2/gate/openstack-tox-py36/a97e64c/ : SUCCESS in 6m 00s\n- openstack-tox-py37 http://logs.openstack.org/27/658227/2/gate/openstack-tox-py37/f057d3d/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/27/658227/2/gate/openstack-tox-docs/6298237/html/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e18436a44ac97a853e03e56fbdddc0b6ac62de34","revisions":{"8a938dda5876e78231f1cff15cf3008146d9e4c0":{"kind":"REWORK","_number":1,"created":"2019-05-09 23:25:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/658227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/658227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/658227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/658227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/658227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/658227/1"}}},"commit":{"parents":[{"commit":"9512ef19800532d6d2f2195c76abf5d85de4857c","subject":"Ensure pdns4 driver is consistent with other drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9512ef19800532d6d2f2195c76abf5d85de4857c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-09 23:25:33.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-09 23:25:33.000000000","tz":-420},"subject":"Added PowerDNS 4 unit tests","message":"Added PowerDNS 4 unit tests\n\nChange-Id: Id251376af55b71d7a5b28ef0839baf89c01aac1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a938dda5876e78231f1cff15cf3008146d9e4c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a938dda5876e78231f1cff15cf3008146d9e4c0"}]},"branch":"refs/heads/master"},"e18436a44ac97a853e03e56fbdddc0b6ac62de34":{"kind":"REWORK","_number":2,"created":"2019-05-20 16:35:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/658227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/658227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/658227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/658227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/658227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/658227/2"}}},"commit":{"parents":[{"commit":"9512ef19800532d6d2f2195c76abf5d85de4857c","subject":"Ensure pdns4 driver is consistent with other drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9512ef19800532d6d2f2195c76abf5d85de4857c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-09 23:25:33.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-20 16:35:48.000000000","tz":-420},"subject":"Added PowerDNS 4 unit tests","message":"Added PowerDNS 4 unit tests\n\nChange-Id: Id251376af55b71d7a5b28ef0839baf89c01aac1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e18436a44ac97a853e03e56fbdddc0b6ac62de34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e18436a44ac97a853e03e56fbdddc0b6ac62de34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
