)]}'
{"id":"openstack%2Fdesignate~659167","triplet_id":"openstack%2Fdesignate~master~I4b57336a103c75df877ff65e645b3b05b3c79a49","project":"openstack/designate","branch":"master","topic":"bandit_exclusion_designate","hashtags":[],"change_id":"I4b57336a103c75df877ff65e645b3b05b3c79a49","subject":"Clean fix for bandit exclusions change","status":"MERGED","created":"2019-05-14 21:14:56.000000000","updated":"2019-05-15 14:16:00.000000000","submitted":"2019-05-15 14:15:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659167-1557929760063-7c1c31b6","meta_rev_id":"43f3fd80264873d6999b76f8c242d55ec392befc","_number":659167,"virtual_id_number":659167,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-15 00:04:16.000000000","_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-05-15 14:15:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-05-15 13:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-15 02:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-05-15 13:09:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 02:08:20.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-05-15 13:09:03.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-15 14:15:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"484ff2601a58a7c80245b7864b95bdefea39d11b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 21:14:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c59b10e2d9d45595c1318098d3ac39b6a19fa3a1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 21:15:47.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 3ead9f84288e2c06c9905d55d560e7c0c61d9045","accounts_in_message":[],"_revision_number":1},{"id":"ee398fdc3e93c4a48a361c8cd0e017b9dcf178f1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 21:15:55.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit f586e321928e44e888f240e64ff9e201d527f9b8","accounts_in_message":[],"_revision_number":1},{"id":"cbeba5ce1e289b0287209315b28d0c086544919f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 22:09:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/67/659167/1/check/build-openstack-api-ref/4571a92/html/ : SUCCESS in 5m 14s\n- designate-bind9 http://logs.openstack.org/67/659167/1/check/designate-bind9/1e6b0dd/ : SUCCESS in 35m 12s\n- designate-pdns4 http://logs.openstack.org/67/659167/1/check/designate-pdns4/b5ed252/ : SUCCESS in 44m 06s\n- designate-bind9-manager-model http://logs.openstack.org/67/659167/1/check/designate-bind9-manager-model/b7bd62b/ : SUCCESS in 53m 18s\n- designate-bind9-py27 http://logs.openstack.org/67/659167/1/check/designate-bind9-py27/835a342/ : SUCCESS in 35m 51s\n- designate-pdns4-py27 http://logs.openstack.org/67/659167/1/check/designate-pdns4-py27/fcb38ce/ : SUCCESS in 44m 32s\n- designate-pdns4-postgres http://logs.openstack.org/67/659167/1/check/designate-pdns4-postgres/33f3a2e/ : SUCCESS in 39m 28s\n- designate-grenade-pdns4 http://logs.openstack.org/67/659167/1/check/designate-grenade-pdns4/125695e/ : SUCCESS in 48m 45s\n- openstack-tox-cover http://logs.openstack.org/67/659167/1/check/openstack-tox-cover/cdbf92f/cover/ : SUCCESS in 6m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/659167/1/check/openstack-tox-lower-constraints/9356fec/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/67/659167/1/check/openstack-tox-pep8/f9fd78b/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/67/659167/1/check/openstack-tox-py27/046e16c/ : SUCCESS in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/67/659167/1/check/openstack-tox-py36/a3afbfc/ : SUCCESS in 6m 16s\n- openstack-tox-py37 http://logs.openstack.org/67/659167/1/check/openstack-tox-py37/cd5a1b0/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/67/659167/1/check/openstack-tox-docs/d2eeb1b/html/ : SUCCESS in 4m 43s\n- requirements-check http://logs.openstack.org/67/659167/1/check/requirements-check/ce03dd4/ : FAILURE in 3m 43s\n- build-openstack-releasenotes http://logs.openstack.org/67/659167/1/check/build-openstack-releasenotes/8b0555f/html/ : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a0147c2615bea92e10fdc347cf63f2bad63e1d82","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 23:11:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"049dfd88ab5f52376569974d1f8e763be326231e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 23:57:32.000000000","message":"Topic set to bandit_exclusion_designate","accounts_in_message":[],"_revision_number":2},{"id":"ad37c6d505054f39203e8da8110d39cc376e2400","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-15 00:04:16.000000000","message":"Patch Set 2:\n\nI believe the choice made on https://review.opendev.org/#/c/658258/ was not the best one.\n\nI\u0027ve re-proposed a similar fix here, which is easily backported (backporting reqs changes is unsightly, if it\u0027s even allowed).\n\nIt didn\u0027t turn out to be as relevant for designate, but the new bandit did have other changes (new rules) that should be run, and will be run in future versions too (keystone had to fix things due to this, that were missed originally with blacklisting, see: https://review.opendev.org/#/c/659164/).\n\nOverall, blacklisting a requirement seems like an overkill reaction to something that\u0027s solved by adding an asterisk and some quotes...","accounts_in_message":[],"_revision_number":2},{"id":"598fd82ea859f61d888ac240858c2eb75fdbb25f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 00:14:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/67/659167/2/check/build-openstack-api-ref/afb8cb4/html/ : SUCCESS in 4m 17s\n- designate-bind9 http://logs.openstack.org/67/659167/2/check/designate-bind9/13257c8/ : SUCCESS in 36m 39s\n- designate-pdns4 http://logs.openstack.org/67/659167/2/check/designate-pdns4/c1a6c66/ : SUCCESS in 35m 30s\n- designate-bind9-manager-model http://logs.openstack.org/67/659167/2/check/designate-bind9-manager-model/2498f8c/ : SUCCESS in 38m 19s\n- designate-bind9-py27 http://logs.openstack.org/67/659167/2/check/designate-bind9-py27/b997316/ : SUCCESS in 42m 38s\n- designate-pdns4-py27 http://logs.openstack.org/67/659167/2/check/designate-pdns4-py27/02f33d5/ : SUCCESS in 36m 30s\n- designate-pdns4-postgres http://logs.openstack.org/67/659167/2/check/designate-pdns4-postgres/2ab6f69/ : SUCCESS in 38m 58s\n- designate-grenade-pdns4 http://logs.openstack.org/67/659167/2/check/designate-grenade-pdns4/b9fc582/ : SUCCESS in 56m 37s\n- openstack-tox-cover http://logs.openstack.org/67/659167/2/check/openstack-tox-cover/826a3e1/cover/ : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/659167/2/check/openstack-tox-lower-constraints/1f01d68/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/67/659167/2/check/openstack-tox-pep8/1191e15/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/67/659167/2/check/openstack-tox-py27/8a37e4d/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/67/659167/2/check/openstack-tox-py36/626c767/ : SUCCESS in 5m 15s\n- openstack-tox-py37 http://logs.openstack.org/67/659167/2/check/openstack-tox-py37/6252983/ : SUCCESS in 6m 40s\n- openstack-tox-docs http://logs.openstack.org/67/659167/2/check/openstack-tox-docs/24f0960/html/ : SUCCESS in 4m 25s\n- requirements-check http://logs.openstack.org/67/659167/2/check/requirements-check/13d4b28/ : SUCCESS in 3m 39s\n- build-openstack-releasenotes http://logs.openstack.org/67/659167/2/check/build-openstack-releasenotes/f4a5c53/html/ : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"a45860178bffc67a7dc7290707ce021c6d352730","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-15 02:08:20.000000000","message":"Patch Set 2: Code-Review+2\n\nPersonally I don\u0027t have an issue with --exclude tests, as while it\u0027s broken in 1.6.0 is something that is likely to work in all other future versions. It\u0027s just a regression bug.\nhttps://github.com/PyCQA/bandit/issues/488\n\nHaving that said, I don\u0027t have anything against making use of their new way of excluding.","accounts_in_message":[],"_revision_number":2},{"id":"ec6fea1f81ef7ce19597965f8143140062e9f7a9","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-15 13:09:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nYeah. this is a simple enough fix","accounts_in_message":[],"_revision_number":2},{"id":"32e1c37409bcf0d5a4ad0fcd5d02d3c544b3c924","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:10:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc55774ba1698c6cc71603f24a0a4e5aba10fa02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:15:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/67/659167/2/gate/build-openstack-api-ref/ba784e3/html/ : SUCCESS in 4m 32s\n- designate-bind9 http://logs.openstack.org/67/659167/2/gate/designate-bind9/084e6ba/ : SUCCESS in 44m 49s\n- designate-pdns4 http://logs.openstack.org/67/659167/2/gate/designate-pdns4/6213bf1/ : SUCCESS in 37m 29s\n- designate-bind9-manager-model http://logs.openstack.org/67/659167/2/gate/designate-bind9-manager-model/8139fc5/ : SUCCESS in 36m 03s\n- designate-bind9-py27 http://logs.openstack.org/67/659167/2/gate/designate-bind9-py27/2ca4080/ : SUCCESS in 32m 29s\n- designate-pdns4-py27 http://logs.openstack.org/67/659167/2/gate/designate-pdns4-py27/36f25ac/ : SUCCESS in 47m 55s\n- designate-pdns4-postgres http://logs.openstack.org/67/659167/2/gate/designate-pdns4-postgres/56e5216/ : SUCCESS in 42m 58s\n- designate-grenade-pdns4 http://logs.openstack.org/67/659167/2/gate/designate-grenade-pdns4/2e4df60/ : SUCCESS in 59m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/659167/2/gate/openstack-tox-lower-constraints/3ca71b7/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 http://logs.openstack.org/67/659167/2/gate/openstack-tox-pep8/cd047ba/ : SUCCESS in 7m 06s\n- openstack-tox-py27 http://logs.openstack.org/67/659167/2/gate/openstack-tox-py27/96d5ea3/ : SUCCESS in 7m 33s\n- openstack-tox-py36 http://logs.openstack.org/67/659167/2/gate/openstack-tox-py36/1e1a4c6/ : SUCCESS in 5m 11s\n- openstack-tox-py37 http://logs.openstack.org/67/659167/2/gate/openstack-tox-py37/e8ab370/ : SUCCESS in 6m 10s\n- openstack-tox-docs http://logs.openstack.org/67/659167/2/gate/openstack-tox-docs/c740e69/html/ : SUCCESS in 4m 51s\n- requirements-check http://logs.openstack.org/67/659167/2/gate/requirements-check/95000ae/ : SUCCESS in 3m 03s\n- build-openstack-releasenotes http://logs.openstack.org/67/659167/2/gate/build-openstack-releasenotes/9f0ad65/html/ : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":2},{"id":"43f3fd80264873d6999b76f8c242d55ec392befc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:16:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dec3639078195e88bbc09168761af96a0f2166b4","revisions":{"fecb6a2eda4d8d72bfcab3cc0e93579bbcbec601":{"kind":"REWORK","_number":1,"created":"2019-05-14 21:14:56.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/67/659167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/67/659167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/659167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/659167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/659167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/67/659167/1"}}},"commit":{"parents":[{"commit":"c16478eb2ebc0fbd4b2db989d257c7a3d1eb3bbc","subject":"Blacklist Bandit 1.6.0 due to directory exclusion bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c16478eb2ebc0fbd4b2db989d257c7a3d1eb3bbc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-14 21:12:26.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-14 21:12:26.000000000","tz":-420},"subject":"Clean fix for bandit exclusions change","message":"Clean fix for bandit exclusions change\n\nDon\u0027t mangle requirements, just fix the glob.\n\nChange-Id: I4b57336a103c75df877ff65e645b3b05b3c79a49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fecb6a2eda4d8d72bfcab3cc0e93579bbcbec601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fecb6a2eda4d8d72bfcab3cc0e93579bbcbec601"}]},"branch":"refs/heads/master"},"dec3639078195e88bbc09168761af96a0f2166b4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-14 23:11:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/67/659167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/67/659167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/659167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/659167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/659167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/67/659167/2"}}},"commit":{"parents":[{"commit":"af51a84477586067fd194aeaaa9826e5b3822a5b","subject":"Cap to sphinx 2.0.0 on Python 2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af51a84477586067fd194aeaaa9826e5b3822a5b"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-14 21:12:26.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-14 23:11:43.000000000","tz":0},"subject":"Clean fix for bandit exclusions change","message":"Clean fix for bandit exclusions change\n\nDon\u0027t mangle requirements, just fix the glob.\n\nChange-Id: I4b57336a103c75df877ff65e645b3b05b3c79a49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dec3639078195e88bbc09168761af96a0f2166b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dec3639078195e88bbc09168761af96a0f2166b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
