)]}'
{"id":"openstack%2Fdesignate~659169","triplet_id":"openstack%2Fdesignate~stable%2Frocky~I4b57336a103c75df877ff65e645b3b05b3c79a49","project":"openstack/designate","branch":"stable/rocky","topic":"bandit_exclusion_designate","hashtags":[],"change_id":"I4b57336a103c75df877ff65e645b3b05b3c79a49","subject":"Clean fix for bandit exclusions change","status":"MERGED","created":"2019-05-14 21:15:55.000000000","updated":"2019-05-17 11:30:46.000000000","submitted":"2019-05-17 11:30:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659169-1558092646301-795c03e7","meta_rev_id":"44d481a3957acdea6a97a309b98f46550cfadb21","_number":659169,"virtual_id_number":659169,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-05-17 11:30:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-05-17 10:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-15 18:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-05-17 10:28:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 18:10:26.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-05-17 10:28:26.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-17 11:30:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e95eb253ebc9007de2bb00304912afd2bd68b5b3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 21:15:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"420df78691a808160f5b906ddae4150d3cb3bb08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 22:13:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/69/659169/1/check/designate-bind9/4e2278a/ : SUCCESS in 40m 00s\n- designate-bind9-manager-model http://logs.openstack.org/69/659169/1/check/designate-bind9-manager-model/5f82045/ : SUCCESS in 40m 42s\n- designate-bind9-py35 http://logs.openstack.org/69/659169/1/check/designate-bind9-py35/c8a7fbb/ : SUCCESS in 40m 19s\n- designate-bind9-py36 http://logs.openstack.org/69/659169/1/check/designate-bind9-py36/383ee13/ : SUCCESS in 37m 06s\n- designate-devstack-pdns4 http://logs.openstack.org/69/659169/1/check/designate-devstack-pdns4/cba5426/ : SUCCESS in 34m 05s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/69/659169/1/check/designate-devstack-pdns4-postgres/c378678/ : SUCCESS in 34m 49s\n- designate-grenade-pdns4 http://logs.openstack.org/69/659169/1/check/designate-grenade-pdns4/c6b6b39/ : SUCCESS in 54m 19s\n- openstack-tox-py36 http://logs.openstack.org/69/659169/1/check/openstack-tox-py36/ecf90b1/ : SUCCESS in 6m 13s\n- openstack-tox-pep8 http://logs.openstack.org/69/659169/1/check/openstack-tox-pep8/8df356f/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/69/659169/1/check/openstack-tox-py27/083dee7/ : SUCCESS in 5m 16s\n- openstack-tox-py35 http://logs.openstack.org/69/659169/1/check/openstack-tox-py35/70e8825/ : SUCCESS in 6m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/69/659169/1/check/build-openstack-sphinx-docs/2a64da2/html/ : SUCCESS in 5m 42s\n- build-openstack-releasenotes http://logs.openstack.org/69/659169/1/check/build-openstack-releasenotes/6b2b40b/html/ : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/659169/1/check/openstack-tox-lower-constraints/d6de262/ : SUCCESS in 5m 03s\n- openstack-tox-cover http://logs.openstack.org/69/659169/1/check/openstack-tox-cover/fb0bd2d/cover/ : SUCCESS in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc3954d80286c80c22b3f09435b7a9bf25f79785","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-14 23:57:46.000000000","message":"Topic set to bandit_exclusion_designate","accounts_in_message":[],"_revision_number":1},{"id":"5d292e4c1a2c78ec8e498ea862a66e54dd4b5299","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-15 18:10:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"824cbe6d6f8234d4fecb80b760690dc8ac0881b0","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-17 10:28:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3debedeca5fdb14868c358bc489269af8a05f273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 10:28:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68b5d697ad41136d6139a4ce0a9463f7b26677b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 11:30:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44d481a3957acdea6a97a309b98f46550cfadb21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 11:30:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/69/659169/1/gate/designate-bind9/9ea3937/ : SUCCESS in 40m 41s\n- designate-bind9-manager-model http://logs.openstack.org/69/659169/1/gate/designate-bind9-manager-model/8a81d2c/ : SUCCESS in 39m 38s\n- designate-bind9-py35 http://logs.openstack.org/69/659169/1/gate/designate-bind9-py35/12947d2/ : SUCCESS in 42m 52s\n- designate-bind9-py36 http://logs.openstack.org/69/659169/1/gate/designate-bind9-py36/a6f9f3a/ : SUCCESS in 38m 39s\n- designate-devstack-pdns4 http://logs.openstack.org/69/659169/1/gate/designate-devstack-pdns4/cd88399/ : SUCCESS in 34m 44s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/69/659169/1/gate/designate-devstack-pdns4-postgres/d5a46cb/ : SUCCESS in 34m 33s\n- designate-grenade-pdns4 http://logs.openstack.org/69/659169/1/gate/designate-grenade-pdns4/e594288/ : SUCCESS in 1h 01m 28s\n- openstack-tox-py36 http://logs.openstack.org/69/659169/1/gate/openstack-tox-py36/85323ea/ : SUCCESS in 5m 42s\n- openstack-tox-pep8 http://logs.openstack.org/69/659169/1/gate/openstack-tox-pep8/1e086ad/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/69/659169/1/gate/openstack-tox-py27/b37c7bf/ : SUCCESS in 6m 17s\n- openstack-tox-py35 http://logs.openstack.org/69/659169/1/gate/openstack-tox-py35/61cbfd4/ : SUCCESS in 5m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/69/659169/1/gate/build-openstack-sphinx-docs/0261012/html/ : SUCCESS in 5m 18s\n- build-openstack-releasenotes http://logs.openstack.org/69/659169/1/gate/build-openstack-releasenotes/0e24c03/html/ : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/659169/1/gate/openstack-tox-lower-constraints/a04a739/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f586e321928e44e888f240e64ff9e201d527f9b8","revisions":{"f586e321928e44e888f240e64ff9e201d527f9b8":{"kind":"REWORK","_number":1,"created":"2019-05-14 21:15:55.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/69/659169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/659169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/659169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/659169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/659169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/659169/1"}}},"commit":{"parents":[{"commit":"4c3d5505c8dd021cf2171c459b0ecb208b3cf51e","subject":"Merge \"Fix DBDeadLock error resulting into 500\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4c3d5505c8dd021cf2171c459b0ecb208b3cf51e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-14 21:12:26.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-14 21:15:55.000000000","tz":0},"subject":"Clean fix for bandit exclusions change","message":"Clean fix for bandit exclusions change\n\nDon\u0027t mangle requirements, just fix the glob.\n\nChange-Id: I4b57336a103c75df877ff65e645b3b05b3c79a49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f586e321928e44e888f240e64ff9e201d527f9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f586e321928e44e888f240e64ff9e201d527f9b8"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
