)]}'
{"id":"openstack%2Fdesignate~659950","triplet_id":"openstack%2Fdesignate~master~Ic6d2c7ab5283df1ad3953e86e65550f92dd0bc6b","project":"openstack/designate","branch":"master","topic":"improved_tests","hashtags":[],"change_id":"Ic6d2c7ab5283df1ad3953e86e65550f92dd0bc6b","subject":"Don\u0027t bind to tcp/udp when running unit-tests","status":"MERGED","created":"2019-05-19 09:25:12.000000000","updated":"2019-05-25 06:55:31.000000000","submitted":"2019-05-25 06:55:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659950-1558767331369-04c46964","meta_rev_id":"4fac6972a3d9f908c23ceccbd31bec5c6c95144f","_number":659950,"virtual_id_number":659950,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-05-25 06:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"recommended":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"all":[{"value":1,"date":"2019-05-20 06:37:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":2,"date":"2019-05-20 10:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-25 05:53:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 06:37:24.000000000","updated_by":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"reviewer":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"state":"REVIEWER"},{"updated":"2019-05-20 10:50:37.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-25 06:55:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88c65129b1a5d3d489ace84decd065b4526b4848","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-19 09:25:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87674ece0ac64775c19ddab4a05570346d5e491b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-19 09:38:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"71c4d172d83d8f622d5a79d34cbec3aedca3743f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 10:42:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/50/659950/2/check/designate-bind9/40204ec/ : SUCCESS in 36m 48s\n- designate-pdns4 http://logs.openstack.org/50/659950/2/check/designate-pdns4/249522d/ : SUCCESS in 35m 12s\n- designate-bind9-manager-model http://logs.openstack.org/50/659950/2/check/designate-bind9-manager-model/7495ee0/ : SUCCESS in 41m 18s\n- designate-bind9-py27 http://logs.openstack.org/50/659950/2/check/designate-bind9-py27/72c2670/ : SUCCESS in 39m 06s\n- designate-pdns4-py27 http://logs.openstack.org/50/659950/2/check/designate-pdns4-py27/0fd88b1/ : SUCCESS in 37m 47s\n- designate-pdns4-postgres http://logs.openstack.org/50/659950/2/check/designate-pdns4-postgres/05a341b/ : SUCCESS in 35m 55s\n- designate-grenade-pdns4 http://logs.openstack.org/50/659950/2/check/designate-grenade-pdns4/3bfe2e5/ : SUCCESS in 58m 10s\n- openstack-tox-cover http://logs.openstack.org/50/659950/2/check/openstack-tox-cover/8134e72/cover/ : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/659950/2/check/openstack-tox-lower-constraints/b429d6a/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/50/659950/2/check/openstack-tox-pep8/c02d685/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/50/659950/2/check/openstack-tox-py27/27a421f/ : SUCCESS in 7m 01s\n- openstack-tox-py36 http://logs.openstack.org/50/659950/2/check/openstack-tox-py36/35995ba/ : SUCCESS in 5m 12s\n- openstack-tox-py37 http://logs.openstack.org/50/659950/2/check/openstack-tox-py37/010a6cf/ : SUCCESS in 6m 45s\n- openstack-tox-docs http://logs.openstack.org/50/659950/2/check/openstack-tox-docs/76a6834/html/ : SUCCESS in 4m 26s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/50/659950/2/check/neutron-tempest-plugin-designate-scenario/fdb8c44/ : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7283ac47142898a75bb5d956230bdd5877860f20","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 06:31:42.000000000","message":"Topic changed from no_sockets_on_tests to improved_tests","accounts_in_message":[],"_revision_number":2},{"id":"69b034ab70b327340d5af5ff3d492ab6d6cfa7a5","author":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"date":"2019-05-20 06:37:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8b60ecd6c630bd05046f1764c1671c98401756f7","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-20 10:50:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fd8e6d3b86c42ebd19177b6e088c1e591b6492bd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-25 05:53:09.000000000","message":"Patch Set 2: Workflow+1\n\nWorking this as this is a low-impact patch. Following this up with another patch to re-add some of the test coverage.\nhttps://review.opendev.org/#/c/661428/","accounts_in_message":[],"_revision_number":2},{"id":"ea2db8cf728b54977964baaabcb508c3c123a5ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 05:53:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cfcb5ecf2a0e08efdaf5f371e8e355466110f4ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 06:55:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/50/659950/2/gate/designate-bind9/10c9025/ : SUCCESS in 39m 16s\n- designate-pdns4 http://logs.openstack.org/50/659950/2/gate/designate-pdns4/4221d70/ : SUCCESS in 36m 38s\n- designate-bind9-manager-model http://logs.openstack.org/50/659950/2/gate/designate-bind9-manager-model/4d5302a/ : SUCCESS in 36m 19s\n- designate-bind9-py27 http://logs.openstack.org/50/659950/2/gate/designate-bind9-py27/87e17c6/ : SUCCESS in 37m 24s\n- designate-pdns4-py27 http://logs.openstack.org/50/659950/2/gate/designate-pdns4-py27/6db905f/ : SUCCESS in 35m 55s\n- designate-pdns4-postgres http://logs.openstack.org/50/659950/2/gate/designate-pdns4-postgres/39ce426/ : SUCCESS in 35m 25s\n- designate-grenade-pdns4 http://logs.openstack.org/50/659950/2/gate/designate-grenade-pdns4/b471b1c/ : SUCCESS in 56m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/659950/2/gate/openstack-tox-lower-constraints/3980f87/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/50/659950/2/gate/openstack-tox-pep8/f63cd2c/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/50/659950/2/gate/openstack-tox-py27/430c0e0/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/50/659950/2/gate/openstack-tox-py36/6dcb34c/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/50/659950/2/gate/openstack-tox-py37/a57850c/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/50/659950/2/gate/openstack-tox-docs/4445b7a/html/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4fac6972a3d9f908c23ceccbd31bec5c6c95144f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 06:55:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b5b593acd82ca866d8df10a2dbed04c6a5be6289","revisions":{"62f134e5b818c880c25889e12d63de20a9e6e1e6":{"kind":"REWORK","_number":1,"created":"2019-05-19 09:25:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/659950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/659950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/659950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/659950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/659950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/659950/1"}}},"commit":{"parents":[{"commit":"b778e942e3eda5a2740f8fb5f60d52470a3cfa97","subject":"Merge \"Ensure pdns4 driver is consistent with other drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b778e942e3eda5a2740f8fb5f60d52470a3cfa97"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-19 09:23:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-19 09:23:09.000000000","tz":-420},"subject":"Reduce the number of sockets needed for unit tests","message":"Reduce the number of sockets needed for unit tests\n\nMost of the time we don\u0027t actually need to bind to\ntcp/udp sockets.\n\nChange-Id: Ic6d2c7ab5283df1ad3953e86e65550f92dd0bc6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62f134e5b818c880c25889e12d63de20a9e6e1e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62f134e5b818c880c25889e12d63de20a9e6e1e6"}]},"branch":"refs/heads/master"},"b5b593acd82ca866d8df10a2dbed04c6a5be6289":{"kind":"REWORK","_number":2,"created":"2019-05-19 09:38:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/659950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/659950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/659950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/659950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/659950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/659950/2"}}},"commit":{"parents":[{"commit":"b778e942e3eda5a2740f8fb5f60d52470a3cfa97","subject":"Merge \"Ensure pdns4 driver is consistent with other drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b778e942e3eda5a2740f8fb5f60d52470a3cfa97"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-19 09:23:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-19 09:36:07.000000000","tz":-420},"subject":"Don\u0027t bind to tcp/udp when running unit-tests","message":"Don\u0027t bind to tcp/udp when running unit-tests\n\nMost of the time we don\u0027t actually need to bind to\ntcp/udp sockets. We should only bind if we really\nhave to.\n\n* Removed unit-tests used to test bindingg to actual ports.\n* Don\u0027t bind to tcp/udp ports if we don\u0027t have to.\n* Added check to make sure client was set before closing.\n\nWe should look at adding functional tests to catch\nissues like described in bug 1566036.\n\nChange-Id: Ic6d2c7ab5283df1ad3953e86e65550f92dd0bc6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5b593acd82ca866d8df10a2dbed04c6a5be6289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5b593acd82ca866d8df10a2dbed04c6a5be6289"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
