)]}'
{"id":"openstack%2Fdesignate~660876","triplet_id":"openstack%2Fdesignate~master~I68d54277dfb7c1cbc14db4e1739137b9a6982b42","project":"openstack/designate","branch":"master","topic":"standardize_unit_tests","hashtags":[],"change_id":"I68d54277dfb7c1cbc14db4e1739137b9a6982b42","subject":"Modernized scheduler tests","status":"MERGED","created":"2019-05-23 04:41:05.000000000","updated":"2019-05-25 02:20:25.000000000","submitted":"2019-05-25 02:20:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660876-1558750825045-50d017d9","meta_rev_id":"255c08a8bd0e42248ea10a75ec9c98178374a838","_number":660876,"virtual_id_number":660876,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-05-25 02:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-05-24 10:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-25 01:24:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 10:54:29.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-05-25 02:20:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88b9c4e7dbe6b54124c6cad8ed40cfe6fc160eb6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-23 04:41:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ff3e4608ba81b725a7ab0b07c6d68570475aa56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 05:45:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/76/660876/1/check/designate-bind9/35ae01f/ : SUCCESS in 41m 33s\n- designate-pdns4 http://logs.openstack.org/76/660876/1/check/designate-pdns4/95a7ad4/ : SUCCESS in 36m 53s\n- designate-bind9-manager-model http://logs.openstack.org/76/660876/1/check/designate-bind9-manager-model/9581bb9/ : SUCCESS in 40m 40s\n- designate-bind9-py27 http://logs.openstack.org/76/660876/1/check/designate-bind9-py27/09af6a7/ : SUCCESS in 38m 54s\n- designate-pdns4-py27 http://logs.openstack.org/76/660876/1/check/designate-pdns4-py27/bc40c2f/ : SUCCESS in 38m 57s\n- designate-pdns4-postgres http://logs.openstack.org/76/660876/1/check/designate-pdns4-postgres/599f3b1/ : SUCCESS in 43m 01s\n- designate-grenade-pdns4 http://logs.openstack.org/76/660876/1/check/designate-grenade-pdns4/ebb306b/ : SUCCESS in 55m 58s\n- openstack-tox-cover http://logs.openstack.org/76/660876/1/check/openstack-tox-cover/78277e5/cover/ : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/660876/1/check/openstack-tox-lower-constraints/cccbd1d/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/76/660876/1/check/openstack-tox-pep8/eeb4045/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/76/660876/1/check/openstack-tox-py27/f343ebf/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/76/660876/1/check/openstack-tox-py36/423d999/ : SUCCESS in 5m 50s\n- openstack-tox-py37 http://logs.openstack.org/76/660876/1/check/openstack-tox-py37/b495f8d/ : SUCCESS in 6m 09s\n- openstack-tox-docs http://logs.openstack.org/76/660876/1/check/openstack-tox-docs/2321a7d/html/ : SUCCESS in 5m 06s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/660876/1/check/neutron-tempest-plugin-designate-scenario/ea8dae2/ : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5d9dac197b79da058362429f7e105b4b79ecfe76","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-23 15:25:30.000000000","message":"Topic changed from scheduler to standardize_unit_tests","accounts_in_message":[],"_revision_number":1},{"id":"cdd80167eb8180c771b1f8a637253d7eab9e608b","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-05-24 10:54:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f235759312cc098c862af9489110d68391b8680","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-25 01:24:17.000000000","message":"Patch Set 1: Workflow+1\n\nMerging this as it\u0027s low-risk.","accounts_in_message":[],"_revision_number":1},{"id":"c323c68722c326ae3c25a10ee8eb9ac0900433d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 01:24:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a96a190c6da374c7e786741ffaed7c8d716e234","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 02:20:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/76/660876/1/gate/designate-bind9/ecd6023/ : SUCCESS in 32m 56s\n- designate-pdns4 http://logs.openstack.org/76/660876/1/gate/designate-pdns4/ef5ea33/ : SUCCESS in 34m 41s\n- designate-bind9-manager-model http://logs.openstack.org/76/660876/1/gate/designate-bind9-manager-model/117637d/ : SUCCESS in 35m 19s\n- designate-bind9-py27 http://logs.openstack.org/76/660876/1/gate/designate-bind9-py27/764f2dc/ : SUCCESS in 37m 18s\n- designate-pdns4-py27 http://logs.openstack.org/76/660876/1/gate/designate-pdns4-py27/c1bab94/ : SUCCESS in 32m 54s\n- designate-pdns4-postgres http://logs.openstack.org/76/660876/1/gate/designate-pdns4-postgres/1c343f7/ : SUCCESS in 33m 24s\n- designate-grenade-pdns4 http://logs.openstack.org/76/660876/1/gate/designate-grenade-pdns4/f0b5d01/ : SUCCESS in 50m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/660876/1/gate/openstack-tox-lower-constraints/d743bcd/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/76/660876/1/gate/openstack-tox-pep8/e29ad31/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/76/660876/1/gate/openstack-tox-py27/95c7895/ : SUCCESS in 5m 29s\n- openstack-tox-py36 http://logs.openstack.org/76/660876/1/gate/openstack-tox-py36/3e25f71/ : SUCCESS in 5m 45s\n- openstack-tox-py37 http://logs.openstack.org/76/660876/1/gate/openstack-tox-py37/7d0d48a/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/76/660876/1/gate/openstack-tox-docs/b1759da/html/ : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"255c08a8bd0e42248ea10a75ec9c98178374a838","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 02:20:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94c38c023512bf6fc6dbc19923d6c9ac9ef1f71f","revisions":{"94c38c023512bf6fc6dbc19923d6c9ac9ef1f71f":{"kind":"REWORK","_number":1,"created":"2019-05-23 04:41:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/660876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/660876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/660876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/660876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/660876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/660876/1"}}},"commit":{"parents":[{"commit":"12fdd24e5bd997c7a444a1a343435fe28219de77","subject":"Merge \"Modernized object tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12fdd24e5bd997c7a444a1a343435fe28219de77"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-23 04:40:03.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-23 04:40:03.000000000","tz":-420},"subject":"Modernized scheduler tests","message":"Modernized scheduler tests\n\n* Renamed folder test_scheduler to scheduler.\n* Cleanedup tests and standardized them.\n\nChange-Id: I68d54277dfb7c1cbc14db4e1739137b9a6982b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/94c38c023512bf6fc6dbc19923d6c9ac9ef1f71f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/94c38c023512bf6fc6dbc19923d6c9ac9ef1f71f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
