)]}'
{"id":"openstack%2Fdesignate~661438","triplet_id":"openstack%2Fdesignate~master~Ieaad0d9b825b6f4f58b14aa6e4ad6c53f4ff30e2","project":"openstack/designate","branch":"master","topic":"standardize_unit_tests","hashtags":[],"change_id":"Ieaad0d9b825b6f4f58b14aa6e4ad6c53f4ff30e2","subject":"Modernized dnsutils tests","status":"MERGED","created":"2019-05-25 07:58:59.000000000","updated":"2019-06-05 03:28:00.000000000","submitted":"2019-06-05 03:28:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661438-1559705280208-0273d778","meta_rev_id":"7cda9d143fe2f89e02b0b1fb37d0a9e3e3982b82","_number":661438,"virtual_id_number":661438,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-06-05 03:27:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-04 17:12:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-04 21:30:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 17:12:16.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-05 03:27:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c0734fa552a2b35395aa4113d459e99ba9cd5ae","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-25 07:58:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66efe234864771deb69757ed15f4b068d1ec4994","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-25 08:00:51.000000000","message":"Topic changed from dnsutils to standardize_unit_tests","accounts_in_message":[],"_revision_number":1},{"id":"4d631fd5455a2d560c6a1579fde31e9d9d952a66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 08:53:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/38/661438/1/check/designate-bind9/c03dc8d/ : SUCCESS in 31m 02s\n- designate-pdns4 http://logs.openstack.org/38/661438/1/check/designate-pdns4/b9d31cc/ : SUCCESS in 30m 10s\n- designate-bind9-manager-model http://logs.openstack.org/38/661438/1/check/designate-bind9-manager-model/992cdc2/ : SUCCESS in 30m 30s\n- designate-bind9-py27 http://logs.openstack.org/38/661438/1/check/designate-bind9-py27/3e6abf1/ : SUCCESS in 35m 47s\n- designate-pdns4-py27 http://logs.openstack.org/38/661438/1/check/designate-pdns4-py27/f80f1e3/ : SUCCESS in 33m 32s\n- designate-pdns4-postgres http://logs.openstack.org/38/661438/1/check/designate-pdns4-postgres/607f69d/ : SUCCESS in 31m 06s\n- designate-grenade-pdns4 http://logs.openstack.org/38/661438/1/check/designate-grenade-pdns4/8ea2d82/ : SUCCESS in 52m 58s\n- openstack-tox-cover http://logs.openstack.org/38/661438/1/check/openstack-tox-cover/4518b58/cover/ : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661438/1/check/openstack-tox-lower-constraints/37c4e7c/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/38/661438/1/check/openstack-tox-pep8/3bfa4d0/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/38/661438/1/check/openstack-tox-py27/0806283/ : SUCCESS in 6m 21s\n- openstack-tox-py36 http://logs.openstack.org/38/661438/1/check/openstack-tox-py36/fde258a/ : SUCCESS in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/38/661438/1/check/openstack-tox-py37/77327cc/ : SUCCESS in 6m 44s\n- openstack-tox-docs http://logs.openstack.org/38/661438/1/check/openstack-tox-docs/5279c73/html/ : SUCCESS in 4m 50s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/38/661438/1/check/neutron-tempest-plugin-designate-scenario/2a012b5/ : SUCCESS in 30m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ad10991859079bf14a6f7d28f5ba45840ff4c41e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-28 22:27:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"407913893ffdfcfcfd508ac7e73ed4d3addbc385","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 23:33:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/38/661438/2/check/designate-bind9/8c755e8/ : SUCCESS in 38m 51s\n- designate-pdns4 http://logs.openstack.org/38/661438/2/check/designate-pdns4/8a3c338/ : SUCCESS in 35m 43s\n- designate-bind9-manager-model http://logs.openstack.org/38/661438/2/check/designate-bind9-manager-model/fedd923/ : SUCCESS in 38m 55s\n- designate-bind9-py27 http://logs.openstack.org/38/661438/2/check/designate-bind9-py27/f4f2a2f/ : SUCCESS in 39m 59s\n- designate-pdns4-py27 http://logs.openstack.org/38/661438/2/check/designate-pdns4-py27/50f91b7/ : SUCCESS in 35m 44s\n- designate-pdns4-postgres http://logs.openstack.org/38/661438/2/check/designate-pdns4-postgres/c6a7c9d/ : SUCCESS in 32m 51s\n- designate-grenade-pdns4 http://logs.openstack.org/38/661438/2/check/designate-grenade-pdns4/13ae18d/ : SUCCESS in 1h 00m 38s\n- openstack-tox-cover http://logs.openstack.org/38/661438/2/check/openstack-tox-cover/dcec330/cover/ : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661438/2/check/openstack-tox-lower-constraints/6915fa6/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/38/661438/2/check/openstack-tox-pep8/2b2b12f/ : SUCCESS in 6m 53s\n- openstack-tox-py27 http://logs.openstack.org/38/661438/2/check/openstack-tox-py27/0d423cf/ : SUCCESS in 5m 57s\n- openstack-tox-py36 http://logs.openstack.org/38/661438/2/check/openstack-tox-py36/fc66531/ : SUCCESS in 8m 00s\n- openstack-tox-py37 http://logs.openstack.org/38/661438/2/check/openstack-tox-py37/7fcff4b/ : SUCCESS in 7m 32s\n- openstack-tox-docs http://logs.openstack.org/38/661438/2/check/openstack-tox-docs/a01a8ea/html/ : SUCCESS in 5m 30s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/38/661438/2/check/neutron-tempest-plugin-designate-scenario/40d12e4/ : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":2},{"id":"28398673689636ff2e2b2bfa42d11820f252a33f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-29 07:42:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8bd43e70faf227555c5a7b86627c78a2b19feace","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-29 07:57:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a5edd857398f16dab734f596962d8dac03b1013d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 08:56:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/38/661438/4/check/designate-bind9/0851dbc/ : SUCCESS in 36m 54s\n- designate-pdns4 http://logs.openstack.org/38/661438/4/check/designate-pdns4/c35616a/ : SUCCESS in 39m 33s\n- designate-bind9-manager-model http://logs.openstack.org/38/661438/4/check/designate-bind9-manager-model/d24c11b/ : SUCCESS in 41m 53s\n- designate-bind9-py27 http://logs.openstack.org/38/661438/4/check/designate-bind9-py27/7e17c08/ : SUCCESS in 56m 10s\n- designate-pdns4-py27 http://logs.openstack.org/38/661438/4/check/designate-pdns4-py27/b6f0120/ : SUCCESS in 56m 53s\n- designate-pdns4-postgres http://logs.openstack.org/38/661438/4/check/designate-pdns4-postgres/84e7acd/ : SUCCESS in 37m 36s\n- designate-grenade-pdns4 http://logs.openstack.org/38/661438/4/check/designate-grenade-pdns4/8822e65/ : SUCCESS in 50m 26s\n- openstack-tox-cover http://logs.openstack.org/38/661438/4/check/openstack-tox-cover/6dcd014/cover/ : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661438/4/check/openstack-tox-lower-constraints/ec0ab3e/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/38/661438/4/check/openstack-tox-pep8/0c5f6e8/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/38/661438/4/check/openstack-tox-py27/5a49400/ : SUCCESS in 5m 43s\n- openstack-tox-py36 http://logs.openstack.org/38/661438/4/check/openstack-tox-py36/5dd2e47/ : SUCCESS in 6m 00s\n- openstack-tox-py37 http://logs.openstack.org/38/661438/4/check/openstack-tox-py37/3df90ed/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/38/661438/4/check/openstack-tox-docs/573e8d9/html/ : SUCCESS in 5m 11s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/38/661438/4/check/neutron-tempest-plugin-designate-scenario/c2579d2/ : SUCCESS in 34m 56s","accounts_in_message":[],"_revision_number":4},{"id":"6e3e7b9bfea83e668c806c9a942b1f5528b3ab20","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-03 21:50:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a8199d17e0b5db9b7e923d90ab6cff4de5486725","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 23:14:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/38/661438/5/check/designate-bind9/49f22ea/ : SUCCESS in 37m 21s\n- designate-pdns4 http://logs.openstack.org/38/661438/5/check/designate-pdns4/c9d8a2c/ : SUCCESS in 33m 42s\n- designate-bind9-manager-model http://logs.openstack.org/38/661438/5/check/designate-bind9-manager-model/71a8c7b/ : SUCCESS in 38m 58s\n- designate-bind9-py27 http://logs.openstack.org/38/661438/5/check/designate-bind9-py27/817fb58/ : SUCCESS in 41m 14s\n- designate-pdns4-py27 http://logs.openstack.org/38/661438/5/check/designate-pdns4-py27/aed9b70/ : SUCCESS in 42m 06s\n- designate-pdns4-postgres http://logs.openstack.org/38/661438/5/check/designate-pdns4-postgres/5d565b2/ : SUCCESS in 41m 53s\n- designate-grenade-pdns4 http://logs.openstack.org/38/661438/5/check/designate-grenade-pdns4/9cb560d/ : SUCCESS in 1h 10m 22s\n- openstack-tox-cover http://logs.openstack.org/38/661438/5/check/openstack-tox-cover/6b10023/cover/ : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661438/5/check/openstack-tox-lower-constraints/1525b63/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/38/661438/5/check/openstack-tox-pep8/b9350a9/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/38/661438/5/check/openstack-tox-py27/2c39001/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/38/661438/5/check/openstack-tox-py36/ebfcc47/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/38/661438/5/check/openstack-tox-py37/07e4482/ : SUCCESS in 5m 53s\n- openstack-tox-docs http://logs.openstack.org/38/661438/5/check/openstack-tox-docs/0c9ab03/html/ : SUCCESS in 4m 39s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/38/661438/5/check/neutron-tempest-plugin-designate-scenario/8401d97/ : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":5},{"id":"ad1518c8642892f6d80c3825f06c2a8d7bff94aa","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-04 17:12:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a2198bb25460f0130251b940069fff2685a2ce2a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-04 21:30:55.000000000","message":"Patch Set 5: Workflow+1\n\nMerging with only one Core reviewers +2 as the change is minor, and adds additional test coverage.","accounts_in_message":[],"_revision_number":5},{"id":"23c51ab08399f45f4f6e7f992be1685ad54156a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 21:31:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7c7cc2ab335d93ceaa81e18742c67c6277e6da7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 22:43:46.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/38/661438/5/gate/designate-bind9/b5f508a/ : SUCCESS in 41m 16s\n- designate-pdns4 http://logs.openstack.org/38/661438/5/gate/designate-pdns4/071acc2/ : FAILURE in 18m 57s\n- designate-bind9-manager-model http://logs.openstack.org/38/661438/5/gate/designate-bind9-manager-model/6bf4283/ : SUCCESS in 41m 52s\n- designate-bind9-py27 http://logs.openstack.org/38/661438/5/gate/designate-bind9-py27/60036a3/ : SUCCESS in 1h 08m 44s\n- designate-pdns4-py27 http://logs.openstack.org/38/661438/5/gate/designate-pdns4-py27/f21d2bd/ : SUCCESS in 33m 38s\n- designate-pdns4-postgres http://logs.openstack.org/38/661438/5/gate/designate-pdns4-postgres/505231a/ : SUCCESS in 36m 23s\n- designate-grenade-pdns4 http://logs.openstack.org/38/661438/5/gate/designate-grenade-pdns4/330cdab/ : FAILURE in 47m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661438/5/gate/openstack-tox-lower-constraints/f2c7b35/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/38/661438/5/gate/openstack-tox-pep8/657913d/ : SUCCESS in 8m 20s\n- openstack-tox-py27 http://logs.openstack.org/38/661438/5/gate/openstack-tox-py27/6d69938/ : SUCCESS in 7m 25s\n- openstack-tox-py36 http://logs.openstack.org/38/661438/5/gate/openstack-tox-py36/8964909/ : SUCCESS in 7m 04s\n- openstack-tox-py37 http://logs.openstack.org/38/661438/5/gate/openstack-tox-py37/d25ed0b/ : SUCCESS in 5m 59s\n- openstack-tox-docs http://logs.openstack.org/38/661438/5/gate/openstack-tox-docs/223828d/html/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":5},{"id":"a33705e3da205bb1b446e7e822cc8204e6a0da84","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-04 23:14:04.000000000","message":"Patch Set 5:\n\nrecheck mirror timeout","accounts_in_message":[],"_revision_number":5},{"id":"4c3dab8a000594adb158c603b2ff44bf9c6e6582","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 00:23:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/38/661438/5/check/designate-bind9/648b8a3/ : SUCCESS in 37m 40s\n- designate-pdns4 http://logs.openstack.org/38/661438/5/check/designate-pdns4/ffb51ef/ : SUCCESS in 39m 34s\n- designate-bind9-manager-model http://logs.openstack.org/38/661438/5/check/designate-bind9-manager-model/dca425f/ : SUCCESS in 41m 38s\n- designate-bind9-py27 http://logs.openstack.org/38/661438/5/check/designate-bind9-py27/e6b800e/ : RETRY_LIMIT in 3m 25s\n- designate-pdns4-py27 http://logs.openstack.org/38/661438/5/check/designate-pdns4-py27/d37dfe6/ : SUCCESS in 38m 29s\n- designate-pdns4-postgres http://logs.openstack.org/38/661438/5/check/designate-pdns4-postgres/8ceac81/ : SUCCESS in 37m 13s\n- designate-grenade-pdns4 http://logs.openstack.org/38/661438/5/check/designate-grenade-pdns4/504390d/ : SUCCESS in 56m 13s\n- openstack-tox-cover http://logs.openstack.org/38/661438/5/check/openstack-tox-cover/2893be6/cover/ : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661438/5/check/openstack-tox-lower-constraints/6d255d1/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/38/661438/5/check/openstack-tox-pep8/bddc0b8/ : SUCCESS in 5m 41s\n- openstack-tox-py27 http://logs.openstack.org/38/661438/5/check/openstack-tox-py27/408900f/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/38/661438/5/check/openstack-tox-py36/2e1db07/ : SUCCESS in 6m 02s\n- openstack-tox-py37 http://logs.openstack.org/38/661438/5/check/openstack-tox-py37/4d9e6fc/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/38/661438/5/check/openstack-tox-docs/1d1859d/html/ : SUCCESS in 5m 07s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/38/661438/5/check/neutron-tempest-plugin-designate-scenario/fc61231/ : SUCCESS in 34m 05s","accounts_in_message":[],"_revision_number":5},{"id":"977eadaeb5448afb255c59c7e795821b6109df45","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-05 01:02:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2dcb12754ea53abc35c820d3d4f9375722058427","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:23:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/38/661438/5/check/designate-bind9/b902a0c/ : SUCCESS in 36m 26s\n- designate-pdns4 http://logs.openstack.org/38/661438/5/check/designate-pdns4/fd53a14/ : SUCCESS in 35m 22s\n- designate-bind9-manager-model http://logs.openstack.org/38/661438/5/check/designate-bind9-manager-model/3d9ce4a/ : SUCCESS in 39m 57s\n- designate-bind9-py27 http://logs.openstack.org/38/661438/5/check/designate-bind9-py27/937dc5b/ : SUCCESS in 38m 57s\n- designate-pdns4-py27 http://logs.openstack.org/38/661438/5/check/designate-pdns4-py27/810733c/ : SUCCESS in 34m 56s\n- designate-pdns4-postgres http://logs.openstack.org/38/661438/5/check/designate-pdns4-postgres/33d0f5a/ : SUCCESS in 39m 42s\n- designate-grenade-pdns4 http://logs.openstack.org/38/661438/5/check/designate-grenade-pdns4/c6b7146/ : SUCCESS in 1h 00m 36s\n- openstack-tox-cover http://logs.openstack.org/38/661438/5/check/openstack-tox-cover/4e022e8/cover/ : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661438/5/check/openstack-tox-lower-constraints/ea1f287/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/38/661438/5/check/openstack-tox-pep8/832d763/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/38/661438/5/check/openstack-tox-py27/f203db5/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/38/661438/5/check/openstack-tox-py36/f75a582/ : SUCCESS in 5m 53s\n- openstack-tox-py37 http://logs.openstack.org/38/661438/5/check/openstack-tox-py37/a342278/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/38/661438/5/check/openstack-tox-docs/b78908d/html/ : SUCCESS in 4m 32s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/38/661438/5/check/neutron-tempest-plugin-designate-scenario/7d93762/ : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":5},{"id":"753a0be6e75035a9724a157ab77b4451f4fa3ece","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:23:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b2912df224e092edb5ba0be953e26e83bde9b201","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 03:27:59.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/38/661438/5/gate/designate-bind9/91fd44d/ : SUCCESS in 41m 31s\n- designate-pdns4 http://logs.openstack.org/38/661438/5/gate/designate-pdns4/ef58b77/ : SUCCESS in 39m 00s\n- designate-bind9-manager-model http://logs.openstack.org/38/661438/5/gate/designate-bind9-manager-model/fc81b56/ : SUCCESS in 42m 48s\n- designate-bind9-py27 http://logs.openstack.org/38/661438/5/gate/designate-bind9-py27/61780f9/ : SUCCESS in 46m 35s\n- designate-pdns4-py27 http://logs.openstack.org/38/661438/5/gate/designate-pdns4-py27/ea3f0cd/ : SUCCESS in 39m 44s\n- designate-pdns4-postgres http://logs.openstack.org/38/661438/5/gate/designate-pdns4-postgres/1eec76f/ : SUCCESS in 40m 10s\n- designate-grenade-pdns4 http://logs.openstack.org/38/661438/5/gate/designate-grenade-pdns4/9efb821/ : SUCCESS in 57m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661438/5/gate/openstack-tox-lower-constraints/a565ffa/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/38/661438/5/gate/openstack-tox-pep8/b279722/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/38/661438/5/gate/openstack-tox-py27/4354831/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/38/661438/5/gate/openstack-tox-py36/845259f/ : SUCCESS in 6m 32s\n- openstack-tox-py37 http://logs.openstack.org/38/661438/5/gate/openstack-tox-py37/fa143b7/ : SUCCESS in 6m 56s\n- openstack-tox-docs http://logs.openstack.org/38/661438/5/gate/openstack-tox-docs/80bc253/html/ : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":5},{"id":"7cda9d143fe2f89e02b0b1fb37d0a9e3e3982b82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 03:28:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"926e4ed0be94af9f6bbe851a1771f13290956ca9","revisions":{"7dceaec7bc681a21568cbc7b2b86d08a69ec5666":{"kind":"REWORK","_number":1,"created":"2019-05-25 07:58:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/661438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/661438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/661438/1"}}},"commit":{"parents":[{"commit":"7d2527014f664b37fab177c7b5dcfaf876e85519","subject":"Merge \"Added basic synchronized_zone tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d2527014f664b37fab177c7b5dcfaf876e85519"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 07:57:44.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 07:58:11.000000000","tz":-420},"subject":"Modernized dnsutils tests","message":"Modernized dnsutils tests\n\n* Moved dnsutils tests to the unit tests folder.\n* Cleaned up and added new dnsutils tests.\n\nChange-Id: Ieaad0d9b825b6f4f58b14aa6e4ad6c53f4ff30e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7dceaec7bc681a21568cbc7b2b86d08a69ec5666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7dceaec7bc681a21568cbc7b2b86d08a69ec5666"}]},"branch":"refs/heads/master"},"54c42ab19f2ab3529d23b576e07691151374d17c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-28 22:27:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/661438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/661438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/661438/2"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 07:57:44.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-28 22:27:01.000000000","tz":0},"subject":"Modernized dnsutils tests","message":"Modernized dnsutils tests\n\n* Moved dnsutils tests to the unit tests folder.\n* Cleaned up and added new dnsutils tests.\n\nChange-Id: Ieaad0d9b825b6f4f58b14aa6e4ad6c53f4ff30e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54c42ab19f2ab3529d23b576e07691151374d17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54c42ab19f2ab3529d23b576e07691151374d17c"}]},"branch":"refs/heads/master"},"ef759b14dca48c6434385ccc12070cb8ab9a7d84":{"kind":"REWORK","_number":3,"created":"2019-05-29 07:42:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/661438/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/661438/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/661438/3"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 07:57:44.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-29 07:42:05.000000000","tz":-420},"subject":"Modernized dnsutils tests","message":"Modernized dnsutils tests\n\n* Moved dnsutils tests to the unit tests folder.\n* Cleaned up and added new dnsutils tests.\n\nChange-Id: Ieaad0d9b825b6f4f58b14aa6e4ad6c53f4ff30e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef759b14dca48c6434385ccc12070cb8ab9a7d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef759b14dca48c6434385ccc12070cb8ab9a7d84"}]},"branch":"refs/heads/master"},"310c6a665f093e7e9eae8fb3b8c1b315dc920b8c":{"kind":"REWORK","_number":4,"created":"2019-05-29 07:57:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/661438/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/661438/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/661438/4"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 07:57:44.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-29 07:57:48.000000000","tz":-420},"subject":"Modernized dnsutils tests","message":"Modernized dnsutils tests\n\n* Moved dnsutils tests to the unit tests folder.\n* Cleaned up and added new dnsutils tests.\n\nChange-Id: Ieaad0d9b825b6f4f58b14aa6e4ad6c53f4ff30e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/310c6a665f093e7e9eae8fb3b8c1b315dc920b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/310c6a665f093e7e9eae8fb3b8c1b315dc920b8c"}]},"branch":"refs/heads/master"},"926e4ed0be94af9f6bbe851a1771f13290956ca9":{"kind":"REWORK","_number":5,"created":"2019-06-03 21:50:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/661438/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/661438/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/661438/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/661438/5"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 07:57:44.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-03 21:46:39.000000000","tz":-420},"subject":"Modernized dnsutils tests","message":"Modernized dnsutils tests\n\n* Moved dnsutils tests to the unit tests folder.\n* Cleaned up and added new dnsutils tests.\n\nChange-Id: Ieaad0d9b825b6f4f58b14aa6e4ad6c53f4ff30e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/926e4ed0be94af9f6bbe851a1771f13290956ca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/926e4ed0be94af9f6bbe851a1771f13290956ca9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
