)]}'
{"id":"openstack%2Fdesignate~661469","triplet_id":"openstack%2Fdesignate~master~I4738793e1f342f992e71f829e886398e1f7b9e67","project":"openstack/designate","branch":"master","topic":"standardize_unit_tests","hashtags":[],"change_id":"I4738793e1f342f992e71f829e886398e1f7b9e67","subject":"Removing unused code and updated unit tests readme","status":"MERGED","created":"2019-05-25 20:37:47.000000000","updated":"2019-05-28 22:03:41.000000000","submitted":"2019-05-28 22:03:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661469-1559081021348-8cbf0c1c","meta_rev_id":"4b6130c804b45bf1ee27ee46efa3ecea687c5bdf","_number":661469,"virtual_id_number":661469,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-05-28 22:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2019-05-27 14:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-28 21:03:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-27 14:27:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-05-28 22:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7440f5ebfd2f70d383b78bf14acb2356407d35a5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-25 20:37:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c40cb0aaf4a3c9650bb4218f15e6756336a61d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 21:38:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/69/661469/1/check/designate-bind9/5cb42b8/ : SUCCESS in 37m 43s\n- designate-pdns4 http://logs.openstack.org/69/661469/1/check/designate-pdns4/c12a3a4/ : SUCCESS in 38m 13s\n- designate-bind9-manager-model http://logs.openstack.org/69/661469/1/check/designate-bind9-manager-model/6770f71/ : SUCCESS in 37m 31s\n- designate-bind9-py27 http://logs.openstack.org/69/661469/1/check/designate-bind9-py27/71192fa/ : SUCCESS in 37m 36s\n- designate-pdns4-py27 http://logs.openstack.org/69/661469/1/check/designate-pdns4-py27/93644c6/ : SUCCESS in 35m 18s\n- designate-pdns4-postgres http://logs.openstack.org/69/661469/1/check/designate-pdns4-postgres/2db9f25/ : SUCCESS in 32m 19s\n- designate-grenade-pdns4 http://logs.openstack.org/69/661469/1/check/designate-grenade-pdns4/90b3d9a/ : SUCCESS in 54m 45s\n- openstack-tox-cover http://logs.openstack.org/69/661469/1/check/openstack-tox-cover/8dc1608/cover/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/661469/1/check/openstack-tox-lower-constraints/87b4899/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/69/661469/1/check/openstack-tox-pep8/f1e1bdd/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/69/661469/1/check/openstack-tox-py27/c8bf4f9/ : SUCCESS in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/69/661469/1/check/openstack-tox-py36/674a0cc/ : SUCCESS in 6m 48s\n- openstack-tox-py37 http://logs.openstack.org/69/661469/1/check/openstack-tox-py37/1799ffb/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/69/661469/1/check/openstack-tox-docs/9448e89/html/ : SUCCESS in 4m 50s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/69/661469/1/check/neutron-tempest-plugin-designate-scenario/7cb47b9/ : SUCCESS in 35m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6dc6299ec75388dac77824b19571aa515b8623c8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-27 03:42:16.000000000","message":"Topic set to standardize_unit_tests","accounts_in_message":[],"_revision_number":1},{"id":"ca0ebe91241c705f7c2dde8c2e24e0c16de8c150","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-05-27 14:27:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57550e078f92d1473ebd20db8d648936415fbad9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-28 21:03:05.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2505b7a2cc3c25b04c161d4c750fdbae168526f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 21:03:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be39166b53d0e06ac21e0e0054b319cbaa78faac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 22:03:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/69/661469/1/gate/designate-bind9/e03860b/ : SUCCESS in 38m 14s\n- designate-pdns4 http://logs.openstack.org/69/661469/1/gate/designate-pdns4/d486077/ : SUCCESS in 37m 57s\n- designate-bind9-manager-model http://logs.openstack.org/69/661469/1/gate/designate-bind9-manager-model/3f29fb9/ : SUCCESS in 42m 57s\n- designate-bind9-py27 http://logs.openstack.org/69/661469/1/gate/designate-bind9-py27/507881d/ : SUCCESS in 36m 04s\n- designate-pdns4-py27 http://logs.openstack.org/69/661469/1/gate/designate-pdns4-py27/73e4f01/ : SUCCESS in 36m 41s\n- designate-pdns4-postgres http://logs.openstack.org/69/661469/1/gate/designate-pdns4-postgres/5b18e39/ : SUCCESS in 41m 41s\n- designate-grenade-pdns4 http://logs.openstack.org/69/661469/1/gate/designate-grenade-pdns4/d9b05d1/ : SUCCESS in 53m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/69/661469/1/gate/openstack-tox-lower-constraints/9a7fe48/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/69/661469/1/gate/openstack-tox-pep8/da39c63/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/69/661469/1/gate/openstack-tox-py27/0eab266/ : SUCCESS in 6m 56s\n- openstack-tox-py36 http://logs.openstack.org/69/661469/1/gate/openstack-tox-py36/8453d84/ : SUCCESS in 5m 49s\n- openstack-tox-py37 http://logs.openstack.org/69/661469/1/gate/openstack-tox-py37/fa46080/ : SUCCESS in 5m 58s\n- openstack-tox-docs http://logs.openstack.org/69/661469/1/gate/openstack-tox-docs/74dcc60/html/ : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4b6130c804b45bf1ee27ee46efa3ecea687c5bdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 22:03:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"201d8be1450b20ade0a5edd4a6465d2b147cb09d","revisions":{"201d8be1450b20ade0a5edd4a6465d2b147cb09d":{"kind":"REWORK","_number":1,"created":"2019-05-25 20:37:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/69/661469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/661469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/661469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/661469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/661469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/661469/1"}}},"commit":{"parents":[{"commit":"4e49b5581fca2bda0d1638efafa88efa8925f05e","subject":"Merge \"Modernized backend tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e49b5581fca2bda0d1638efafa88efa8925f05e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 20:36:04.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 20:36:04.000000000","tz":-420},"subject":"Removing unused code and updated unit tests readme","message":"Removing unused code and updated unit tests readme\n\n* Removed unused ipa test.\n* Removed redundant resource test.\n* Updated README.\n\nChange-Id: I4738793e1f342f992e71f829e886398e1f7b9e67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/201d8be1450b20ade0a5edd4a6465d2b147cb09d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/201d8be1450b20ade0a5edd4a6465d2b147cb09d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
