)]}'
{"id":"openstack%2Fdesignate~661470","triplet_id":"openstack%2Fdesignate~master~I048980a74129c603af779ca3f0f9a5ddf750f118","project":"openstack/designate","branch":"master","topic":"standardize_unit_tests","hashtags":[],"change_id":"I048980a74129c603af779ca3f0f9a5ddf750f118","subject":"Modernized agent tests","status":"MERGED","created":"2019-05-25 22:02:16.000000000","updated":"2019-06-11 14:07:31.000000000","submitted":"2019-06-11 14:07:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661470-1560262051315-8840775f","meta_rev_id":"6113aef00f1bf1e48b40f20754f7a70216af673d","_number":661470,"virtual_id_number":661470,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-06-11 14:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-10 00:12:57.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-11 12:50:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-06-11 12:50:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 12:50:36.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-11 14:07:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b5bd740e50738be7485bfe87194821d36990a84","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-25 22:02:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"918053506381325d755787d82fed38e748d618fa","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-25 22:02:49.000000000","message":"Topic set to standardize_unit_tests","accounts_in_message":[],"_revision_number":1},{"id":"1da3634605cdc596eee0a7083affafbca4b581eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 23:00:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/70/661470/1/check/designate-bind9/a47f876/ : SUCCESS in 34m 23s\n- designate-pdns4 http://logs.openstack.org/70/661470/1/check/designate-pdns4/1f15813/ : SUCCESS in 34m 08s\n- designate-bind9-manager-model http://logs.openstack.org/70/661470/1/check/designate-bind9-manager-model/900f861/ : SUCCESS in 36m 22s\n- designate-bind9-py27 http://logs.openstack.org/70/661470/1/check/designate-bind9-py27/a0b7fe8/ : SUCCESS in 35m 01s\n- designate-pdns4-py27 http://logs.openstack.org/70/661470/1/check/designate-pdns4-py27/6093ddf/ : SUCCESS in 30m 13s\n- designate-pdns4-postgres http://logs.openstack.org/70/661470/1/check/designate-pdns4-postgres/fc4a8c2/ : SUCCESS in 34m 48s\n- designate-grenade-pdns4 http://logs.openstack.org/70/661470/1/check/designate-grenade-pdns4/a180633/ : SUCCESS in 54m 28s\n- openstack-tox-cover http://logs.openstack.org/70/661470/1/check/openstack-tox-cover/9ca6cba/cover/ : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/661470/1/check/openstack-tox-lower-constraints/c08dcc3/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/70/661470/1/check/openstack-tox-pep8/d400b46/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/70/661470/1/check/openstack-tox-py27/ea0801b/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/70/661470/1/check/openstack-tox-py36/328bdd0/ : SUCCESS in 5m 17s\n- openstack-tox-py37 http://logs.openstack.org/70/661470/1/check/openstack-tox-py37/02961f6/ : SUCCESS in 5m 58s\n- openstack-tox-docs http://logs.openstack.org/70/661470/1/check/openstack-tox-docs/a88dfa2/html/ : SUCCESS in 4m 57s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/70/661470/1/check/neutron-tempest-plugin-designate-scenario/e464181/ : SUCCESS in 30m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f678497bca5c710861fd962fc3bcdc3c3abddcde","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-25 23:30:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a61e53750a8c31419fe113110809d0d56b52e7cd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-26 00:01:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f0c2e3939dd32882fef71d25506b958288a0b712","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-26 00:17:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"120846a4728cb4982bc01977afb2d79363eb3d40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-26 01:11:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/70/661470/4/check/designate-bind9/4b8aee0/ : SUCCESS in 37m 06s\n- designate-pdns4 http://logs.openstack.org/70/661470/4/check/designate-pdns4/53de5b0/ : SUCCESS in 27m 14s\n- designate-bind9-manager-model http://logs.openstack.org/70/661470/4/check/designate-bind9-manager-model/715bc15/ : SUCCESS in 38m 36s\n- designate-bind9-py27 http://logs.openstack.org/70/661470/4/check/designate-bind9-py27/b0851f9/ : SUCCESS in 40m 26s\n- designate-pdns4-py27 http://logs.openstack.org/70/661470/4/check/designate-pdns4-py27/64845a6/ : SUCCESS in 34m 57s\n- designate-pdns4-postgres http://logs.openstack.org/70/661470/4/check/designate-pdns4-postgres/1e90fae/ : SUCCESS in 48m 19s\n- designate-grenade-pdns4 http://logs.openstack.org/70/661470/4/check/designate-grenade-pdns4/10046df/ : SUCCESS in 51m 59s\n- openstack-tox-cover http://logs.openstack.org/70/661470/4/check/openstack-tox-cover/78508bd/cover/ : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/661470/4/check/openstack-tox-lower-constraints/01505c4/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/70/661470/4/check/openstack-tox-pep8/e0d222c/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/70/661470/4/check/openstack-tox-py27/3240b8f/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/70/661470/4/check/openstack-tox-py36/71237a6/ : SUCCESS in 5m 31s\n- openstack-tox-py37 http://logs.openstack.org/70/661470/4/check/openstack-tox-py37/b0dc7fc/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/70/661470/4/check/openstack-tox-docs/8c1bde9/html/ : SUCCESS in 4m 29s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/70/661470/4/check/neutron-tempest-plugin-designate-scenario/a167fd9/ : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":4},{"id":"5e961c61efb5676a94d69e5c74d3bfb097a6fcfa","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-28 22:26:11.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b1b4d20f9752fcc5440f77f15311bd7a657b733f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 23:43:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/70/661470/5/check/designate-bind9/421a5d5/ : SUCCESS in 39m 44s\n- designate-pdns4 http://logs.openstack.org/70/661470/5/check/designate-pdns4/9579180/ : SUCCESS in 38m 45s\n- designate-bind9-manager-model http://logs.openstack.org/70/661470/5/check/designate-bind9-manager-model/13d957b/ : SUCCESS in 38m 40s\n- designate-bind9-py27 http://logs.openstack.org/70/661470/5/check/designate-bind9-py27/7244796/ : SUCCESS in 54m 52s\n- designate-pdns4-py27 http://logs.openstack.org/70/661470/5/check/designate-pdns4-py27/42b58cd/ : SUCCESS in 36m 06s\n- designate-pdns4-postgres http://logs.openstack.org/70/661470/5/check/designate-pdns4-postgres/3fbdf87/ : SUCCESS in 49m 53s\n- designate-grenade-pdns4 http://logs.openstack.org/70/661470/5/check/designate-grenade-pdns4/9613265/ : SUCCESS in 1h 16m 13s\n- openstack-tox-cover http://logs.openstack.org/70/661470/5/check/openstack-tox-cover/3fa3d03/cover/ : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/661470/5/check/openstack-tox-lower-constraints/2624aa6/ : SUCCESS in 7m 31s\n- openstack-tox-pep8 http://logs.openstack.org/70/661470/5/check/openstack-tox-pep8/f1e3dc4/ : SUCCESS in 7m 35s\n- openstack-tox-py27 http://logs.openstack.org/70/661470/5/check/openstack-tox-py27/0cfec88/ : SUCCESS in 5m 32s\n- openstack-tox-py36 http://logs.openstack.org/70/661470/5/check/openstack-tox-py36/6bf18e2/ : SUCCESS in 11m 30s\n- openstack-tox-py37 http://logs.openstack.org/70/661470/5/check/openstack-tox-py37/442454c/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/70/661470/5/check/openstack-tox-docs/e735e48/html/ : SUCCESS in 4m 39s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/70/661470/5/check/neutron-tempest-plugin-designate-scenario/1ddd583/ : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":5},{"id":"f776111539d1840c562e58b5ebfc882105a52133","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-29 07:38:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"267a0122a4aac21cd77d70943e1a4e5a43cbe5e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 08:43:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/70/661470/6/check/designate-bind9/7290fbe/ : SUCCESS in 37m 57s\n- designate-pdns4 http://logs.openstack.org/70/661470/6/check/designate-pdns4/75410dc/ : SUCCESS in 36m 49s\n- designate-bind9-manager-model http://logs.openstack.org/70/661470/6/check/designate-bind9-manager-model/fa54daf/ : SUCCESS in 37m 41s\n- designate-bind9-py27 http://logs.openstack.org/70/661470/6/check/designate-bind9-py27/14fa735/ : SUCCESS in 40m 23s\n- designate-pdns4-py27 http://logs.openstack.org/70/661470/6/check/designate-pdns4-py27/51404c5/ : SUCCESS in 41m 00s\n- designate-pdns4-postgres http://logs.openstack.org/70/661470/6/check/designate-pdns4-postgres/62c3dd8/ : SUCCESS in 39m 12s\n- designate-grenade-pdns4 http://logs.openstack.org/70/661470/6/check/designate-grenade-pdns4/c5bc97f/ : SUCCESS in 58m 46s\n- openstack-tox-cover http://logs.openstack.org/70/661470/6/check/openstack-tox-cover/ca69823/cover/ : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/661470/6/check/openstack-tox-lower-constraints/8775897/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/70/661470/6/check/openstack-tox-pep8/eafc6a7/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/70/661470/6/check/openstack-tox-py27/4701394/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/70/661470/6/check/openstack-tox-py36/8b37152/ : SUCCESS in 5m 17s\n- openstack-tox-py37 http://logs.openstack.org/70/661470/6/check/openstack-tox-py37/92025b0/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/70/661470/6/check/openstack-tox-docs/ae5c569/html/ : SUCCESS in 5m 10s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/70/661470/6/check/neutron-tempest-plugin-designate-scenario/4bb4c30/ : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":6},{"id":"aa20daddf0384131b8c349b6992e1a5f49348cd1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-03 21:42:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6c9fdeb8fd78a00d111bc68efae4bcd4433ab104","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 23:12:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/70/661470/7/check/designate-bind9/1832519/ : SUCCESS in 42m 24s\n- designate-pdns4 http://logs.openstack.org/70/661470/7/check/designate-pdns4/4023698/ : SUCCESS in 41m 28s\n- designate-bind9-manager-model http://logs.openstack.org/70/661470/7/check/designate-bind9-manager-model/9d64a8b/ : SUCCESS in 38m 16s\n- designate-bind9-py27 http://logs.openstack.org/70/661470/7/check/designate-bind9-py27/c17b574/ : SUCCESS in 40m 57s\n- designate-pdns4-py27 http://logs.openstack.org/70/661470/7/check/designate-pdns4-py27/c61e9e1/ : SUCCESS in 38m 46s\n- designate-pdns4-postgres http://logs.openstack.org/70/661470/7/check/designate-pdns4-postgres/42b80c3/ : SUCCESS in 41m 30s\n- designate-grenade-pdns4 http://logs.openstack.org/70/661470/7/check/designate-grenade-pdns4/38d0172/ : SUCCESS in 59m 06s\n- openstack-tox-cover http://logs.openstack.org/70/661470/7/check/openstack-tox-cover/fb7671e/cover/ : SUCCESS in 7m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/661470/7/check/openstack-tox-lower-constraints/a11b1c2/ : SUCCESS in 6m 46s\n- openstack-tox-pep8 http://logs.openstack.org/70/661470/7/check/openstack-tox-pep8/78e0bde/ : SUCCESS in 6m 41s\n- openstack-tox-py27 http://logs.openstack.org/70/661470/7/check/openstack-tox-py27/12da99a/ : SUCCESS in 7m 44s\n- openstack-tox-py36 http://logs.openstack.org/70/661470/7/check/openstack-tox-py36/f491ddb/ : SUCCESS in 8m 29s\n- openstack-tox-py37 http://logs.openstack.org/70/661470/7/check/openstack-tox-py37/5f55072/ : SUCCESS in 7m 23s\n- openstack-tox-docs http://logs.openstack.org/70/661470/7/check/openstack-tox-docs/57797b5/html/ : SUCCESS in 5m 15s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/70/661470/7/check/neutron-tempest-plugin-designate-scenario/e67f1fd/ : SUCCESS in 34m 27s","accounts_in_message":[],"_revision_number":7},{"id":"d476d72a0928b3041f0bc9d57b954b7f0da3ae11","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-04 17:11:57.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"e67c00799486ec0e28cd55dd39232c40f750220c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-09 06:44:14.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"cdf0df62ddaebff19bd9624a5c7f2ad9a608e519","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 07:50:16.000000000","message":"Patch Set 7: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"034da126c6d38d8f28abfeee722e505c2ac0ca0a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-09 08:24:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f0b3e2a7ab394b9f99e17f9ec5900b1217f2e8cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 09:11:31.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/70/661470/8/check/designate-bind9/35775aa/ : SUCCESS in 37m 37s\n- designate-pdns4 http://logs.openstack.org/70/661470/8/check/designate-pdns4/4ef7d76/ : SUCCESS in 35m 09s\n- designate-bind9-manager-model http://logs.openstack.org/70/661470/8/check/designate-bind9-manager-model/3c0f3aa/ : SUCCESS in 33m 31s\n- designate-bind9-py27 http://logs.openstack.org/70/661470/8/check/designate-bind9-py27/7981587/ : SUCCESS in 35m 08s\n- designate-pdns4-py27 http://logs.openstack.org/70/661470/8/check/designate-pdns4-py27/47dd737/ : SUCCESS in 33m 38s\n- designate-pdns4-postgres http://logs.openstack.org/70/661470/8/check/designate-pdns4-postgres/b2abc66/ : SUCCESS in 35m 37s\n- designate-grenade-pdns4 http://logs.openstack.org/70/661470/8/check/designate-grenade-pdns4/f9a0535/ : SUCCESS in 45m 31s\n- openstack-tox-cover http://logs.openstack.org/70/661470/8/check/openstack-tox-cover/0b8a0db/cover/ : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/661470/8/check/openstack-tox-lower-constraints/9849e48/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/70/661470/8/check/openstack-tox-pep8/65771fc/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/70/661470/8/check/openstack-tox-py27/423828f/ : SUCCESS in 5m 56s\n- openstack-tox-py36 http://logs.openstack.org/70/661470/8/check/openstack-tox-py36/8aa3793/ : SUCCESS in 5m 34s\n- openstack-tox-py37 http://logs.openstack.org/70/661470/8/check/openstack-tox-py37/361df21/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/70/661470/8/check/openstack-tox-docs/ae106b1/html/ : SUCCESS in 5m 17s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/70/661470/8/check/neutron-tempest-plugin-designate-scenario/f0e1eb8/ : SUCCESS in 31m 00s","accounts_in_message":[],"_revision_number":8},{"id":"a211736d5e915e4a4a7aeb5744db00a07b8ebf2c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-10 00:12:57.000000000","message":"Patch Set 8:\n\nHad to fix a merge conflict.","accounts_in_message":[],"_revision_number":8},{"id":"2d41e0a7b4e165ff5e10771a8dccf12c92edad51","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-11 12:50:36.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"66a88c7163e8b3f0d23b039e4af00a5753ef9b38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 12:50:51.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"cd1a6a9e86f7cdffa349f5eb3077ef651256474c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 14:07:31.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/70/661470/8/gate/designate-bind9/262fadc/ : SUCCESS in 42m 09s\n- designate-pdns4 http://logs.openstack.org/70/661470/8/gate/designate-pdns4/43411ff/ : SUCCESS in 38m 47s\n- designate-bind9-manager-model http://logs.openstack.org/70/661470/8/gate/designate-bind9-manager-model/8a03b86/ : SUCCESS in 43m 07s\n- designate-bind9-py27 http://logs.openstack.org/70/661470/8/gate/designate-bind9-py27/924cff5/ : SUCCESS in 46m 01s\n- designate-pdns4-py27 http://logs.openstack.org/70/661470/8/gate/designate-pdns4-py27/8d95895/ : SUCCESS in 36m 47s\n- designate-pdns4-postgres http://logs.openstack.org/70/661470/8/gate/designate-pdns4-postgres/9c37a73/ : SUCCESS in 44m 00s\n- designate-grenade-pdns4 http://logs.openstack.org/70/661470/8/gate/designate-grenade-pdns4/0b14d90/ : SUCCESS in 1h 10m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/661470/8/gate/openstack-tox-lower-constraints/ba787b3/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/70/661470/8/gate/openstack-tox-pep8/72a89be/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/70/661470/8/gate/openstack-tox-py27/a740bb5/ : SUCCESS in 5m 48s\n- openstack-tox-py36 http://logs.openstack.org/70/661470/8/gate/openstack-tox-py36/76c74d5/ : SUCCESS in 5m 57s\n- openstack-tox-py37 http://logs.openstack.org/70/661470/8/gate/openstack-tox-py37/8baab9e/ : SUCCESS in 5m 52s\n- openstack-tox-docs http://logs.openstack.org/70/661470/8/gate/openstack-tox-docs/359c446/html/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":8},{"id":"6113aef00f1bf1e48b40f20754f7a70216af673d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 14:07:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d60bd5369af23a06b5467981c5c5ce74a6eb6fc4","revisions":{"3f1ca012f41d71e98eb66a88344fd8056cda84d4":{"kind":"REWORK","_number":1,"created":"2019-05-25 22:02:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/661470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/661470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/661470/1"}}},"commit":{"parents":[{"commit":"4e49b5581fca2bda0d1638efafa88efa8925f05e","subject":"Merge \"Modernized backend tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e49b5581fca2bda0d1638efafa88efa8925f05e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 22:01:29.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 22:01:29.000000000","tz":-420},"subject":"Modernized agent tests","message":"Modernized agent tests\n\n* Moved all tests from test_agent to unit.agent.\n* Cleaned up tests and standardized them.\n\nChange-Id: I048980a74129c603af779ca3f0f9a5ddf750f118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f1ca012f41d71e98eb66a88344fd8056cda84d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f1ca012f41d71e98eb66a88344fd8056cda84d4"}]},"branch":"refs/heads/master"},"03fc832f444481baa27a3951757ee281efb99042":{"kind":"REWORK","_number":2,"created":"2019-05-25 23:30:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/661470/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/661470/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/661470/2"}}},"commit":{"parents":[{"commit":"4e49b5581fca2bda0d1638efafa88efa8925f05e","subject":"Merge \"Modernized backend tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e49b5581fca2bda0d1638efafa88efa8925f05e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 22:01:29.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 23:30:23.000000000","tz":-420},"subject":"Modernized agent tests","message":"Modernized agent tests\n\n* Moved all tests from test_agent to unit.agent.\n* Cleaned up tests and standardized them.\n\nChange-Id: I048980a74129c603af779ca3f0f9a5ddf750f118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03fc832f444481baa27a3951757ee281efb99042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03fc832f444481baa27a3951757ee281efb99042"}]},"branch":"refs/heads/master"},"e06ae0f9a983432297548187617caa50d204dcbc":{"kind":"REWORK","_number":3,"created":"2019-05-26 00:01:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/661470/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/661470/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/661470/3"}}},"commit":{"parents":[{"commit":"4e49b5581fca2bda0d1638efafa88efa8925f05e","subject":"Merge \"Modernized backend tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e49b5581fca2bda0d1638efafa88efa8925f05e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 22:01:29.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-26 00:00:54.000000000","tz":-420},"subject":"Modernized agent tests","message":"Modernized agent tests\n\n* Moved all tests from test_agent to unit.agent.\n* Cleaned up tests and standardized them.\n\nChange-Id: I048980a74129c603af779ca3f0f9a5ddf750f118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e06ae0f9a983432297548187617caa50d204dcbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e06ae0f9a983432297548187617caa50d204dcbc"}]},"branch":"refs/heads/master"},"d2a54a5953c6e874ed39f8573a23c227810cd105":{"kind":"REWORK","_number":4,"created":"2019-05-26 00:17:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/661470/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/661470/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/661470/4"}}},"commit":{"parents":[{"commit":"4e49b5581fca2bda0d1638efafa88efa8925f05e","subject":"Merge \"Modernized backend tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e49b5581fca2bda0d1638efafa88efa8925f05e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 22:01:29.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-26 00:16:54.000000000","tz":-420},"subject":"Modernized agent tests","message":"Modernized agent tests\n\n* Moved all tests from test_agent to unit.agent.\n* Cleaned up tests and standardized them.\n\nChange-Id: I048980a74129c603af779ca3f0f9a5ddf750f118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2a54a5953c6e874ed39f8573a23c227810cd105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2a54a5953c6e874ed39f8573a23c227810cd105"}]},"branch":"refs/heads/master"},"3b73de7e5cfea45d73cd12f6876ad27024e68ac5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-05-28 22:26:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/661470/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/661470/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/661470/5"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 22:01:29.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-28 22:26:11.000000000","tz":0},"subject":"Modernized agent tests","message":"Modernized agent tests\n\n* Moved all tests from test_agent to unit.agent.\n* Cleaned up tests and standardized them.\n\nChange-Id: I048980a74129c603af779ca3f0f9a5ddf750f118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3b73de7e5cfea45d73cd12f6876ad27024e68ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3b73de7e5cfea45d73cd12f6876ad27024e68ac5"}]},"branch":"refs/heads/master"},"59705badf2219a6ac6c7e96340c7100f0f72c4cd":{"kind":"REWORK","_number":6,"created":"2019-05-29 07:38:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/661470/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/661470/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/661470/6"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 22:01:29.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-29 07:38:43.000000000","tz":-420},"subject":"Modernized agent tests","message":"Modernized agent tests\n\n* Moved all tests from test_agent to unit.agent.\n* Cleaned up tests and standardized them.\n\nChange-Id: I048980a74129c603af779ca3f0f9a5ddf750f118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59705badf2219a6ac6c7e96340c7100f0f72c4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59705badf2219a6ac6c7e96340c7100f0f72c4cd"}]},"branch":"refs/heads/master"},"b9c58f9dc5f2f65651a4f60b7335544dcdc9a1ea":{"kind":"REWORK","_number":7,"created":"2019-06-03 21:42:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/661470/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/661470/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/661470/7"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 22:01:29.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-03 21:42:08.000000000","tz":-420},"subject":"Modernized agent tests","message":"Modernized agent tests\n\n* Moved all tests from test_agent to unit.agent.\n* Cleaned up tests and standardized them.\n\nChange-Id: I048980a74129c603af779ca3f0f9a5ddf750f118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9c58f9dc5f2f65651a4f60b7335544dcdc9a1ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9c58f9dc5f2f65651a4f60b7335544dcdc9a1ea"}]},"branch":"refs/heads/master"},"d60bd5369af23a06b5467981c5c5ce74a6eb6fc4":{"kind":"REWORK","_number":8,"created":"2019-06-09 08:24:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/661470/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/661470/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/661470/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/661470/8"}}},"commit":{"parents":[{"commit":"b81755750a9e6b6b991aa41b0738e762c7dca71b","subject":"Merge \"Modernized mdns tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b81755750a9e6b6b991aa41b0738e762c7dca71b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-05-25 22:01:29.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 08:23:57.000000000","tz":-420},"subject":"Modernized agent tests","message":"Modernized agent tests\n\n* Moved all tests from test_agent to unit.agent.\n* Cleaned up tests and standardized them.\n\nChange-Id: I048980a74129c603af779ca3f0f9a5ddf750f118\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d60bd5369af23a06b5467981c5c5ce74a6eb6fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d60bd5369af23a06b5467981c5c5ce74a6eb6fc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
