)]}'
{"id":"openstack%2Fdesignate~662647","triplet_id":"openstack%2Fdesignate~master~I6e8affbabeca4162f11fb4ba47ffa2e960103ccd","project":"openstack/designate","branch":"master","topic":"deprecated_warnings","hashtags":[],"change_id":"I6e8affbabeca4162f11fb4ba47ffa2e960103ccd","subject":"Updated context to use project_id","status":"MERGED","created":"2019-06-02 20:20:03.000000000","updated":"2019-06-17 09:25:37.000000000","submitted":"2019-06-17 09:25:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662647-1560763537013-018dc456","meta_rev_id":"94d814a4b438d4da69912f8756fdc5c89b094203","_number":662647,"virtual_id_number":662647,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-06-17 09:25:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-06-16 19:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-16 22:47:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 17:09:17.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-16 19:18:57.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-06-17 09:25:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65e1ad10c9975ade197c452785e77c0132d32e3f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-02 20:20:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4891278e7b72dea24d6c0874b859d658ebbdb97","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-02 20:23:33.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4862ce74e9e6d808dbc4f8f2229aaf145fe758b3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-02 20:52:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"31c587a9c8bee13acd931de7f88d48dda553e67c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 21:43:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/47/662647/3/check/designate-bind9/4621865/ : SUCCESS in 43m 08s\n- designate-pdns4 http://logs.openstack.org/47/662647/3/check/designate-pdns4/1b3e935/ : SUCCESS in 44m 40s\n- designate-bind9-manager-model http://logs.openstack.org/47/662647/3/check/designate-bind9-manager-model/2d63842/ : SUCCESS in 48m 02s\n- designate-bind9-py27 http://logs.openstack.org/47/662647/3/check/designate-bind9-py27/5814c27/ : SUCCESS in 38m 20s\n- designate-pdns4-py27 http://logs.openstack.org/47/662647/3/check/designate-pdns4-py27/f563b80/ : SUCCESS in 46m 30s\n- designate-pdns4-postgres http://logs.openstack.org/47/662647/3/check/designate-pdns4-postgres/ec0c03e/ : SUCCESS in 32m 10s\n- designate-grenade-pdns4 http://logs.openstack.org/47/662647/3/check/designate-grenade-pdns4/2f57732/ : SUCCESS in 48m 15s\n- openstack-tox-cover http://logs.openstack.org/47/662647/3/check/openstack-tox-cover/8692d7a/cover/ : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/662647/3/check/openstack-tox-lower-constraints/bd9c6c9/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/47/662647/3/check/openstack-tox-pep8/ed3816e/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/47/662647/3/check/openstack-tox-py27/14d3296/ : SUCCESS in 5m 59s\n- openstack-tox-py36 http://logs.openstack.org/47/662647/3/check/openstack-tox-py36/7d1922a/ : SUCCESS in 7m 22s\n- openstack-tox-py37 http://logs.openstack.org/47/662647/3/check/openstack-tox-py37/6552781/ : SUCCESS in 8m 20s\n- openstack-tox-docs http://logs.openstack.org/47/662647/3/check/openstack-tox-docs/1102acd/html/ : SUCCESS in 6m 54s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/47/662647/3/check/neutron-tempest-plugin-designate-scenario/b72a0a8/ : SUCCESS in 35m 06s","accounts_in_message":[],"_revision_number":3},{"id":"0f4530613d4dfb6cc27286f71cf3772af65a50a5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-03 05:40:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2dfeb2dfae6a7ea6cbae69329c6424fe5e43bba4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-03 05:46:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6fdbfcb362c9ac5c2a83590af666a9ce3e81b299","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 06:48:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/47/662647/5/check/designate-bind9/9294a4b/ : SUCCESS in 38m 14s\n- designate-pdns4 http://logs.openstack.org/47/662647/5/check/designate-pdns4/b590779/ : SUCCESS in 56m 43s\n- designate-bind9-manager-model http://logs.openstack.org/47/662647/5/check/designate-bind9-manager-model/7171e2e/ : SUCCESS in 38m 56s\n- designate-bind9-py27 http://logs.openstack.org/47/662647/5/check/designate-bind9-py27/7b7be91/ : SUCCESS in 1h 00m 04s\n- designate-pdns4-py27 http://logs.openstack.org/47/662647/5/check/designate-pdns4-py27/056d0ca/ : SUCCESS in 36m 05s\n- designate-pdns4-postgres http://logs.openstack.org/47/662647/5/check/designate-pdns4-postgres/227b7da/ : SUCCESS in 32m 52s\n- designate-grenade-pdns4 http://logs.openstack.org/47/662647/5/check/designate-grenade-pdns4/d6e6508/ : SUCCESS in 52m 10s\n- openstack-tox-cover http://logs.openstack.org/47/662647/5/check/openstack-tox-cover/7b6f2a1/cover/ : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/662647/5/check/openstack-tox-lower-constraints/b6c657e/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/47/662647/5/check/openstack-tox-pep8/c8575bb/ : SUCCESS in 6m 59s\n- openstack-tox-py27 http://logs.openstack.org/47/662647/5/check/openstack-tox-py27/7be57b1/ : SUCCESS in 6m 10s\n- openstack-tox-py36 http://logs.openstack.org/47/662647/5/check/openstack-tox-py36/42a56cb/ : SUCCESS in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/47/662647/5/check/openstack-tox-py37/44916da/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/47/662647/5/check/openstack-tox-docs/604937c/html/ : SUCCESS in 5m 45s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/47/662647/5/check/neutron-tempest-plugin-designate-scenario/505a8e8/ : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":5},{"id":"5b2e06a616a0f6e0b9a875ac4f8e36154e376d8c","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-04 17:09:17.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e073d2f122c2767a5aac046624b547ab38928222","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-09 22:48:07.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1ede1498a1dd76a4611ebd0d2afa8667193b3482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 23:59:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/47/662647/6/check/designate-bind9/891868d/ : SUCCESS in 39m 17s\n- designate-pdns4 http://logs.openstack.org/47/662647/6/check/designate-pdns4/2204e98/ : SUCCESS in 37m 44s\n- designate-bind9-manager-model http://logs.openstack.org/47/662647/6/check/designate-bind9-manager-model/9f06ef2/ : SUCCESS in 38m 57s\n- designate-bind9-py27 http://logs.openstack.org/47/662647/6/check/designate-bind9-py27/ddda6b2/ : SUCCESS in 33m 24s\n- designate-pdns4-py27 http://logs.openstack.org/47/662647/6/check/designate-pdns4-py27/c46f84b/ : SUCCESS in 35m 32s\n- designate-pdns4-postgres http://logs.openstack.org/47/662647/6/check/designate-pdns4-postgres/cf32f23/ : SUCCESS in 29m 14s\n- designate-grenade-pdns4 http://logs.openstack.org/47/662647/6/check/designate-grenade-pdns4/0b8968f/ : SUCCESS in 59m 10s\n- openstack-tox-cover http://logs.openstack.org/47/662647/6/check/openstack-tox-cover/1999bb9/cover/ : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/662647/6/check/openstack-tox-lower-constraints/1271b03/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/47/662647/6/check/openstack-tox-pep8/7788641/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/47/662647/6/check/openstack-tox-py27/2757582/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/47/662647/6/check/openstack-tox-py36/1819fc6/ : SUCCESS in 5m 32s\n- openstack-tox-py37 http://logs.openstack.org/47/662647/6/check/openstack-tox-py37/3a0fb1b/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/47/662647/6/check/openstack-tox-docs/3a231e0/html/ : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/47/662647/6/check/neutron-tempest-plugin-designate-scenario/8eeb848/ : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":6},{"id":"9d273465e1c60078655a81114f356b1a5e77f963","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-06-13 08:53:49.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nAlmost +2 except for a small question","accounts_in_message":[],"_revision_number":6},{"id":"8ca795eedffccfcb10e8774268f767e1fbedbb5f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-13 18:08:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7b8bf06993a7e12b4ea4c1fa6595cab53b360d00","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-13 18:09:40.000000000","message":"Patch Set 7:\n\n\u003e (1 comment)\n \u003e \n \u003e Almost +2 except for a small question\n\n \u003e (1 comment)\n \u003e \n \u003e Almost +2 except for a small question\n\nThose changes were unintentional. Probably relics from some earlier changes.","accounts_in_message":[],"_revision_number":7},{"id":"55efea66897c0a7f4bb09b049b67d8e9e70aff2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 19:19:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/47/662647/7/check/designate-bind9/7666fda/ : SUCCESS in 42m 29s\n- designate-pdns4 http://logs.openstack.org/47/662647/7/check/designate-pdns4/3e6c4b4/ : SUCCESS in 37m 32s\n- designate-bind9-manager-model http://logs.openstack.org/47/662647/7/check/designate-bind9-manager-model/76948b0/ : SUCCESS in 43m 49s\n- designate-bind9-py27 http://logs.openstack.org/47/662647/7/check/designate-bind9-py27/1141227/ : SUCCESS in 34m 31s\n- designate-pdns4-py27 http://logs.openstack.org/47/662647/7/check/designate-pdns4-py27/2fc4aa6/ : FAILURE in 17m 20s\n- designate-pdns4-postgres http://logs.openstack.org/47/662647/7/check/designate-pdns4-postgres/f46c53e/ : FAILURE in 17m 33s\n- designate-grenade-pdns4 http://logs.openstack.org/47/662647/7/check/designate-grenade-pdns4/ee0bcca/ : SUCCESS in 59m 52s\n- openstack-tox-cover http://logs.openstack.org/47/662647/7/check/openstack-tox-cover/ff21968/cover/ : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/662647/7/check/openstack-tox-lower-constraints/6957e2f/ : SUCCESS in 5m 11s\n- openstack-tox-pep8 http://logs.openstack.org/47/662647/7/check/openstack-tox-pep8/aae0638/ : SUCCESS in 6m 17s\n- openstack-tox-py27 http://logs.openstack.org/47/662647/7/check/openstack-tox-py27/8a90819/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/47/662647/7/check/openstack-tox-py36/73304b9/ : SUCCESS in 8m 43s\n- openstack-tox-py37 http://logs.openstack.org/47/662647/7/check/openstack-tox-py37/23a0a99/ : SUCCESS in 6m 24s\n- openstack-tox-docs http://logs.openstack.org/47/662647/7/check/openstack-tox-docs/bdb6fc3/html/ : SUCCESS in 5m 17s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/47/662647/7/check/neutron-tempest-plugin-designate-scenario/cae8e15/ : FAILURE in 19m 19s","accounts_in_message":[],"_revision_number":7},{"id":"878b0eb28728f0fa2c3fed74858d9e386e67f5f1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-13 19:50:46.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"359bbcbb17f46b783c1ef67004472bfbdd979fb9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-13 20:28:12.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"34630f8de783a16d269a8c65e58487235d21f09a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-13 20:29:47.000000000","message":"Patch Set 8:\n\nWas hitting mirror errors again.\n\u003e E: Failed to fetch X  404  Not Found [IP: Y 80]\nTook the opportunity to rebase and used that to stop the CI early.","accounts_in_message":[],"_revision_number":8},{"id":"45c25af5678ab98280a1205950277dd9d92c6b98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 21:53:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/47/662647/8/check/designate-bind9/689cc5e/ : FAILURE in 16m 31s\n- designate-pdns4 http://logs.openstack.org/47/662647/8/check/designate-pdns4/1bb6184/ : FAILURE in 20m 27s\n- designate-bind9-manager-model http://logs.openstack.org/47/662647/8/check/designate-bind9-manager-model/f8dabfb/ : SUCCESS in 38m 30s\n- designate-bind9-py27 http://logs.openstack.org/47/662647/8/check/designate-bind9-py27/0d8cbc5/ : SUCCESS in 42m 09s\n- designate-pdns4-py27 http://logs.openstack.org/47/662647/8/check/designate-pdns4-py27/144d8d6/ : SUCCESS in 39m 27s\n- designate-pdns4-postgres http://logs.openstack.org/47/662647/8/check/designate-pdns4-postgres/1266428/ : SUCCESS in 48m 00s\n- designate-grenade-pdns4 http://logs.openstack.org/47/662647/8/check/designate-grenade-pdns4/d2c76a6/ : SUCCESS in 1h 10m 26s\n- openstack-tox-cover http://logs.openstack.org/47/662647/8/check/openstack-tox-cover/c612ba7/cover/ : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/662647/8/check/openstack-tox-lower-constraints/8aa7055/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/47/662647/8/check/openstack-tox-pep8/7c58a72/ : SUCCESS in 5m 50s\n- openstack-tox-py27 http://logs.openstack.org/47/662647/8/check/openstack-tox-py27/fd22e60/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/47/662647/8/check/openstack-tox-py36/6e32f9e/ : SUCCESS in 8m 18s\n- openstack-tox-py37 http://logs.openstack.org/47/662647/8/check/openstack-tox-py37/3ca7f2b/ : SUCCESS in 6m 42s\n- openstack-tox-docs http://logs.openstack.org/47/662647/8/check/openstack-tox-docs/eb2c38b/html/ : SUCCESS in 6m 07s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/47/662647/8/check/neutron-tempest-plugin-designate-scenario/9f9098f/ : SUCCESS in 27m 15s","accounts_in_message":[],"_revision_number":8},{"id":"7ba3ee700836b6dd49ce1b47a2bc7d06a5ade35c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-13 22:40:55.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"11c90212b77d458cedefb6053cf09371c0537904","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-13 23:31:38.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"20c8e5e167fac510bf2ba29750f5109c7f98e0b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 23:52:18.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/47/662647/8/check/designate-bind9/4d7a0f4/ : SUCCESS in 42m 38s\n- designate-pdns4 http://logs.openstack.org/47/662647/8/check/designate-pdns4/9218665/ : SUCCESS in 35m 55s\n- designate-bind9-manager-model http://logs.openstack.org/47/662647/8/check/designate-bind9-manager-model/c19c918/ : SUCCESS in 46m 53s\n- designate-bind9-py27 http://logs.openstack.org/47/662647/8/check/designate-bind9-py27/27219bf/ : SUCCESS in 45m 27s\n- designate-pdns4-py27 http://logs.openstack.org/47/662647/8/check/designate-pdns4-py27/5c74f93/ : SUCCESS in 44m 04s\n- designate-pdns4-postgres http://logs.openstack.org/47/662647/8/check/designate-pdns4-postgres/4a2e1c0/ : SUCCESS in 39m 37s\n- designate-grenade-pdns4 http://logs.openstack.org/47/662647/8/check/designate-grenade-pdns4/223ba3c/ : SUCCESS in 57m 52s\n- openstack-tox-cover http://logs.openstack.org/47/662647/8/check/openstack-tox-cover/907dca2/cover/ : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/662647/8/check/openstack-tox-lower-constraints/7ecabf1/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/47/662647/8/check/openstack-tox-pep8/75b7c4f/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/47/662647/8/check/openstack-tox-py27/746bb72/ : FAILURE in 8m 12s\n- openstack-tox-py36 http://logs.openstack.org/47/662647/8/check/openstack-tox-py36/cd5ea85/ : SUCCESS in 5m 17s\n- openstack-tox-py37 http://logs.openstack.org/47/662647/8/check/openstack-tox-py37/5319450/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/47/662647/8/check/openstack-tox-docs/02f0e8d/html/ : SUCCESS in 4m 57s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/47/662647/8/check/neutron-tempest-plugin-designate-scenario/4e65807/ : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":8},{"id":"75bedec802f964f0c03f2620dba7509cdd4c9039","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 01:52:36.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"81378e94b9e50a356d266b7308a3f28dde9fc390","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:03:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/47/662647/8/check/designate-bind9/6e53633/ : SUCCESS in 41m 03s\n- designate-pdns4 http://logs.openstack.org/47/662647/8/check/designate-pdns4/2a6fdaf/ : SUCCESS in 39m 31s\n- designate-bind9-manager-model http://logs.openstack.org/47/662647/8/check/designate-bind9-manager-model/12d92ec/ : SUCCESS in 38m 37s\n- designate-bind9-py27 http://logs.openstack.org/47/662647/8/check/designate-bind9-py27/c489d44/ : SUCCESS in 36m 36s\n- designate-pdns4-py27 http://logs.openstack.org/47/662647/8/check/designate-pdns4-py27/7c3bbaa/ : SUCCESS in 39m 58s\n- designate-pdns4-postgres http://logs.openstack.org/47/662647/8/check/designate-pdns4-postgres/da9e170/ : SUCCESS in 38m 58s\n- designate-grenade-pdns4 http://logs.openstack.org/47/662647/8/check/designate-grenade-pdns4/9534edb/ : SUCCESS in 1h 03m 40s\n- openstack-tox-cover http://logs.openstack.org/47/662647/8/check/openstack-tox-cover/fefb716/cover/ : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/662647/8/check/openstack-tox-lower-constraints/c5f17ef/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/47/662647/8/check/openstack-tox-pep8/7eb33f2/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/47/662647/8/check/openstack-tox-py27/05a895b/ : SUCCESS in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/47/662647/8/check/openstack-tox-py36/fbf2b30/ : SUCCESS in 5m 38s\n- openstack-tox-py37 http://logs.openstack.org/47/662647/8/check/openstack-tox-py37/794d102/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/47/662647/8/check/openstack-tox-docs/133f2de/html/ : SUCCESS in 4m 36s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/47/662647/8/check/neutron-tempest-plugin-designate-scenario/ae08966/ : SUCCESS in 35m 30s","accounts_in_message":[],"_revision_number":8},{"id":"ffd489f80e8cb0e34ccc5d010fa391d5da9963c7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-06-16 19:18:57.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"b7fa746bfb05d2b63aa3f150c42a15459fb113e0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-16 22:47:39.000000000","message":"Patch Set 8: Workflow+1\n\nMerging this as it was already approved by Graham Hayes with only a small modification done to the unit-tests.","accounts_in_message":[],"_revision_number":8},{"id":"7bca089cfa4a6b1f54c450778df5df0a02777eb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-16 22:47:51.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6353f1ade5a355532177bd554773a919c68e8255","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 00:08:00.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/47/662647/8/gate/designate-bind9/6f5db6c/ : SUCCESS in 38m 42s\n- designate-pdns4 http://logs.openstack.org/47/662647/8/gate/designate-pdns4/9cec316/ : SUCCESS in 35m 37s\n- designate-bind9-manager-model http://logs.openstack.org/47/662647/8/gate/designate-bind9-manager-model/3f2e73c/ : SUCCESS in 38m 21s\n- designate-bind9-py27 http://logs.openstack.org/47/662647/8/gate/designate-bind9-py27/38c0ab9/ : SUCCESS in 39m 56s\n- designate-pdns4-py27 http://logs.openstack.org/47/662647/8/gate/designate-pdns4-py27/9fba450/ : SUCCESS in 32m 39s\n- designate-pdns4-postgres http://logs.openstack.org/47/662647/8/gate/designate-pdns4-postgres/0c09a59/ : SUCCESS in 45m 33s\n- designate-grenade-pdns4 http://logs.openstack.org/47/662647/8/gate/designate-grenade-pdns4/8fe3cc9/ : SUCCESS in 58m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/662647/8/gate/openstack-tox-lower-constraints/d70591d/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/47/662647/8/gate/openstack-tox-pep8/ec453d2/ : SUCCESS in 5m 59s\n- openstack-tox-py27 http://logs.openstack.org/47/662647/8/gate/openstack-tox-py27/d7b3049/ : SUCCESS in 7m 29s\n- openstack-tox-py36 http://logs.openstack.org/47/662647/8/gate/openstack-tox-py36/b567f0c/ : SUCCESS in 6m 14s\n- openstack-tox-py37 http://logs.openstack.org/47/662647/8/gate/openstack-tox-py37/44914b2/ : SUCCESS in 8m 33s\n- openstack-tox-docs http://logs.openstack.org/47/662647/8/gate/openstack-tox-docs/692ee10/ : FAILURE in 6m 25s","accounts_in_message":[],"_revision_number":8},{"id":"27dd50bc512622cd6f33d0c775a27b3a24a254fa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-06-17 06:52:37.000000000","message":"Patch Set 8:\n\nrecheck doc fix merged","accounts_in_message":[],"_revision_number":8},{"id":"85a63c382463cb756dca53edad4ca35b2aab4925","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 08:04:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/47/662647/8/check/designate-bind9/5f67f83/ : SUCCESS in 38m 18s\n- designate-pdns4 http://logs.openstack.org/47/662647/8/check/designate-pdns4/4acb10a/ : SUCCESS in 40m 27s\n- designate-bind9-manager-model http://logs.openstack.org/47/662647/8/check/designate-bind9-manager-model/4f75194/ : SUCCESS in 39m 24s\n- designate-bind9-py27 http://logs.openstack.org/47/662647/8/check/designate-bind9-py27/7fa7517/ : SUCCESS in 38m 25s\n- designate-pdns4-py27 http://logs.openstack.org/47/662647/8/check/designate-pdns4-py27/b2186d6/ : SUCCESS in 39m 13s\n- designate-pdns4-postgres http://logs.openstack.org/47/662647/8/check/designate-pdns4-postgres/754b530/ : SUCCESS in 38m 17s\n- designate-grenade-pdns4 http://logs.openstack.org/47/662647/8/check/designate-grenade-pdns4/cdddf5a/ : SUCCESS in 56m 56s\n- openstack-tox-cover http://logs.openstack.org/47/662647/8/check/openstack-tox-cover/f85bc27/cover/ : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/662647/8/check/openstack-tox-lower-constraints/851f7e9/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/47/662647/8/check/openstack-tox-pep8/cbebd9c/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/47/662647/8/check/openstack-tox-py27/f06c66c/ : SUCCESS in 7m 59s\n- openstack-tox-py36 http://logs.openstack.org/47/662647/8/check/openstack-tox-py36/125ac0d/ : SUCCESS in 6m 17s\n- openstack-tox-py37 http://logs.openstack.org/47/662647/8/check/openstack-tox-py37/aa7bbbf/ : SUCCESS in 6m 10s\n- openstack-tox-docs http://logs.openstack.org/47/662647/8/check/openstack-tox-docs/1e5cd38/html/ : SUCCESS in 4m 54s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/47/662647/8/check/neutron-tempest-plugin-designate-scenario/820d84e/ : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":8},{"id":"5e3ca590ab8e39cdcb2504afe8f093800709fdab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 08:04:46.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f6278a1ab8db3bee619b9eeb8b0f7e040be8eb1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 09:25:36.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/47/662647/8/gate/designate-bind9/0f78640/ : SUCCESS in 43m 09s\n- designate-pdns4 http://logs.openstack.org/47/662647/8/gate/designate-pdns4/a5fdbb3/ : SUCCESS in 42m 15s\n- designate-bind9-manager-model http://logs.openstack.org/47/662647/8/gate/designate-bind9-manager-model/1ac7a4c/ : SUCCESS in 46m 30s\n- designate-bind9-py27 http://logs.openstack.org/47/662647/8/gate/designate-bind9-py27/87a0e82/ : SUCCESS in 39m 58s\n- designate-pdns4-py27 http://logs.openstack.org/47/662647/8/gate/designate-pdns4-py27/838b845/ : SUCCESS in 38m 52s\n- designate-pdns4-postgres http://logs.openstack.org/47/662647/8/gate/designate-pdns4-postgres/cd77377/ : SUCCESS in 38m 03s\n- designate-grenade-pdns4 http://logs.openstack.org/47/662647/8/gate/designate-grenade-pdns4/690d86c/ : SUCCESS in 1h 00m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/662647/8/gate/openstack-tox-lower-constraints/90ef1d6/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/47/662647/8/gate/openstack-tox-pep8/9a2b887/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/47/662647/8/gate/openstack-tox-py27/aa7a642/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/47/662647/8/gate/openstack-tox-py36/37b0ce0/ : SUCCESS in 5m 52s\n- openstack-tox-py37 http://logs.openstack.org/47/662647/8/gate/openstack-tox-py37/32d8f9a/ : SUCCESS in 7m 33s\n- openstack-tox-docs http://logs.openstack.org/47/662647/8/gate/openstack-tox-docs/9651045/html/ : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":8},{"id":"9e8b0e82d5a1eed42f81f77ee10f3c1ffcb8739a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 09:25:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"03b390dee5bf80e486b4bfaac9ffddab157e742a","revisions":{"630a2258df37ff732a121e8d4b66ce9b980d69de":{"kind":"REWORK","_number":1,"created":"2019-06-02 20:20:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/662647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/662647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/662647/1"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:18:49.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:18:49.000000000","tz":-420},"subject":"Updated context to use project_id","message":"Updated context to use project_id\n\nFixed a bunch of deprecated warnings from oslo.context.\n\n* Use project_id instead of tenant.\n\nChange-Id: I6e8affbabeca4162f11fb4ba47ffa2e960103ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/630a2258df37ff732a121e8d4b66ce9b980d69de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/630a2258df37ff732a121e8d4b66ce9b980d69de"}]},"branch":"refs/heads/master"},"713d800952dcf44115b8dcbb225c325319e20ef4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-02 20:23:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/662647/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/662647/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/662647/2"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:18:49.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:20:53.000000000","tz":-420},"subject":"Updated context to use project_id","message":"Updated context to use project_id\n\nFixed a bunch of deprecated warnings from oslo.context.\n\n* Use project_id instead of tenant.\n* Use user_id, domain_id etc.\n\nChange-Id: I6e8affbabeca4162f11fb4ba47ffa2e960103ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/713d800952dcf44115b8dcbb225c325319e20ef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/713d800952dcf44115b8dcbb225c325319e20ef4"}]},"branch":"refs/heads/master"},"75781847c95d3b49c4013af2271bbc9ff44861e5":{"kind":"REWORK","_number":3,"created":"2019-06-02 20:52:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/662647/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/662647/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/662647/3"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:18:49.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:52:23.000000000","tz":-420},"subject":"Updated context to use project_id","message":"Updated context to use project_id\n\nFixed a bunch of deprecated warnings from oslo.context.\n\n* Use project_id instead of tenant.\n* Use user_id, domain_id etc.\n\nChange-Id: I6e8affbabeca4162f11fb4ba47ffa2e960103ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75781847c95d3b49c4013af2271bbc9ff44861e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75781847c95d3b49c4013af2271bbc9ff44861e5"}]},"branch":"refs/heads/master"},"3d265a26e47c54c390d6b5519ff19a08514357ff":{"kind":"REWORK","_number":4,"created":"2019-06-03 05:40:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/662647/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/662647/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/662647/4"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:18:49.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-03 05:40:09.000000000","tz":-420},"subject":"Updated context to use project_id","message":"Updated context to use project_id\n\nFixed a bunch of deprecated warnings from oslo.context.\n\n* Use project_id instead of tenant.\n* Use user_id, domain_id etc.\n\nChange-Id: I6e8affbabeca4162f11fb4ba47ffa2e960103ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3d265a26e47c54c390d6b5519ff19a08514357ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3d265a26e47c54c390d6b5519ff19a08514357ff"}]},"branch":"refs/heads/master"},"859706098c504a64a9f3254e6422f065cb7ddbd8":{"kind":"REWORK","_number":5,"created":"2019-06-03 05:46:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/662647/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/662647/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/662647/5"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:18:49.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-03 05:46:31.000000000","tz":-420},"subject":"Updated context to use project_id","message":"Updated context to use project_id\n\nFixed a bunch of deprecated warnings from oslo.context.\n\n* Use project_id instead of tenant.\n* Use user_id, domain_id etc.\n\nChange-Id: I6e8affbabeca4162f11fb4ba47ffa2e960103ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/859706098c504a64a9f3254e6422f065cb7ddbd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/859706098c504a64a9f3254e6422f065cb7ddbd8"}]},"branch":"refs/heads/master"},"e584c41ec89d65a7c8b4eac841f1d03be573847a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-09 22:48:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/662647/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/662647/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/662647/6"}}},"commit":{"parents":[{"commit":"174365947394ef304a3aa729658765bc3d36eabf","subject":"Merge \"Remove KS v2 support and add region support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/174365947394ef304a3aa729658765bc3d36eabf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:18:49.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 22:48:07.000000000","tz":0},"subject":"Updated context to use project_id","message":"Updated context to use project_id\n\nFixed a bunch of deprecated warnings from oslo.context.\n\n* Use project_id instead of tenant.\n* Use user_id, domain_id etc.\n\nChange-Id: I6e8affbabeca4162f11fb4ba47ffa2e960103ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e584c41ec89d65a7c8b4eac841f1d03be573847a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e584c41ec89d65a7c8b4eac841f1d03be573847a"}]},"branch":"refs/heads/master"},"5ee0fd1a4117dd294f6134210b7c4bd82799c132":{"kind":"REWORK","_number":7,"created":"2019-06-13 18:08:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/662647/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/662647/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/662647/7"}}},"commit":{"parents":[{"commit":"174365947394ef304a3aa729658765bc3d36eabf","subject":"Merge \"Remove KS v2 support and add region support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/174365947394ef304a3aa729658765bc3d36eabf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:18:49.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-13 18:08:48.000000000","tz":-420},"subject":"Updated context to use project_id","message":"Updated context to use project_id\n\nFixed a bunch of deprecated warnings from oslo.context.\n\n* Use project_id instead of tenant.\n* Use user_id, domain_id etc.\n\nChange-Id: I6e8affbabeca4162f11fb4ba47ffa2e960103ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5ee0fd1a4117dd294f6134210b7c4bd82799c132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5ee0fd1a4117dd294f6134210b7c4bd82799c132"}]},"branch":"refs/heads/master"},"03b390dee5bf80e486b4bfaac9ffddab157e742a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-06-13 20:28:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/47/662647/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/662647/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/662647/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/662647/8"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 20:18:49.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-13 20:28:12.000000000","tz":0},"subject":"Updated context to use project_id","message":"Updated context to use project_id\n\nFixed a bunch of deprecated warnings from oslo.context.\n\n* Use project_id instead of tenant.\n* Use user_id, domain_id etc.\n\nChange-Id: I6e8affbabeca4162f11fb4ba47ffa2e960103ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03b390dee5bf80e486b4bfaac9ffddab157e742a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03b390dee5bf80e486b4bfaac9ffddab157e742a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
