)]}'
{"id":"openstack%2Fdesignate~662651","triplet_id":"openstack%2Fdesignate~master~I2129d682f5b88e38090b24de827ccb05f78f4782","project":"openstack/designate","branch":"master","topic":"standardize_unit_tests","hashtags":[],"change_id":"I2129d682f5b88e38090b24de827ccb05f78f4782","subject":"Renamed unit.test_api to unit.api","status":"MERGED","created":"2019-06-02 22:44:02.000000000","updated":"2019-06-04 23:18:04.000000000","submitted":"2019-06-04 23:18:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662651-1559690284814-d9d96933","meta_rev_id":"e4ce074d6872a007864059ffa1b509eb07d8d31b","_number":662651,"virtual_id_number":662651,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-06-04 23:18:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-04 17:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-04 21:31:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 17:19:04.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-04 23:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1a72bcd719de45f27f0688ac832ac08cd85b18f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-02 22:44:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52451c1952e4c0c4ef3fe352f21971326f3ca9a3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-02 22:45:07.000000000","message":"Topic changed from api to standardize_unit_tests","accounts_in_message":[],"_revision_number":1},{"id":"7b0b96242c1608bfcec8a06c2b40bf3feae71d66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 23:43:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/51/662651/1/check/designate-bind9/b138b8c/ : SUCCESS in 40m 09s\n- designate-pdns4 http://logs.openstack.org/51/662651/1/check/designate-pdns4/c5b957b/ : SUCCESS in 35m 22s\n- designate-bind9-manager-model http://logs.openstack.org/51/662651/1/check/designate-bind9-manager-model/e84c7c3/ : SUCCESS in 38m 41s\n- designate-bind9-py27 http://logs.openstack.org/51/662651/1/check/designate-bind9-py27/e82a38d/ : SUCCESS in 40m 11s\n- designate-pdns4-py27 http://logs.openstack.org/51/662651/1/check/designate-pdns4-py27/341e4da/ : SUCCESS in 35m 13s\n- designate-pdns4-postgres http://logs.openstack.org/51/662651/1/check/designate-pdns4-postgres/6c71ca8/ : SUCCESS in 35m 56s\n- designate-grenade-pdns4 http://logs.openstack.org/51/662651/1/check/designate-grenade-pdns4/1353a99/ : SUCCESS in 54m 30s\n- openstack-tox-cover http://logs.openstack.org/51/662651/1/check/openstack-tox-cover/cc3355f/cover/ : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/662651/1/check/openstack-tox-lower-constraints/e18303f/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/51/662651/1/check/openstack-tox-pep8/9ec9c80/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/51/662651/1/check/openstack-tox-py27/263e714/ : SUCCESS in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/51/662651/1/check/openstack-tox-py36/4dd0d63/ : SUCCESS in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/51/662651/1/check/openstack-tox-py37/b5dc810/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/51/662651/1/check/openstack-tox-docs/4b8f752/html/ : SUCCESS in 4m 26s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/51/662651/1/check/neutron-tempest-plugin-designate-scenario/bc15923/ : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4583fbffae39627179e2a46fbeabbb75b496d9c0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-03 21:27:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"116aded160d841bc83311df131b9f19f120e4f5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 22:30:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/51/662651/2/check/designate-bind9/b4cc897/ : SUCCESS in 36m 29s\n- designate-pdns4 http://logs.openstack.org/51/662651/2/check/designate-pdns4/89cb987/ : SUCCESS in 32m 33s\n- designate-bind9-manager-model http://logs.openstack.org/51/662651/2/check/designate-bind9-manager-model/ff83d95/ : SUCCESS in 39m 35s\n- designate-bind9-py27 http://logs.openstack.org/51/662651/2/check/designate-bind9-py27/38828ea/ : SUCCESS in 42m 46s\n- designate-pdns4-py27 http://logs.openstack.org/51/662651/2/check/designate-pdns4-py27/dfeff7f/ : SUCCESS in 42m 16s\n- designate-pdns4-postgres http://logs.openstack.org/51/662651/2/check/designate-pdns4-postgres/a93ffa7/ : SUCCESS in 36m 17s\n- designate-grenade-pdns4 http://logs.openstack.org/51/662651/2/check/designate-grenade-pdns4/29d7ea9/ : SUCCESS in 55m 27s\n- openstack-tox-cover http://logs.openstack.org/51/662651/2/check/openstack-tox-cover/6e97ede/cover/ : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/662651/2/check/openstack-tox-lower-constraints/1923ff7/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/51/662651/2/check/openstack-tox-pep8/84d186a/ : SUCCESS in 9m 42s\n- openstack-tox-py27 http://logs.openstack.org/51/662651/2/check/openstack-tox-py27/ba17bcf/ : SUCCESS in 6m 10s\n- openstack-tox-py36 http://logs.openstack.org/51/662651/2/check/openstack-tox-py36/17e7c20/ : SUCCESS in 5m 40s\n- openstack-tox-py37 http://logs.openstack.org/51/662651/2/check/openstack-tox-py37/8d8f5bb/ : SUCCESS in 7m 10s\n- openstack-tox-docs http://logs.openstack.org/51/662651/2/check/openstack-tox-docs/c1d175e/html/ : SUCCESS in 6m 14s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/51/662651/2/check/neutron-tempest-plugin-designate-scenario/f9606c3/ : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":2},{"id":"cc410c881a7a061ca6aafa0d1cbb46e2bda8cc07","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-04 17:19:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5c2ad7d04d99ee23f46eacb7f8bac12af442feb6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-04 21:31:40.000000000","message":"Patch Set 2: Workflow+1\n\nMerging with only one Core reviewers +2 as the change are minor.","accounts_in_message":[],"_revision_number":2},{"id":"e93d7a8454704983adf01bd98f2125350f4cd35c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 21:31:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ca4421e06795fa447eac69d65f765be66d0bcb27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 23:18:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/51/662651/2/gate/designate-bind9/e308f1a/ : SUCCESS in 35m 55s\n- designate-pdns4 http://logs.openstack.org/51/662651/2/gate/designate-pdns4/f7786d1/ : SUCCESS in 38m 25s\n- designate-bind9-manager-model http://logs.openstack.org/51/662651/2/gate/designate-bind9-manager-model/ed4c338/ : SUCCESS in 37m 28s\n- designate-bind9-py27 http://logs.openstack.org/51/662651/2/gate/designate-bind9-py27/c4c17d6/ : SUCCESS in 40m 08s\n- designate-pdns4-py27 http://logs.openstack.org/51/662651/2/gate/designate-pdns4-py27/403323d/ : SUCCESS in 36m 23s\n- designate-pdns4-postgres http://logs.openstack.org/51/662651/2/gate/designate-pdns4-postgres/8ff14b2/ : SUCCESS in 39m 41s\n- designate-grenade-pdns4 http://logs.openstack.org/51/662651/2/gate/designate-grenade-pdns4/5088d63/ : SUCCESS in 55m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/662651/2/gate/openstack-tox-lower-constraints/fe3bacf/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/51/662651/2/gate/openstack-tox-pep8/b51d986/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/51/662651/2/gate/openstack-tox-py27/d94b9e2/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/51/662651/2/gate/openstack-tox-py36/bb04cc8/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/51/662651/2/gate/openstack-tox-py37/250d652/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/51/662651/2/gate/openstack-tox-docs/2274bd9/html/ : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e4ce074d6872a007864059ffa1b509eb07d8d31b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 23:18:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"778bb8cd16140a96309bf65259fab21ab82ad667","revisions":{"c2d1a2cf0ab84ac4d7c60bf053f50c77c1e88896":{"kind":"REWORK","_number":1,"created":"2019-06-02 22:44:02.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/51/662651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/51/662651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/662651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/662651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/662651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/51/662651/1"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 22:42:55.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 22:42:55.000000000","tz":-420},"subject":"Renamed unit.test_api to unit.api","message":"Renamed unit.test_api to unit.api\n\nAdditional changes in a follow up patch, but\nwant to make them smaller so they are easier\nto review.\n\n- Minor cleanups.\n\nChange-Id: I2129d682f5b88e38090b24de827ccb05f78f4782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2d1a2cf0ab84ac4d7c60bf053f50c77c1e88896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2d1a2cf0ab84ac4d7c60bf053f50c77c1e88896"}]},"branch":"refs/heads/master"},"778bb8cd16140a96309bf65259fab21ab82ad667":{"kind":"REWORK","_number":2,"created":"2019-06-03 21:27:53.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/51/662651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/51/662651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/662651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/662651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/662651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/51/662651/2"}}},"commit":{"parents":[{"commit":"7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe","subject":"Merge \"Fix api-ref building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe1c1e9b64efc3ecdbe8af72fd5827b84de02fe"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-02 22:42:55.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-03 21:27:40.000000000","tz":-420},"subject":"Renamed unit.test_api to unit.api","message":"Renamed unit.test_api to unit.api\n\nAdditional changes in a follow up patch, but\nwant to make them smaller so they are easier\nto review.\n\n- Minor cleanups.\n\nChange-Id: I2129d682f5b88e38090b24de827ccb05f78f4782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/778bb8cd16140a96309bf65259fab21ab82ad667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/778bb8cd16140a96309bf65259fab21ab82ad667"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
