)]}'
{"id":"openstack%2Fdesignate~662725","triplet_id":"openstack%2Fdesignate~stable%2Fqueens~I307a8f7dd8b8a83effa447a846db3288efa32dba","project":"openstack/designate","branch":"stable/queens","topic":"bug/1768824","hashtags":[],"change_id":"I307a8f7dd8b8a83effa447a846db3288efa32dba","subject":"Add unique constraint to service_statuses","status":"ABANDONED","created":"2019-06-03 09:35:18.000000000","updated":"2021-12-24 04:28:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f9dee551c6661246ab4428ca18992d3c4936972b","_number":662725,"virtual_id_number":662725,"owner":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-06-04 17:49:20.000000000","_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2019-06-03 10:52:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":-1,"date":"2019-06-04 06:38:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-04 09:38:20.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 10:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-04 06:38:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-06-04 09:38:20.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"8b557a052a18b56d15a1f9d8aba9d3a18fc24001","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-03 09:35:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e98a2cfd21b0b3a91cc950599d67a5b0cedd2cef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 10:52:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-devstack-bind9 http://logs.openstack.org/25/662725/1/check/designate-devstack-bind9/84ac587/ : SUCCESS in 37m 06s\n- designate-devstack-pdns4 http://logs.openstack.org/25/662725/1/check/designate-devstack-pdns4/9b8bb0a/ : SUCCESS in 36m 59s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/25/662725/1/check/designate-devstack-pdns4-postgres/e449b74/ : SUCCESS in 34m 34s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/25/662725/1/check/designate-devstack-pdns4-manager-model/83a7469/ : FAILURE in 40m 27s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/25/662725/1/check/designate-devstack-pdns4-py35/6fc71c0/ : SUCCESS in 37m 49s\n- designate-grenade-pdns4 http://logs.openstack.org/25/662725/1/check/designate-grenade-pdns4/f968ed8/ : SUCCESS in 1h 03m 01s\n- openstack-tox-pep8 http://logs.openstack.org/25/662725/1/check/openstack-tox-pep8/ecbd5cb/ : FAILURE in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/25/662725/1/check/openstack-tox-py27/4c251d1/ : FAILURE in 7m 47s\n- openstack-tox-py35 http://logs.openstack.org/25/662725/1/check/openstack-tox-py35/4e1216a/ : FAILURE in 5m 15s\n- build-openstack-sphinx-docs http://logs.openstack.org/25/662725/1/check/build-openstack-sphinx-docs/505fed4/html/ : SUCCESS in 7m 18s\n- build-openstack-releasenotes http://logs.openstack.org/25/662725/1/check/build-openstack-releasenotes/2d83738/html/ : SUCCESS in 6m 31s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/25/662725/1/check/neutron-tempest-plugin-designate-scenario/49fe998/ : FAILURE in 28m 39s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/25/662725/1/check/openstack-tox-cover/cb81866/ : FAILURE in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e96164cd53b0e40e92c8163bf3b4f62946e16728","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-06-04 06:38:10.000000000","message":"Patch Set 1: Code-Review-1\n\nI think that backporting this DB migration to a stable branch is not advisable since it may fail when the existing data don\u0027t match the constraints.","accounts_in_message":[],"_revision_number":1},{"id":"14871512affc21e761f3674a330e665fa94a94e0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-04 09:38:20.000000000","message":"Patch Set 1:\n\n\u003e I think that backporting this DB migration to a stable branch is\n \u003e not advisable since it may fail when the existing data don\u0027t match\n \u003e the constraints.\n\nI kinda agree, but not sure what is worse. If you hit this bug, it\u0027s dangerous to run Designate, as it will create an infinite (only limited by disk/memory) number of service records. Not sure which is better.","accounts_in_message":[],"_revision_number":1},{"id":"f3d65b5866020e2e705dd171ecde40ed396b26ac","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-04 17:49:20.000000000","message":"Patch Set 1:\n\nAs has been stated, without this patch you can run into the problem of designate killing your database by flooding the service_statuses table. Upgrading to this patch only takes effect if the migration is executed and that will expectedly fail if you are hitting the bug. But isn\u0027t that actually useful? i.e. it serves as a red flag that you have a serious problem that needs resolving and the solution is documented in the LP bug. So personally I am favour of having this backported because whatever the outcome i believe it is advantageous to have this patch.","accounts_in_message":[],"_revision_number":1},{"id":"f9dee551c6661246ab4428ca18992d3c4936972b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:28:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e186c890f72d99a70fa913d093a2016da6bff04e","revisions":{"e186c890f72d99a70fa913d093a2016da6bff04e":{"kind":"REWORK","_number":1,"created":"2019-06-03 09:35:18.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/25/662725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/662725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/662725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/662725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/662725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/662725/1"}}},"commit":{"parents":[{"commit":"d26dbdfe09b6a669bac243809c2adfcea2e92be3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d26dbdfe09b6a669bac243809c2adfcea2e92be3"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2018-05-14 20:50:03.000000000","tz":-420},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2019-06-03 09:35:16.000000000","tz":60},"subject":"Add unique constraint to service_statuses","message":"Add unique constraint to service_statuses\n\nIn the current implementation, if for any reason a duplicate\nservice entry gets created, the call to update that service\nwill fail endlessly, and instead cause the service to create\nnew entries everytime update_service_status gets called. Causing\nit to fill the database with duplicate entries.\n\nThis patch adds a unique constraint to the service_statuses\ntable based on the service_name and hostname, to ensure that\nthis cannot happen.\n\nIn addition we add a new test to the storage driver and further\nexpanded the central service test coverage.\n\nChange-Id: I307a8f7dd8b8a83effa447a846db3288efa32dba\nCloses-Bug: #1768824\n(cherry picked from commit 1924abff40341337bd00659ff01a14ea91701475)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e186c890f72d99a70fa913d093a2016da6bff04e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e186c890f72d99a70fa913d093a2016da6bff04e"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
