)]}'
{"id":"openstack%2Fdesignate~662761","triplet_id":"openstack%2Fdesignate~stable%2Fqueens~Ie0350b034f0eb03749138aadd0951d30073214c0","project":"openstack/designate","branch":"stable/queens","topic":"bug/1768824","hashtags":[],"change_id":"Ie0350b034f0eb03749138aadd0951d30073214c0","subject":"Add designate-status command for upgrade checks","status":"ABANDONED","created":"2019-06-03 14:11:07.000000000","updated":"2021-07-06 10:14:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a02f13b9ad127d804c28cb8940e38a98f92b7817","_number":662761,"virtual_id_number":662761,"owner":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 14:11:07.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-06-03 15:29:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-13 22:52:56.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"d0f57bb1dd255518a8fd195e9f59aa0c8d283e8d","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-03 14:11:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a91f8f457459db0372fd886419b702a99591c3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:29:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/61/662761/1/check/test-release-openstack/a89d5f1/ : SUCCESS in 3m 25s\n- designate-devstack-bind9 http://logs.openstack.org/61/662761/1/check/designate-devstack-bind9/9e70279/ : SUCCESS in 36m 32s\n- designate-devstack-pdns4 http://logs.openstack.org/61/662761/1/check/designate-devstack-pdns4/8f326ad/ : SUCCESS in 33m 07s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/61/662761/1/check/designate-devstack-pdns4-postgres/602c1c3/ : SUCCESS in 40m 53s\n- designate-devstack-pdns4-manager-model http://logs.openstack.org/61/662761/1/check/designate-devstack-pdns4-manager-model/b6882d8/ : SUCCESS in 40m 01s\n- designate-devstack-pdns4-py35 http://logs.openstack.org/61/662761/1/check/designate-devstack-pdns4-py35/e99f193/ : RETRY_LIMIT in 3m 28s\n- designate-grenade-pdns4 http://logs.openstack.org/61/662761/1/check/designate-grenade-pdns4/c535574/ : RETRY_LIMIT in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/61/662761/1/check/openstack-tox-pep8/1bc6011/ : FAILURE in 5m 56s\n- openstack-tox-py27 http://logs.openstack.org/61/662761/1/check/openstack-tox-py27/dbc967f/ : SUCCESS in 6m 09s\n- openstack-tox-py35 http://logs.openstack.org/61/662761/1/check/openstack-tox-py35/aa480ef/ : RETRY_LIMIT in 1m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/662761/1/check/build-openstack-sphinx-docs/c6fe3ff/ : FAILURE in 6m 03s\n- requirements-check http://logs.openstack.org/61/662761/1/check/requirements-check/4f11abb/ : RETRY_LIMIT in 1m 21s\n- build-openstack-releasenotes http://logs.openstack.org/61/662761/1/check/build-openstack-releasenotes/a930e7d/html/ : SUCCESS in 6m 19s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/61/662761/1/check/neutron-tempest-plugin-designate-scenario/1f454a4/ : FAILURE in 27m 54s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/61/662761/1/check/openstack-tox-cover/c6033dc/ : RETRY_LIMIT in 1m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6adb8d3fa0c0650d140bff0356e4685a9e960e5a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-16 15:52:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0e87cb627577b7480d78fa4c4eec2fc7aabc368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 16:54:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8c46fb89a3f94d62a01624a10e909e42 : SUCCESS in 3m 30s\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/4ddc2ec8f1524eb3aac18264c93c0db0 : SUCCESS in 41m 42s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/a2b53a711a99436db59cd733fad9077c : SUCCESS in 36m 18s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/f45f361e9ec44fe9bf72d7303e2a3a35 : SUCCESS in 39m 54s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/03d3c043a796405bb3eccddf7f999cd4 : SUCCESS in 42m 54s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/56c1f9a9d9ec4a3581313bed46094afe : SUCCESS in 42m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/501b1197374243a2a1effb22dfc4746f : SUCCESS in 51m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67889dc1c8441818aa626a69c20b8de : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fd4019a97734e36a878f4b8b563b97a : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9c5a6142d332452883d2ecd637500016 : SUCCESS in 5m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/536eaf21979143d2a522297aa9bf1cc6 : FAILURE in 5m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a1181266f6fb4771a5871039c2197a93 : FAILURE in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/753811c5805c4c1a9ba545e49a38e06b : SUCCESS in 5m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c0e42ec34da94604ace02bde94a56227 : FAILURE in 22m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ca59ac3aece49989a3ae3ec845368df : SUCCESS in 8m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7186f8d5c1e18cdd97d6d1bb7555891d3015bde6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-13 22:52:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"72a143e0de26839e0160fcaee150ff821fc31645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 23:52:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/70a6c0390be5414ca4b8c2bf5bab5b0e : SUCCESS in 2m 54s\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/3335eb76c21748f68b0d8c2cc19f4938 : SUCCESS in 34m 06s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/aa05e23a640a4e8c87dfa742d48b262d : SUCCESS in 31m 35s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/26a42e25a8b24816b0ba4ffae05f37d4 : SUCCESS in 30m 22s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/d1600b53c8764bf39298bf41388b1b2a : SUCCESS in 32m 52s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/7f53a1bc7e5c4c46b2c23a7870f0d7b3 : FAILURE in 36m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/293db5c4550f4aee8a44f199482f64d1 : FAILURE in 49m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e23b0ade334da096840030e5e86e13 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5cde06711a974d799e2861b67f1fe107 : SUCCESS in 5m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/87f13f1b76e3425a95c0b08adbbb3253 : SUCCESS in 6m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/060f0a6d98a64460ab38913e02ee050a : FAILURE in 6m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/64eef1d387e74b57862eba1f227179da : FAILURE in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac20f27841884ed68d044fb16a406734 : SUCCESS in 4m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9c6a905250d40ada7cad18aab641463 : FAILURE in 18m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f24bc041a52477382fbd8edbc5df620 : SUCCESS in 7m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"feea67345b8c63213f2170c0cd898d7352d4eae9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-07-06 10:13:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a02f13b9ad127d804c28cb8940e38a98f92b7817","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-07-06 10:14:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bcfb6fff3d1bd0322fb0db40f8e3a35f4d6efd58","revisions":{"194f9b486b00a5b2758048cb498ead3145b13f45":{"kind":"REWORK","_number":1,"created":"2019-06-03 14:11:07.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/61/662761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/662761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/662761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/662761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/662761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/662761/1"}}},"commit":{"parents":[{"commit":"d26dbdfe09b6a669bac243809c2adfcea2e92be3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d26dbdfe09b6a669bac243809c2adfcea2e92be3"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2018-09-19 20:34:26.000000000","tz":0},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2019-06-03 14:10:57.000000000","tz":60},"subject":"Add designate-status command for upgrade checks","message":"Add designate-status command for upgrade checks\n\nThis checks for duplicate entries in the service_statuses table\nwhich will be an error after change\nI307a8f7dd8b8a83effa447a846db3288efa32dba.\n\nRelated-Bug: 1768824\n\nStory: 2003657\nTask: 26127\n\nChange-Id: Ie0350b034f0eb03749138aadd0951d30073214c0\nCo-authored-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n(cherry picked from commit 1842802de4a05a057eb7f3c94bd01a4b1f0b9143)\n(cherry picked from commit cc38d9625ffd87f1b717ccb2aa6a0159f71b08a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/194f9b486b00a5b2758048cb498ead3145b13f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/194f9b486b00a5b2758048cb498ead3145b13f45"}]},"branch":"refs/heads/stable/queens"},"bcfb6fff3d1bd0322fb0db40f8e3a35f4d6efd58":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-06 10:13:10.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/61/662761/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/662761/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/662761/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/662761/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/662761/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/662761/2"}}},"commit":{"parents":[{"commit":"076600dd9107baf567c3dc032a0901dfeda73784","subject":"Worker should send NOTIFY also to all servers in \u0027also_notifies\u0027 pool settings.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/076600dd9107baf567c3dc032a0901dfeda73784"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2018-09-19 20:34:26.000000000","tz":0},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2021-07-06 10:13:10.000000000","tz":0},"subject":"Add designate-status command for upgrade checks","message":"Add designate-status command for upgrade checks\n\nThis checks for duplicate entries in the service_statuses table\nwhich will be an error after change\nI307a8f7dd8b8a83effa447a846db3288efa32dba.\n\nRelated-Bug: 1768824\n\nStory: 2003657\nTask: 26127\n\nChange-Id: Ie0350b034f0eb03749138aadd0951d30073214c0\nCo-authored-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n(cherry picked from commit 1842802de4a05a057eb7f3c94bd01a4b1f0b9143)\n(cherry picked from commit cc38d9625ffd87f1b717ccb2aa6a0159f71b08a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bcfb6fff3d1bd0322fb0db40f8e3a35f4d6efd58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bcfb6fff3d1bd0322fb0db40f8e3a35f4d6efd58"}]},"branch":"refs/heads/stable/queens","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
