)]}'
{"id":"openstack%2Fdesignate~664167","triplet_id":"openstack%2Fdesignate~master~I27d2774bf89fefcbf2015dd1a92fd96c7a2057eb","project":"openstack/designate","branch":"master","topic":"improved_tests","hashtags":[],"change_id":"I27d2774bf89fefcbf2015dd1a92fd96c7a2057eb","subject":"Improving test coverage of designate backend","status":"MERGED","created":"2019-06-09 23:07:31.000000000","updated":"2019-06-18 13:57:55.000000000","submitted":"2019-06-18 13:57:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664167-1560866275378-44d5107d","meta_rev_id":"14c8f348bbcfa7ab466c9d1de5687f5276cf7d90","_number":664167,"virtual_id_number":664167,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":2,"date":"2019-06-18 13:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-18 11:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-06-18 00:32:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-06-18 11:58:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 00:32:24.000000000","updated_by":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"reviewer":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"state":"REVIEWER"},{"updated":"2019-06-18 11:58:52.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-18 13:57:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f311726639e6ee09ebf6af6026096ad318fdc948","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-09 23:07:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"572e4c5a09ee3bbb3e9afd749562efadce006750","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-09 23:43:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfb2388622888f4818db7972f49fe04fee14bee9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 00:30:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/67/664167/2/check/designate-bind9/86c4ae0/ : SUCCESS in 34m 58s\n- designate-pdns4 http://logs.openstack.org/67/664167/2/check/designate-pdns4/515ff12/ : SUCCESS in 34m 55s\n- designate-bind9-manager-model http://logs.openstack.org/67/664167/2/check/designate-bind9-manager-model/e82fdcb/ : SUCCESS in 36m 18s\n- designate-bind9-py27 http://logs.openstack.org/67/664167/2/check/designate-bind9-py27/68c36cf/ : SUCCESS in 34m 12s\n- designate-pdns4-py27 http://logs.openstack.org/67/664167/2/check/designate-pdns4-py27/a8eb472/ : SUCCESS in 33m 02s\n- designate-pdns4-postgres http://logs.openstack.org/67/664167/2/check/designate-pdns4-postgres/80c9593/ : SUCCESS in 35m 16s\n- designate-grenade-pdns4 http://logs.openstack.org/67/664167/2/check/designate-grenade-pdns4/09cc1d0/ : SUCCESS in 45m 07s\n- openstack-tox-cover http://logs.openstack.org/67/664167/2/check/openstack-tox-cover/b26bff9/cover/ : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/664167/2/check/openstack-tox-lower-constraints/2c19385/ : SUCCESS in 6m 50s\n- openstack-tox-pep8 http://logs.openstack.org/67/664167/2/check/openstack-tox-pep8/a72bfd5/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/67/664167/2/check/openstack-tox-py27/a92ebbd/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/67/664167/2/check/openstack-tox-py36/74d449b/ : SUCCESS in 5m 43s\n- openstack-tox-py37 http://logs.openstack.org/67/664167/2/check/openstack-tox-py37/61286cc/ : SUCCESS in 6m 03s\n- openstack-tox-docs http://logs.openstack.org/67/664167/2/check/openstack-tox-docs/58c738b/html/ : SUCCESS in 5m 07s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/67/664167/2/check/neutron-tempest-plugin-designate-scenario/a893ab5/ : SUCCESS in 31m 22s","accounts_in_message":[],"_revision_number":2},{"id":"dafe0c328146ad4bbcf30e9bb91b783f05d34e65","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-10 18:18:31.000000000","message":"Topic changed from designate_backend_tests to improved_tests","accounts_in_message":[],"_revision_number":2},{"id":"b9fd608bc536599c847db4bcd1c64a65ae606088","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 17:00:20.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c1e4d1098991c7083eb185b30d2f7368d98c844a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:09:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/67/664167/3/check/designate-bind9/642c322/ : SUCCESS in 37m 23s\n- designate-pdns4 http://logs.openstack.org/67/664167/3/check/designate-pdns4/a80c3df/ : SUCCESS in 37m 02s\n- designate-bind9-manager-model http://logs.openstack.org/67/664167/3/check/designate-bind9-manager-model/df5e7e2/ : SUCCESS in 41m 49s\n- designate-bind9-py27 http://logs.openstack.org/67/664167/3/check/designate-bind9-py27/e29cf7b/ : SUCCESS in 40m 55s\n- designate-pdns4-py27 http://logs.openstack.org/67/664167/3/check/designate-pdns4-py27/4e57c8e/ : SUCCESS in 40m 21s\n- designate-pdns4-postgres http://logs.openstack.org/67/664167/3/check/designate-pdns4-postgres/f163751/ : SUCCESS in 38m 19s\n- designate-grenade-pdns4 http://logs.openstack.org/67/664167/3/check/designate-grenade-pdns4/0c4e8b1/ : SUCCESS in 1h 00m 06s\n- openstack-tox-cover http://logs.openstack.org/67/664167/3/check/openstack-tox-cover/d391757/cover/ : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/664167/3/check/openstack-tox-lower-constraints/eb51781/ : SUCCESS in 7m 29s\n- openstack-tox-pep8 http://logs.openstack.org/67/664167/3/check/openstack-tox-pep8/15e9d1b/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/67/664167/3/check/openstack-tox-py27/ef62577/ : SUCCESS in 7m 43s\n- openstack-tox-py36 http://logs.openstack.org/67/664167/3/check/openstack-tox-py36/2753df9/ : SUCCESS in 5m 42s\n- openstack-tox-py37 http://logs.openstack.org/67/664167/3/check/openstack-tox-py37/17b654f/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/67/664167/3/check/openstack-tox-docs/c4dc8c2/html/ : SUCCESS in 4m 36s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/67/664167/3/check/neutron-tempest-plugin-designate-scenario/0889ed3/ : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":3},{"id":"b049bab04f9bc6cb3b1d2a5ccfab374f1643c7bf","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-06-18 00:32:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a5caeada13cfd443a0748429e377e5b51a9b5268","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-18 11:58:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6d314bfca64ee076f9d846786dc84ba92c637ddc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 11:59:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"570dfb76fff426c480b24a390f033660a3a00be2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 13:57:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/67/664167/3/gate/designate-bind9/cffeed1/ : SUCCESS in 40m 45s\n- designate-pdns4 http://logs.openstack.org/67/664167/3/gate/designate-pdns4/715beb4/ : SUCCESS in 42m 28s\n- designate-bind9-manager-model http://logs.openstack.org/67/664167/3/gate/designate-bind9-manager-model/dffb4f6/ : SUCCESS in 40m 10s\n- designate-bind9-py27 http://logs.openstack.org/67/664167/3/gate/designate-bind9-py27/a4c3b4f/ : SUCCESS in 49m 07s\n- designate-pdns4-py27 http://logs.openstack.org/67/664167/3/gate/designate-pdns4-py27/52af6c4/ : SUCCESS in 31m 39s\n- designate-pdns4-postgres http://logs.openstack.org/67/664167/3/gate/designate-pdns4-postgres/fea58e2/ : SUCCESS in 38m 53s\n- designate-grenade-pdns4 http://logs.openstack.org/67/664167/3/gate/designate-grenade-pdns4/0198281/ : SUCCESS in 1h 06m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/664167/3/gate/openstack-tox-lower-constraints/2655219/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/67/664167/3/gate/openstack-tox-pep8/be25719/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/67/664167/3/gate/openstack-tox-py27/34ddeee/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/67/664167/3/gate/openstack-tox-py36/1969cc3/ : SUCCESS in 5m 17s\n- openstack-tox-py37 http://logs.openstack.org/67/664167/3/gate/openstack-tox-py37/88a18ee/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/67/664167/3/gate/openstack-tox-docs/dd1ce60/html/ : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":3},{"id":"14c8f348bbcfa7ab466c9d1de5687f5276cf7d90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 13:57:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eb2443603bbceec1c3a2f2371e66d734da046d9e","revisions":{"40ccc1539a58a0a28a5caac620f44156f387b7b5":{"kind":"REWORK","_number":1,"created":"2019-06-09 23:07:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/67/664167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/67/664167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/664167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/664167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/664167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/67/664167/1"}}},"commit":{"parents":[{"commit":"e584c41ec89d65a7c8b4eac841f1d03be573847a","subject":"Updated context to use project_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e584c41ec89d65a7c8b4eac841f1d03be573847a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:07:08.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:07:08.000000000","tz":-420},"subject":"Improving test coverage of designate backend","message":"Improving test coverage of designate backend\n\nChange-Id: I27d2774bf89fefcbf2015dd1a92fd96c7a2057eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/40ccc1539a58a0a28a5caac620f44156f387b7b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/40ccc1539a58a0a28a5caac620f44156f387b7b5"}]},"branch":"refs/heads/master"},"8ea44b74702a32c9f9b4c423e9e1ebbc6a9439dc":{"kind":"REWORK","_number":2,"created":"2019-06-09 23:43:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/67/664167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/67/664167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/664167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/664167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/664167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/67/664167/2"}}},"commit":{"parents":[{"commit":"e584c41ec89d65a7c8b4eac841f1d03be573847a","subject":"Updated context to use project_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e584c41ec89d65a7c8b4eac841f1d03be573847a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:07:08.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:43:05.000000000","tz":-420},"subject":"Improving test coverage of designate backend","message":"Improving test coverage of designate backend\n\nChange-Id: I27d2774bf89fefcbf2015dd1a92fd96c7a2057eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8ea44b74702a32c9f9b4c423e9e1ebbc6a9439dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8ea44b74702a32c9f9b4c423e9e1ebbc6a9439dc"}]},"branch":"refs/heads/master"},"eb2443603bbceec1c3a2f2371e66d734da046d9e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-17 17:00:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/67/664167/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/67/664167/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/664167/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/664167/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/67/664167/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/67/664167/3"}}},"commit":{"parents":[{"commit":"ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6","subject":"Merge \"Fixed renaming topic breaking service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:07:08.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-17 17:00:06.000000000","tz":-420},"subject":"Improving test coverage of designate backend","message":"Improving test coverage of designate backend\n\nChange-Id: I27d2774bf89fefcbf2015dd1a92fd96c7a2057eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb2443603bbceec1c3a2f2371e66d734da046d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb2443603bbceec1c3a2f2371e66d734da046d9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
