)]}'
{"id":"openstack%2Fdesignate~664173","triplet_id":"openstack%2Fdesignate~master~I9af77adb8a0b5c2ee076eeffe993100d8c65196c","project":"openstack/designate","branch":"master","topic":"improved_tests","hashtags":[],"change_id":"I9af77adb8a0b5c2ee076eeffe993100d8c65196c","subject":"Remove unused code in utils and improved utils testing","status":"MERGED","created":"2019-06-09 23:41:26.000000000","updated":"2019-07-02 04:25:51.000000000","submitted":"2019-07-02 04:25:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664173-1562041551804-40190feb","meta_rev_id":"175151f9e0207c36a7e6a7a68f2de85696c70a6d","_number":664173,"virtual_id_number":664173,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":2,"date":"2019-07-02 04:25:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-07-01 15:25:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-06-28 18:34:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-02 03:20:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 18:34:29.000000000","updated_by":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"reviewer":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"state":"REVIEWER"},{"updated":"2019-07-01 15:25:37.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-07-02 04:25:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6797f25454084d818cf3bffd8a5a7c588924801d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-09 23:41:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85890f43cac3fbf872e8eddf409c1ccd28ea7d88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 00:46:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/73/664173/1/check/designate-bind9/cd68590/ : SUCCESS in 40m 21s\n- designate-pdns4 http://logs.openstack.org/73/664173/1/check/designate-pdns4/064cfe0/ : SUCCESS in 38m 36s\n- designate-bind9-manager-model http://logs.openstack.org/73/664173/1/check/designate-bind9-manager-model/2472efa/ : SUCCESS in 36m 54s\n- designate-bind9-py27 http://logs.openstack.org/73/664173/1/check/designate-bind9-py27/cdca021/ : SUCCESS in 42m 56s\n- designate-pdns4-py27 http://logs.openstack.org/73/664173/1/check/designate-pdns4-py27/3d73af1/ : SUCCESS in 36m 49s\n- designate-pdns4-postgres http://logs.openstack.org/73/664173/1/check/designate-pdns4-postgres/55d1d83/ : SUCCESS in 38m 55s\n- designate-grenade-pdns4 http://logs.openstack.org/73/664173/1/check/designate-grenade-pdns4/aab1c55/ : SUCCESS in 59m 12s\n- openstack-tox-cover http://logs.openstack.org/73/664173/1/check/openstack-tox-cover/74661ac/cover/ : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/664173/1/check/openstack-tox-lower-constraints/95336fa/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/73/664173/1/check/openstack-tox-pep8/1321960/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/73/664173/1/check/openstack-tox-py27/53883e3/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/73/664173/1/check/openstack-tox-py36/48cff0a/ : SUCCESS in 5m 30s\n- openstack-tox-py37 http://logs.openstack.org/73/664173/1/check/openstack-tox-py37/62a68e3/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/73/664173/1/check/openstack-tox-docs/8c01057/html/ : SUCCESS in 4m 26s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/73/664173/1/check/neutron-tempest-plugin-designate-scenario/ff48b57/ : SUCCESS in 33m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bb593c2fd10230746fe954f149d48941b8163551","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-10 18:17:01.000000000","message":"Topic set to improved_tests","accounts_in_message":[],"_revision_number":1},{"id":"6fd10476bb1e22f312556f10ed541a57cbfaf36e","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-06-18 01:00:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ec645dc5a4403b7e6f9f86247dcaa6df7187d1e1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 19:51:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"82607ef8e17820b48543f47dcf654d10ff407d8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:09:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/73/664173/2/check/designate-bind9/258b313/ : SUCCESS in 40m 07s\n- designate-pdns4 http://logs.openstack.org/73/664173/2/check/designate-pdns4/8ff627f/ : SUCCESS in 41m 03s\n- designate-bind9-manager-model http://logs.openstack.org/73/664173/2/check/designate-bind9-manager-model/735abf5/ : SUCCESS in 45m 33s\n- designate-bind9-py27 http://logs.openstack.org/73/664173/2/check/designate-bind9-py27/6620a5a/ : SUCCESS in 1h 09m 19s\n- designate-pdns4-py27 http://logs.openstack.org/73/664173/2/check/designate-pdns4-py27/0a4a332/ : SUCCESS in 31m 55s\n- designate-pdns4-postgres http://logs.openstack.org/73/664173/2/check/designate-pdns4-postgres/ba0f4e4/ : SUCCESS in 34m 09s\n- designate-grenade-pdns4 http://logs.openstack.org/73/664173/2/check/designate-grenade-pdns4/56f8af5/ : SUCCESS in 58m 51s\n- openstack-tox-cover http://logs.openstack.org/73/664173/2/check/openstack-tox-cover/def71ef/cover/ : SUCCESS in 7m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/664173/2/check/openstack-tox-lower-constraints/ef2eeae/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/73/664173/2/check/openstack-tox-pep8/b6b7ba0/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/73/664173/2/check/openstack-tox-py27/0bf0c65/ : SUCCESS in 7m 35s\n- openstack-tox-py36 http://logs.openstack.org/73/664173/2/check/openstack-tox-py36/f7b8e5c/ : SUCCESS in 9m 01s\n- openstack-tox-py37 http://logs.openstack.org/73/664173/2/check/openstack-tox-py37/3618aea/ : SUCCESS in 5m 59s\n- openstack-tox-docs http://logs.openstack.org/73/664173/2/check/openstack-tox-docs/bb7b15b/html/ : SUCCESS in 4m 59s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/73/664173/2/check/neutron-tempest-plugin-designate-scenario/c09df04/ : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":2},{"id":"59b9f38e23a5cdfc2a3e50ae4816784286680027","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-24 00:56:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b87ffe11304e589d7f90e95e3da9a3a596d8b789","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 02:05:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/73/664173/3/check/designate-bind9/2d983be/ : SUCCESS in 38m 15s\n- designate-pdns4 http://logs.openstack.org/73/664173/3/check/designate-pdns4/899fac9/ : SUCCESS in 41m 48s\n- designate-bind9-manager-model http://logs.openstack.org/73/664173/3/check/designate-bind9-manager-model/70a6223/ : SUCCESS in 37m 56s\n- designate-bind9-py27 http://logs.openstack.org/73/664173/3/check/designate-bind9-py27/f4125c1/ : SUCCESS in 42m 24s\n- designate-pdns4-py27 http://logs.openstack.org/73/664173/3/check/designate-pdns4-py27/da6042a/ : SUCCESS in 43m 40s\n- designate-pdns4-postgres http://logs.openstack.org/73/664173/3/check/designate-pdns4-postgres/9502be9/ : SUCCESS in 36m 35s\n- designate-grenade-pdns4 http://logs.openstack.org/73/664173/3/check/designate-grenade-pdns4/e0b68d2/ : SUCCESS in 57m 04s\n- openstack-tox-cover http://logs.openstack.org/73/664173/3/check/openstack-tox-cover/2ad1e5f/cover/ : SUCCESS in 6m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/664173/3/check/openstack-tox-lower-constraints/5e8e43a/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/73/664173/3/check/openstack-tox-pep8/7ec8de0/ : SUCCESS in 5m 44s\n- openstack-tox-py27 http://logs.openstack.org/73/664173/3/check/openstack-tox-py27/6dc071f/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/73/664173/3/check/openstack-tox-py36/d92e375/ : SUCCESS in 5m 49s\n- openstack-tox-py37 http://logs.openstack.org/73/664173/3/check/openstack-tox-py37/ec0c5a0/ : SUCCESS in 7m 47s\n- openstack-tox-docs http://logs.openstack.org/73/664173/3/check/openstack-tox-docs/6f173dd/html/ : SUCCESS in 5m 09s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/73/664173/3/check/neutron-tempest-plugin-designate-scenario/06c7638/ : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":3},{"id":"b853764332d6ba293cbe77118fa9fdc7b94da009","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-27 18:13:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7be88370de6eb52b93b00486a3b45425de15985c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 19:34:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/73/664173/4/check/designate-bind9/58d0b5c/ : SUCCESS in 44m 07s\n- designate-pdns4 http://logs.openstack.org/73/664173/4/check/designate-pdns4/39f2298/ : SUCCESS in 38m 33s\n- designate-bind9-manager-model http://logs.openstack.org/73/664173/4/check/designate-bind9-manager-model/c09f964/ : SUCCESS in 46m 29s\n- designate-bind9-py27 http://logs.openstack.org/73/664173/4/check/designate-bind9-py27/12496c4/ : SUCCESS in 41m 47s\n- designate-pdns4-py27 http://logs.openstack.org/73/664173/4/check/designate-pdns4-py27/6ed152a/ : SUCCESS in 36m 25s\n- designate-pdns4-postgres http://logs.openstack.org/73/664173/4/check/designate-pdns4-postgres/a903e3f/ : SUCCESS in 36m 51s\n- designate-grenade-pdns4 http://logs.openstack.org/73/664173/4/check/designate-grenade-pdns4/6a4c315/ : SUCCESS in 1h 01m 12s\n- openstack-tox-cover http://logs.openstack.org/73/664173/4/check/openstack-tox-cover/a112dca/cover/ : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/664173/4/check/openstack-tox-lower-constraints/bc39fd8/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/73/664173/4/check/openstack-tox-pep8/7a91c59/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/73/664173/4/check/openstack-tox-py27/96bf1e7/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/73/664173/4/check/openstack-tox-py36/cf3b6a3/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/73/664173/4/check/openstack-tox-py37/7676d1c/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/73/664173/4/check/openstack-tox-docs/c784fc7/html/ : SUCCESS in 3m 26s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/73/664173/4/check/neutron-tempest-plugin-designate-scenario/1e099f8/ : SUCCESS in 38m 07s","accounts_in_message":[],"_revision_number":4},{"id":"385f5e515358b784d9ed90904744a8f9d89d86fc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 03:28:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"581cb50c61d194838a9fef220560f0e8b4ff1664","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 03:35:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1ed446f3d3835065f3b111fb6ed568e9cb162a44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 04:37:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/73/664173/6/check/designate-bind9/a09fad0/ : SUCCESS in 33m 18s\n- designate-pdns4 http://logs.openstack.org/73/664173/6/check/designate-pdns4/ce5d57a/ : SUCCESS in 39m 00s\n- designate-bind9-manager-model http://logs.openstack.org/73/664173/6/check/designate-bind9-manager-model/35c0e6b/ : SUCCESS in 35m 27s\n- designate-bind9-py27 http://logs.openstack.org/73/664173/6/check/designate-bind9-py27/f62fcc0/ : SUCCESS in 41m 59s\n- designate-pdns4-py27 http://logs.openstack.org/73/664173/6/check/designate-pdns4-py27/e6d0d5d/ : SUCCESS in 37m 32s\n- designate-pdns4-postgres http://logs.openstack.org/73/664173/6/check/designate-pdns4-postgres/01113d1/ : SUCCESS in 37m 59s\n- designate-grenade-pdns4 http://logs.openstack.org/73/664173/6/check/designate-grenade-pdns4/1b6d161/ : SUCCESS in 56m 19s\n- openstack-tox-cover http://logs.openstack.org/73/664173/6/check/openstack-tox-cover/0a82ed3/cover/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/664173/6/check/openstack-tox-lower-constraints/016ee62/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/73/664173/6/check/openstack-tox-pep8/96a029b/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/73/664173/6/check/openstack-tox-py27/2d37240/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/73/664173/6/check/openstack-tox-py36/5d6f053/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/73/664173/6/check/openstack-tox-py37/5dfe763/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/73/664173/6/check/openstack-tox-docs/a341975/html/ : SUCCESS in 3m 46s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/73/664173/6/check/neutron-tempest-plugin-designate-scenario/f1fcfc2/ : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":6},{"id":"350cd304813bf82d075b6d52819e6a0fcc1b0a1b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 04:54:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"887d2c80b71ee3827a2bc7595b1b113329218623","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 04:57:56.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"89d86e15f36e2281b43c45c57b5d8a7f6a281e4c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 04:59:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"533d4a5572d05124c585b9c8d16729a897658f3e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 05:10:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"33e26a0b4dbc5e13cc3348088e2e33c5a306ee09","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 05:50:05.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"391d1032d4cc52a8d99270982ae54bd424df83d4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 05:56:54.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"59db23da06329480d1b982604a6d25661100583f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 06:54:05.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/73/664173/12/check/designate-bind9/4bcc81b/ : SUCCESS in 40m 20s\n- designate-pdns4 http://logs.openstack.org/73/664173/12/check/designate-pdns4/29908b8/ : SUCCESS in 38m 49s\n- designate-bind9-manager-model http://logs.openstack.org/73/664173/12/check/designate-bind9-manager-model/06f527a/ : SUCCESS in 37m 58s\n- designate-bind9-py27 http://logs.openstack.org/73/664173/12/check/designate-bind9-py27/d4813d2/ : SUCCESS in 43m 07s\n- designate-pdns4-py27 http://logs.openstack.org/73/664173/12/check/designate-pdns4-py27/909de7d/ : SUCCESS in 36m 17s\n- designate-pdns4-postgres http://logs.openstack.org/73/664173/12/check/designate-pdns4-postgres/b167662/ : SUCCESS in 36m 46s\n- designate-grenade-pdns4 http://logs.openstack.org/73/664173/12/check/designate-grenade-pdns4/4831793/ : SUCCESS in 52m 39s\n- openstack-tox-cover http://logs.openstack.org/73/664173/12/check/openstack-tox-cover/c625096/cover/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/664173/12/check/openstack-tox-lower-constraints/7bf7ae2/ : SUCCESS in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/73/664173/12/check/openstack-tox-pep8/292a86a/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/73/664173/12/check/openstack-tox-py27/f5a8f57/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/73/664173/12/check/openstack-tox-py36/b525094/ : SUCCESS in 4m 52s\n- openstack-tox-py37 http://logs.openstack.org/73/664173/12/check/openstack-tox-py37/aa2ee52/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/73/664173/12/check/openstack-tox-docs/6c9d55a/html/ : SUCCESS in 3m 17s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/73/664173/12/check/neutron-tempest-plugin-designate-scenario/25e18c8/ : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":12},{"id":"3d7a29be4f4563eacd05c57786fb7db1dfe6c6d9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 17:26:07.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4d7ff2228f571db82643eeaf59fd952c30d1b08f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 17:29:30.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d52b4cbc9f6b8f69ee3c114482b9a7135e8acc8b","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-06-28 18:34:29.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"79708a6fce2730b848e1b6d0fb42feb504a63c17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 18:36:44.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/73/664173/14/check/designate-bind9/505f401/ : SUCCESS in 44m 56s\n- designate-pdns4 http://logs.openstack.org/73/664173/14/check/designate-pdns4/494ce94/ : SUCCESS in 37m 35s\n- designate-bind9-manager-model http://logs.openstack.org/73/664173/14/check/designate-bind9-manager-model/cb0ecff/ : SUCCESS in 39m 50s\n- designate-bind9-py27 http://logs.openstack.org/73/664173/14/check/designate-bind9-py27/0568462/ : SUCCESS in 42m 18s\n- designate-pdns4-py27 http://logs.openstack.org/73/664173/14/check/designate-pdns4-py27/17e60e8/ : SUCCESS in 38m 35s\n- designate-pdns4-postgres http://logs.openstack.org/73/664173/14/check/designate-pdns4-postgres/5405638/ : SUCCESS in 37m 45s\n- designate-grenade-pdns4 http://logs.openstack.org/73/664173/14/check/designate-grenade-pdns4/6ca6b78/ : SUCCESS in 59m 34s\n- openstack-tox-cover http://logs.openstack.org/73/664173/14/check/openstack-tox-cover/7a8c81b/ : FAILURE in 5m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/664173/14/check/openstack-tox-lower-constraints/9fa1a7e/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/73/664173/14/check/openstack-tox-pep8/547be1a/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/73/664173/14/check/openstack-tox-py27/cfc7b34/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/73/664173/14/check/openstack-tox-py36/bd05ccb/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/73/664173/14/check/openstack-tox-py37/81b4c42/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/73/664173/14/check/openstack-tox-docs/98671a6/html/ : SUCCESS in 4m 21s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/73/664173/14/check/neutron-tempest-plugin-designate-scenario/60654be/ : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":14},{"id":"6b3970b60b78393074fd507c5c1960808782a17e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-28 18:38:49.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"5fb14f0f6a5cf771603d1991b2e1f049274029b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 19:47:56.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/73/664173/14/check/designate-bind9/5150475/ : SUCCESS in 36m 42s\n- designate-pdns4 http://logs.openstack.org/73/664173/14/check/designate-pdns4/84e494f/ : SUCCESS in 33m 35s\n- designate-bind9-manager-model http://logs.openstack.org/73/664173/14/check/designate-bind9-manager-model/1c0b2a5/ : SUCCESS in 39m 33s\n- designate-bind9-py27 http://logs.openstack.org/73/664173/14/check/designate-bind9-py27/0ddc6a3/ : SUCCESS in 35m 53s\n- designate-pdns4-py27 http://logs.openstack.org/73/664173/14/check/designate-pdns4-py27/8fb2a81/ : SUCCESS in 34m 26s\n- designate-pdns4-postgres http://logs.openstack.org/73/664173/14/check/designate-pdns4-postgres/6870377/ : SUCCESS in 37m 20s\n- designate-grenade-pdns4 http://logs.openstack.org/73/664173/14/check/designate-grenade-pdns4/703c3f6/ : SUCCESS in 1h 02m 30s\n- openstack-tox-cover http://logs.openstack.org/73/664173/14/check/openstack-tox-cover/46dc797/cover/ : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/664173/14/check/openstack-tox-lower-constraints/6cf2fe4/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/73/664173/14/check/openstack-tox-pep8/a6ee8fb/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/73/664173/14/check/openstack-tox-py27/7ad0342/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/73/664173/14/check/openstack-tox-py36/90f5f41/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/73/664173/14/check/openstack-tox-py37/64f6eb9/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/73/664173/14/check/openstack-tox-docs/1ddd212/html/ : SUCCESS in 3m 32s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/73/664173/14/check/neutron-tempest-plugin-designate-scenario/bf87565/ : SUCCESS in 30m 18s","accounts_in_message":[],"_revision_number":14},{"id":"10793d8c3ce9c06143b9e8ee8bbf633318a6df89","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-07-01 15:25:37.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"c34aae7dde43f8d8d59f6a87e05a049668dc7ea2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-02 03:20:07.000000000","message":"Patch Set 14: Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"49f9222430456023876de24941bc904b12aba59d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 03:20:19.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"4908df6cc0adc98217e2b8534fd0cb95b1f831a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 04:25:51.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/73/664173/14/gate/designate-bind9/d684d5e/ : SUCCESS in 44m 54s\n- designate-pdns4 http://logs.openstack.org/73/664173/14/gate/designate-pdns4/4d03bc2/ : SUCCESS in 38m 24s\n- designate-bind9-manager-model http://logs.openstack.org/73/664173/14/gate/designate-bind9-manager-model/374267f/ : SUCCESS in 40m 24s\n- designate-bind9-py27 http://logs.openstack.org/73/664173/14/gate/designate-bind9-py27/57cb824/ : SUCCESS in 45m 41s\n- designate-pdns4-py27 http://logs.openstack.org/73/664173/14/gate/designate-pdns4-py27/c7ff64e/ : SUCCESS in 36m 16s\n- designate-pdns4-postgres http://logs.openstack.org/73/664173/14/gate/designate-pdns4-postgres/6540a5b/ : SUCCESS in 42m 55s\n- designate-grenade-pdns4 http://logs.openstack.org/73/664173/14/gate/designate-grenade-pdns4/5c64326/ : SUCCESS in 1h 00m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/664173/14/gate/openstack-tox-lower-constraints/2bd6272/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/73/664173/14/gate/openstack-tox-pep8/cdafedf/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/73/664173/14/gate/openstack-tox-py27/789926e/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/73/664173/14/gate/openstack-tox-py36/8a7b841/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/73/664173/14/gate/openstack-tox-py37/6695cdf/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/73/664173/14/gate/openstack-tox-docs/dca2bc8/html/ : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":14},{"id":"175151f9e0207c36a7e6a7a68f2de85696c70a6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 04:25:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"ee04eaf2af1f0a39df2feadac8b882b26414a714","revisions":{"236a2ccf30c1d09c20743018a146ebce5bf33642":{"kind":"REWORK","_number":1,"created":"2019-06-09 23:41:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/1"}}},"commit":{"parents":[{"commit":"174365947394ef304a3aa729658765bc3d36eabf","subject":"Merge \"Remove KS v2 support and add region support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/174365947394ef304a3aa729658765bc3d36eabf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:40:42.000000000","tz":-420},"subject":"Remove unused code in utils","message":"Remove unused code in utils\n\n* Removed utility code not used anywhere.\n* Added some additional tests on previously untested code.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/236a2ccf30c1d09c20743018a146ebce5bf33642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/236a2ccf30c1d09c20743018a146ebce5bf33642"}]},"branch":"refs/heads/master"},"870e51fb7630e4aa44f74d54bdc8f7f0b81f1c2c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-18 19:51:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/2"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 19:51:34.000000000","tz":0},"subject":"Remove unused code in utils","message":"Remove unused code in utils\n\n* Removed utility code not used anywhere.\n* Added some additional tests on previously untested code.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/870e51fb7630e4aa44f74d54bdc8f7f0b81f1c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/870e51fb7630e4aa44f74d54bdc8f7f0b81f1c2c"}]},"branch":"refs/heads/master"},"572b8312661ff1a081cc232ffc3e62e0e4fea8be":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-24 00:56:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/3"}}},"commit":{"parents":[{"commit":"a1d2f371093a90d0bbbbf1b3bcba04ebeb1a92a4","subject":"Merge \"Only register dynamic pool options when running designate-manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a1d2f371093a90d0bbbbf1b3bcba04ebeb1a92a4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-24 00:56:50.000000000","tz":0},"subject":"Remove unused code in utils","message":"Remove unused code in utils\n\n* Removed utility code not used anywhere.\n* Added some additional tests on previously untested code.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/572b8312661ff1a081cc232ffc3e62e0e4fea8be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/572b8312661ff1a081cc232ffc3e62e0e4fea8be"}]},"branch":"refs/heads/master"},"873cffcaf89fe4262501aa3aadb50fafbd1b0c2b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-27 18:13:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/4"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-27 18:13:58.000000000","tz":0},"subject":"Remove unused code in utils","message":"Remove unused code in utils\n\n* Removed utility code not used anywhere.\n* Added some additional tests on previously untested code.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/873cffcaf89fe4262501aa3aadb50fafbd1b0c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/873cffcaf89fe4262501aa3aadb50fafbd1b0c2b"}]},"branch":"refs/heads/master"},"e4f008ffceb8ccea3e8421144ec26f0af25a1a57":{"kind":"REWORK","_number":5,"created":"2019-06-28 03:28:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/5"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-28 03:28:00.000000000","tz":-420},"subject":"Remove unused code in utils","message":"Remove unused code in utils\n\n* Removed utility code not used anywhere.\n* Added some additional tests on previously untested code.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e4f008ffceb8ccea3e8421144ec26f0af25a1a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e4f008ffceb8ccea3e8421144ec26f0af25a1a57"}]},"branch":"refs/heads/master"},"4c3a5564dc2926b10b0fc371880c7886bbeecebc":{"kind":"REWORK","_number":6,"created":"2019-06-28 03:35:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/6"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-28 03:35:33.000000000","tz":-420},"subject":"Remove unused code in utils","message":"Remove unused code in utils\n\n* Removed utility code not used anywhere.\n* Added some additional tests on previously untested code.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4c3a5564dc2926b10b0fc371880c7886bbeecebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4c3a5564dc2926b10b0fc371880c7886bbeecebc"}]},"branch":"refs/heads/master"},"3f5a411fdd77cd75f2bf37fbb26861c35358e013":{"kind":"REWORK","_number":7,"created":"2019-06-28 04:54:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/7"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-28 04:53:44.000000000","tz":-420},"subject":"Remove unused code in utils","message":"Remove unused code in utils\n\n* Removed utility code not referenced anywhere.\n* Added some additional tests on previously untested code.\n* Moved existing utils tests to designate.tests.unit.\n* Cleaned up existing tests.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f5a411fdd77cd75f2bf37fbb26861c35358e013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f5a411fdd77cd75f2bf37fbb26861c35358e013"}]},"branch":"refs/heads/master"},"1fe9cefc5cf5200c1f1328d4045d80d39d42bdc1":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-06-28 04:57:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/8"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-28 04:55:44.000000000","tz":-420},"subject":"Remove unused code in utils and improved utils testing","message":"Remove unused code in utils and improved utils testing\n\n* Added some additional tests on previously untested code.\n* Cleaned up existing tests.\n* Moved existing utils tests to designate.tests.unit.\n* Moved test utility code into a separate utility file.\n* Removed utility code no longer referenced anywhere.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fe9cefc5cf5200c1f1328d4045d80d39d42bdc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fe9cefc5cf5200c1f1328d4045d80d39d42bdc1"}]},"branch":"refs/heads/master"},"71a0d3428fe6d3dfe5c75061c128db974cc7a8cf":{"kind":"REWORK","_number":9,"created":"2019-06-28 04:59:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/9"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-28 04:59:15.000000000","tz":-420},"subject":"Remove unused code in utils and improved utils testing","message":"Remove unused code in utils and improved utils testing\n\n* Added some additional tests on previously untested code.\n* Cleaned up existing tests.\n* Moved existing utils tests to designate.tests.unit.\n* Moved test utility code into a separate utility file.\n* Removed utility code no longer referenced anywhere.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/71a0d3428fe6d3dfe5c75061c128db974cc7a8cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/71a0d3428fe6d3dfe5c75061c128db974cc7a8cf"}]},"branch":"refs/heads/master"},"173625d588c82032f150887085a1fe095cc8cd4d":{"kind":"REWORK","_number":10,"created":"2019-06-28 05:10:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/10"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-28 05:10:28.000000000","tz":-420},"subject":"Remove unused code in utils and improved utils testing","message":"Remove unused code in utils and improved utils testing\n\n* Added some additional tests on previously untested code.\n* Cleaned up existing tests.\n* Moved existing utils tests to designate.tests.unit.\n* Moved test utility code into a separate utility file.\n* Removed utility code no longer referenced anywhere.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/173625d588c82032f150887085a1fe095cc8cd4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/173625d588c82032f150887085a1fe095cc8cd4d"}]},"branch":"refs/heads/master"},"1d349058640a221809e6bb991d8de73085bb5443":{"kind":"REWORK","_number":11,"created":"2019-06-28 05:50:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/11"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-28 05:49:43.000000000","tz":-420},"subject":"Remove unused code in utils and improved utils testing","message":"Remove unused code in utils and improved utils testing\n\n* Added some additional tests on previously untested code.\n* Cleaned up existing tests.\n* Moved existing utils tests to designate.tests.unit.\n* Moved test utility code into a separate utility file.\n* Removed utility code no longer referenced anywhere.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1d349058640a221809e6bb991d8de73085bb5443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1d349058640a221809e6bb991d8de73085bb5443"}]},"branch":"refs/heads/master"},"4568854c43535b65ab3a93bdd5ca397e88837266":{"kind":"REWORK","_number":12,"created":"2019-06-28 05:56:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/12"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-28 05:56:48.000000000","tz":-420},"subject":"Remove unused code in utils and improved utils testing","message":"Remove unused code in utils and improved utils testing\n\n* Added some additional tests on previously untested code.\n* Cleaned up existing tests.\n* Moved existing utils tests to designate.tests.unit.\n* Moved test utility code into a separate utility file.\n* Removed utility code no longer referenced anywhere.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4568854c43535b65ab3a93bdd5ca397e88837266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4568854c43535b65ab3a93bdd5ca397e88837266"}]},"branch":"refs/heads/master"},"38346246158f2ad26ceaa26054a2e937def86df9":{"kind":"REWORK","_number":13,"created":"2019-06-28 17:26:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/13"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-28 17:26:06.000000000","tz":-420},"subject":"Remove unused code in utils and improved utils testing","message":"Remove unused code in utils and improved utils testing\n\n* Added some additional tests on previously untested code.\n* Cleaned up existing tests.\n* Moved existing utils tests to designate.tests.unit.\n* Moved test utility code into a separate utility file.\n* Removed utility code no longer referenced anywhere.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/38346246158f2ad26ceaa26054a2e937def86df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/38346246158f2ad26ceaa26054a2e937def86df9"}]},"branch":"refs/heads/master"},"ee04eaf2af1f0a39df2feadac8b882b26414a714":{"kind":"REWORK","_number":14,"created":"2019-06-28 17:29:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/664173/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/664173/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/664173/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/664173/14"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:15:15.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-28 17:29:38.000000000","tz":-420},"subject":"Remove unused code in utils and improved utils testing","message":"Remove unused code in utils and improved utils testing\n\n* Added some additional tests on previously untested code.\n* Cleaned up existing tests.\n* Moved existing utils tests to designate.tests.unit.\n* Moved test utility code into a separate utility file.\n* Removed utility code no longer referenced anywhere.\n\nChange-Id: I9af77adb8a0b5c2ee076eeffe993100d8c65196c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee04eaf2af1f0a39df2feadac8b882b26414a714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee04eaf2af1f0a39df2feadac8b882b26414a714"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
