)]}'
{"id":"openstack%2Fdesignate~664176","triplet_id":"openstack%2Fdesignate~master~Iab163987fb8b27b21d4870eeb59f5d5f84229f91","project":"openstack/designate","branch":"master","topic":"deprecated_warnings","hashtags":[],"change_id":"Iab163987fb8b27b21d4870eeb59f5d5f84229f91","subject":"Fixed oslo deprecated warning when running devstack","status":"MERGED","created":"2019-06-10 00:01:21.000000000","updated":"2019-06-19 15:02:49.000000000","submitted":"2019-06-19 15:02:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664176-1560956569422-b95397b7","meta_rev_id":"51edfd0c48898d64909ec359a5a0be6b24edf304","_number":664176,"virtual_id_number":664176,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-06-19 15:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-19 13:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-06-19 13:41:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 13:41:11.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-19 15:02:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d584f022db50aba30c8c663dddfe781f7c92f403","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-10 00:01:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"943eaf238b1e4de214008a6dce848a66d964fe22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 00:54:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/76/664176/1/check/designate-bind9/cd64783/ : SUCCESS in 37m 04s\n- designate-pdns4 http://logs.openstack.org/76/664176/1/check/designate-pdns4/ca31fa0/ : SUCCESS in 35m 52s\n- designate-bind9-manager-model http://logs.openstack.org/76/664176/1/check/designate-bind9-manager-model/62f5a7d/ : SUCCESS in 36m 55s\n- designate-bind9-py27 http://logs.openstack.org/76/664176/1/check/designate-bind9-py27/bcbdba2/ : SUCCESS in 38m 50s\n- designate-pdns4-py27 http://logs.openstack.org/76/664176/1/check/designate-pdns4-py27/fd4d566/ : SUCCESS in 36m 51s\n- designate-pdns4-postgres http://logs.openstack.org/76/664176/1/check/designate-pdns4-postgres/37832e6/ : SUCCESS in 38m 19s\n- designate-grenade-pdns4 http://logs.openstack.org/76/664176/1/check/designate-grenade-pdns4/e1c5923/ : SUCCESS in 52m 01s\n- openstack-tox-cover http://logs.openstack.org/76/664176/1/check/openstack-tox-cover/4f03bf5/cover/ : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/664176/1/check/openstack-tox-lower-constraints/df9bf0e/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/76/664176/1/check/openstack-tox-pep8/959a576/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/76/664176/1/check/openstack-tox-py27/3ff6cd6/ : SUCCESS in 9m 00s\n- openstack-tox-py36 http://logs.openstack.org/76/664176/1/check/openstack-tox-py36/33590a8/ : SUCCESS in 5m 48s\n- openstack-tox-py37 http://logs.openstack.org/76/664176/1/check/openstack-tox-py37/4cc3085/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/76/664176/1/check/openstack-tox-docs/a15e972/html/ : SUCCESS in 4m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/664176/1/check/neutron-tempest-plugin-designate-scenario/15f0e39/ : SUCCESS in 33m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2845d1252d292f59642a784fbb6f689ec46e705e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-10 18:17:40.000000000","message":"Topic changed from notification_deprecated to deprecated_warnings","accounts_in_message":[],"_revision_number":1},{"id":"b2191e106cbf59d7ee90deca68d955fa4b091303","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-11 16:48:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"659ca32d3a5eabf7378c6f82f982f143aa05d9be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 18:09:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/76/664176/2/check/designate-bind9/bd153c3/ : SUCCESS in 40m 00s\n- designate-pdns4 http://logs.openstack.org/76/664176/2/check/designate-pdns4/de5fb76/ : SUCCESS in 41m 12s\n- designate-bind9-manager-model http://logs.openstack.org/76/664176/2/check/designate-bind9-manager-model/3b771c1/ : SUCCESS in 40m 33s\n- designate-bind9-py27 http://logs.openstack.org/76/664176/2/check/designate-bind9-py27/c11776d/ : SUCCESS in 41m 43s\n- designate-pdns4-py27 http://logs.openstack.org/76/664176/2/check/designate-pdns4-py27/a8168c4/ : SUCCESS in 36m 35s\n- designate-pdns4-postgres http://logs.openstack.org/76/664176/2/check/designate-pdns4-postgres/6b0aeef/ : SUCCESS in 37m 18s\n- designate-grenade-pdns4 http://logs.openstack.org/76/664176/2/check/designate-grenade-pdns4/0ab2a56/ : SUCCESS in 1h 11m 44s\n- openstack-tox-cover http://logs.openstack.org/76/664176/2/check/openstack-tox-cover/285bd6e/cover/ : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/664176/2/check/openstack-tox-lower-constraints/4928e69/ : SUCCESS in 5m 54s\n- openstack-tox-pep8 http://logs.openstack.org/76/664176/2/check/openstack-tox-pep8/5a7e903/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/76/664176/2/check/openstack-tox-py27/2a7235a/ : SUCCESS in 5m 45s\n- openstack-tox-py36 http://logs.openstack.org/76/664176/2/check/openstack-tox-py36/6515427/ : SUCCESS in 6m 12s\n- openstack-tox-py37 http://logs.openstack.org/76/664176/2/check/openstack-tox-py37/c2efb73/ : SUCCESS in 6m 33s\n- openstack-tox-docs http://logs.openstack.org/76/664176/2/check/openstack-tox-docs/09d8485/html/ : SUCCESS in 5m 47s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/664176/2/check/neutron-tempest-plugin-designate-scenario/12d7374/ : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e27ad9385e7bf4966e7937c767739fecef106f38","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 19:51:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f6ebe2aa42c62a77c7709936e0e796040a361b29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:01:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/76/664176/3/check/designate-bind9/75752ce/ : SUCCESS in 40m 22s\n- designate-pdns4 http://logs.openstack.org/76/664176/3/check/designate-pdns4/0856818/ : SUCCESS in 43m 32s\n- designate-bind9-manager-model http://logs.openstack.org/76/664176/3/check/designate-bind9-manager-model/b2411c5/ : SUCCESS in 38m 40s\n- designate-bind9-py27 http://logs.openstack.org/76/664176/3/check/designate-bind9-py27/bc69137/ : SUCCESS in 43m 26s\n- designate-pdns4-py27 http://logs.openstack.org/76/664176/3/check/designate-pdns4-py27/1a03d17/ : SUCCESS in 37m 28s\n- designate-pdns4-postgres http://logs.openstack.org/76/664176/3/check/designate-pdns4-postgres/b917573/ : SUCCESS in 41m 19s\n- designate-grenade-pdns4 http://logs.openstack.org/76/664176/3/check/designate-grenade-pdns4/e15e94b/ : SUCCESS in 59m 18s\n- openstack-tox-cover http://logs.openstack.org/76/664176/3/check/openstack-tox-cover/bd67a74/cover/ : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/664176/3/check/openstack-tox-lower-constraints/5d1fae8/ : SUCCESS in 6m 13s\n- openstack-tox-pep8 http://logs.openstack.org/76/664176/3/check/openstack-tox-pep8/68e2e00/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/76/664176/3/check/openstack-tox-py27/dc44f47/ : SUCCESS in 5m 54s\n- openstack-tox-py36 http://logs.openstack.org/76/664176/3/check/openstack-tox-py36/b2b1cfa/ : SUCCESS in 6m 27s\n- openstack-tox-py37 http://logs.openstack.org/76/664176/3/check/openstack-tox-py37/1721cf8/ : SUCCESS in 6m 29s\n- openstack-tox-docs http://logs.openstack.org/76/664176/3/check/openstack-tox-docs/33d8d72/html/ : SUCCESS in 5m 11s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/664176/3/check/neutron-tempest-plugin-designate-scenario/d45d835/ : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ffdf649bd625de5d42322856ab709eed53d5114f","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-19 13:41:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9b57e338dc896b99087c39f6e273d8905e13dee0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 13:41:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e11bf1d278bc2a348dbfde616659a6ac0a51d62f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:02:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"51edfd0c48898d64909ec359a5a0be6b24edf304","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:02:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/76/664176/3/gate/designate-bind9/f373169/ : SUCCESS in 36m 31s\n- designate-pdns4 http://logs.openstack.org/76/664176/3/gate/designate-pdns4/7e78f0c/ : SUCCESS in 1h 05m 59s\n- designate-bind9-manager-model http://logs.openstack.org/76/664176/3/gate/designate-bind9-manager-model/f993b84/ : SUCCESS in 34m 53s\n- designate-bind9-py27 http://logs.openstack.org/76/664176/3/gate/designate-bind9-py27/8fa03d6/ : SUCCESS in 41m 54s\n- designate-pdns4-py27 http://logs.openstack.org/76/664176/3/gate/designate-pdns4-py27/403f55f/ : SUCCESS in 49m 27s\n- designate-pdns4-postgres http://logs.openstack.org/76/664176/3/gate/designate-pdns4-postgres/6974aef/ : SUCCESS in 43m 12s\n- designate-grenade-pdns4 http://logs.openstack.org/76/664176/3/gate/designate-grenade-pdns4/c91d465/ : SUCCESS in 1h 03m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/664176/3/gate/openstack-tox-lower-constraints/63d4e4c/ : SUCCESS in 6m 54s\n- openstack-tox-pep8 http://logs.openstack.org/76/664176/3/gate/openstack-tox-pep8/7a9c07d/ : SUCCESS in 7m 20s\n- openstack-tox-py27 http://logs.openstack.org/76/664176/3/gate/openstack-tox-py27/1b72e8d/ : SUCCESS in 5m 59s\n- openstack-tox-py36 http://logs.openstack.org/76/664176/3/gate/openstack-tox-py36/b3583b0/ : SUCCESS in 8m 35s\n- openstack-tox-py37 http://logs.openstack.org/76/664176/3/gate/openstack-tox-py37/32e98ad/ : SUCCESS in 9m 32s\n- openstack-tox-docs http://logs.openstack.org/76/664176/3/gate/openstack-tox-docs/1d2236c/html/ : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a31982b53e230835dd7729b26046fdcbfeb1fd75","revisions":{"ea1ef85709b4e06528ece9de0a673a984fcbc7fb":{"kind":"REWORK","_number":1,"created":"2019-06-10 00:01:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/664176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/664176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/664176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/664176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/664176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/664176/1"}}},"commit":{"parents":[{"commit":"174365947394ef304a3aa729658765bc3d36eabf","subject":"Merge \"Remove KS v2 support and add region support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/174365947394ef304a3aa729658765bc3d36eabf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:58:46.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 00:01:09.000000000","tz":-420},"subject":"Fixed oslo deprecated warning when running devstack","message":"Fixed oslo deprecated warning when running devstack\n\nMoved notification driver and topics from DEFAULT to the\noslo_messaging_notifications group.\n\nChange-Id: Iab163987fb8b27b21d4870eeb59f5d5f84229f91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ea1ef85709b4e06528ece9de0a673a984fcbc7fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ea1ef85709b4e06528ece9de0a673a984fcbc7fb"}]},"branch":"refs/heads/master"},"e31e7bc679bf54bad352f905461e8cc90c0ddd83":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-11 16:48:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/664176/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/664176/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/664176/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/664176/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/664176/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/664176/2"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:58:46.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 16:48:17.000000000","tz":0},"subject":"Fixed oslo deprecated warning when running devstack","message":"Fixed oslo deprecated warning when running devstack\n\nMoved notification driver and topics from DEFAULT to the\noslo_messaging_notifications group.\n\nChange-Id: Iab163987fb8b27b21d4870eeb59f5d5f84229f91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e31e7bc679bf54bad352f905461e8cc90c0ddd83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e31e7bc679bf54bad352f905461e8cc90c0ddd83"}]},"branch":"refs/heads/master"},"a31982b53e230835dd7729b26046fdcbfeb1fd75":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-18 19:51:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/664176/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/664176/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/664176/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/664176/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/664176/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/664176/3"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-09 23:58:46.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 19:51:14.000000000","tz":0},"subject":"Fixed oslo deprecated warning when running devstack","message":"Fixed oslo deprecated warning when running devstack\n\nMoved notification driver and topics from DEFAULT to the\noslo_messaging_notifications group.\n\nChange-Id: Iab163987fb8b27b21d4870eeb59f5d5f84229f91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a31982b53e230835dd7729b26046fdcbfeb1fd75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a31982b53e230835dd7729b26046fdcbfeb1fd75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
