)]}'
{"id":"openstack%2Fdesignate~664181","triplet_id":"openstack%2Fdesignate~master~I100ab865410ef09fb380b9add5b319c72cc5b1a3","project":"openstack/designate","branch":"master","topic":"config_improvements","hashtags":[],"change_id":"I100ab865410ef09fb380b9add5b319c72cc5b1a3","subject":"Removing unused and duplicate config options","status":"ABANDONED","created":"2019-06-10 01:53:00.000000000","updated":"2019-06-14 02:38:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f9fff2cba6c90dd3de5d9c35bd07b6e56649d20a","_number":664181,"virtual_id_number":664181,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-11 20:07:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 20:07:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb1c5d1f0289ac8ac2a51d524db5dd552e4b00ef","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-10 01:53:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89c6ad6ead0eb7cce195a575967bea57226cb109","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 03:02:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/81/664181/1/check/designate-bind9/f4b6897/ : SUCCESS in 39m 54s\n- designate-pdns4 http://logs.openstack.org/81/664181/1/check/designate-pdns4/76e544c/ : SUCCESS in 39m 28s\n- designate-bind9-manager-model http://logs.openstack.org/81/664181/1/check/designate-bind9-manager-model/28563cd/ : SUCCESS in 43m 28s\n- designate-bind9-py27 http://logs.openstack.org/81/664181/1/check/designate-bind9-py27/55009c6/ : SUCCESS in 43m 11s\n- designate-pdns4-py27 http://logs.openstack.org/81/664181/1/check/designate-pdns4-py27/207a250/ : SUCCESS in 41m 56s\n- designate-pdns4-postgres http://logs.openstack.org/81/664181/1/check/designate-pdns4-postgres/8d9ee97/ : SUCCESS in 38m 03s\n- designate-grenade-pdns4 http://logs.openstack.org/81/664181/1/check/designate-grenade-pdns4/be70769/ : SUCCESS in 57m 03s\n- openstack-tox-cover http://logs.openstack.org/81/664181/1/check/openstack-tox-cover/555768d/cover/ : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/664181/1/check/openstack-tox-lower-constraints/460391e/ : SUCCESS in 8m 42s\n- openstack-tox-pep8 http://logs.openstack.org/81/664181/1/check/openstack-tox-pep8/026ebc9/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/81/664181/1/check/openstack-tox-py27/c0bb917/ : SUCCESS in 6m 45s\n- openstack-tox-py36 http://logs.openstack.org/81/664181/1/check/openstack-tox-py36/446b8ab/ : SUCCESS in 6m 10s\n- openstack-tox-py37 http://logs.openstack.org/81/664181/1/check/openstack-tox-py37/3edf070/ : SUCCESS in 7m 09s\n- openstack-tox-docs http://logs.openstack.org/81/664181/1/check/openstack-tox-docs/4cc6f26/html/ : SUCCESS in 6m 59s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/81/664181/1/check/neutron-tempest-plugin-designate-scenario/6a7000a/ : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5d5dfe902d1867ba98dfd305b52ce082500cda63","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-10 18:18:03.000000000","message":"Topic changed from unused_configs to config_improvements","accounts_in_message":[],"_revision_number":1},{"id":"43cd206a52e6fd2eb6a2430985c0f36fa6ca1b96","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-11 18:53:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1d762a86c7e217d64840bada6a6d4d2d2ba6ada2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:07:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/81/664181/2/check/designate-bind9/ad0b066/ : SUCCESS in 39m 57s\n- designate-pdns4 http://logs.openstack.org/81/664181/2/check/designate-pdns4/51bbe57/ : SUCCESS in 42m 55s\n- designate-bind9-manager-model http://logs.openstack.org/81/664181/2/check/designate-bind9-manager-model/e0f51cc/ : SUCCESS in 35m 04s\n- designate-bind9-py27 http://logs.openstack.org/81/664181/2/check/designate-bind9-py27/80d8a94/ : SUCCESS in 41m 18s\n- designate-pdns4-py27 http://logs.openstack.org/81/664181/2/check/designate-pdns4-py27/391a828/ : SUCCESS in 37m 30s\n- designate-pdns4-postgres http://logs.openstack.org/81/664181/2/check/designate-pdns4-postgres/e038c30/ : SUCCESS in 38m 17s\n- designate-grenade-pdns4 http://logs.openstack.org/81/664181/2/check/designate-grenade-pdns4/cfc5bce/ : SUCCESS in 1h 06m 34s\n- openstack-tox-cover http://logs.openstack.org/81/664181/2/check/openstack-tox-cover/742d175/cover/ : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/664181/2/check/openstack-tox-lower-constraints/4187518/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/81/664181/2/check/openstack-tox-pep8/576cf05/ : SUCCESS in 5m 17s\n- openstack-tox-py27 http://logs.openstack.org/81/664181/2/check/openstack-tox-py27/f502f52/ : SUCCESS in 7m 38s\n- openstack-tox-py36 http://logs.openstack.org/81/664181/2/check/openstack-tox-py36/696eaf8/ : SUCCESS in 6m 03s\n- openstack-tox-py37 http://logs.openstack.org/81/664181/2/check/openstack-tox-py37/7b67850/ : SUCCESS in 7m 57s\n- openstack-tox-docs http://logs.openstack.org/81/664181/2/check/openstack-tox-docs/8290d14/html/ : SUCCESS in 5m 08s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/81/664181/2/check/neutron-tempest-plugin-designate-scenario/2f87183/ : SUCCESS in 39m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f9fff2cba6c90dd3de5d9c35bd07b6e56649d20a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 02:38:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f3c1fe197824325d0a55b0e2acd6b6c4d5c0b33","revisions":{"0a5e2164ef5f14f7a616cc33e627498b9c2b6cab":{"kind":"REWORK","_number":1,"created":"2019-06-10 01:53:00.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/81/664181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/664181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/664181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/664181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/664181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/664181/1"}}},"commit":{"parents":[{"commit":"174365947394ef304a3aa729658765bc3d36eabf","subject":"Merge \"Remove KS v2 support and add region support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/174365947394ef304a3aa729658765bc3d36eabf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 01:52:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 01:52:49.000000000","tz":-420},"subject":"Removing unused and duplicate config options","message":"Removing unused and duplicate config options\n\nChange-Id: I100ab865410ef09fb380b9add5b319c72cc5b1a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a5e2164ef5f14f7a616cc33e627498b9c2b6cab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a5e2164ef5f14f7a616cc33e627498b9c2b6cab"}]},"branch":"refs/heads/master"},"4f3c1fe197824325d0a55b0e2acd6b6c4d5c0b33":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-11 18:53:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/81/664181/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/664181/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/664181/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/664181/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/664181/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/664181/2"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 01:52:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 18:53:38.000000000","tz":0},"subject":"Removing unused and duplicate config options","message":"Removing unused and duplicate config options\n\nChange-Id: I100ab865410ef09fb380b9add5b319c72cc5b1a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f3c1fe197824325d0a55b0e2acd6b6c4d5c0b33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f3c1fe197824325d0a55b0e2acd6b6c4d5c0b33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
