)]}'
{"id":"openstack%2Fdesignate~664182","triplet_id":"openstack%2Fdesignate~master~Id802d981de5577a54ff1b5b71e9e134d1175bbf4","project":"openstack/designate","branch":"master","topic":"config_improvements","hashtags":[],"change_id":"Id802d981de5577a54ff1b5b71e9e134d1175bbf4","subject":"Change configuration to be consistent with underlines","status":"MERGED","created":"2019-06-10 02:03:53.000000000","updated":"2019-07-01 19:27:06.000000000","submitted":"2019-07-01 19:27:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664182-1562009226238-a98d8c21","meta_rev_id":"47e48b8a1111f9f39d0411dac7f1ba6bde8c3fef","_number":664182,"virtual_id_number":664182,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":2,"date":"2019-07-01 19:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-07-01 15:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-06-27 18:14:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-01 18:10:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 01:05:11.000000000","updated_by":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"reviewer":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"state":"REVIEWER"},{"updated":"2019-07-01 15:26:27.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-07-01 19:27:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49da06d385e93c4d4713f54f1edba4263869fce2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-10 02:03:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08bdab403481e58654a1bfa7ad7768cb9c43ce77","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-10 02:09:46.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6e2a705a2df3fc1bc6af0eca9993712e1c8dc041","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 03:15:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/82/664182/2/check/designate-bind9/b68a06a/ : SUCCESS in 42m 47s\n- designate-pdns4 http://logs.openstack.org/82/664182/2/check/designate-pdns4/4c3744c/ : SUCCESS in 36m 18s\n- designate-bind9-manager-model http://logs.openstack.org/82/664182/2/check/designate-bind9-manager-model/2c7c145/ : SUCCESS in 44m 53s\n- designate-bind9-py27 http://logs.openstack.org/82/664182/2/check/designate-bind9-py27/fa7da7d/ : SUCCESS in 43m 55s\n- designate-pdns4-py27 http://logs.openstack.org/82/664182/2/check/designate-pdns4-py27/813b3a6/ : SUCCESS in 39m 45s\n- designate-pdns4-postgres http://logs.openstack.org/82/664182/2/check/designate-pdns4-postgres/108d76a/ : SUCCESS in 36m 02s\n- designate-grenade-pdns4 http://logs.openstack.org/82/664182/2/check/designate-grenade-pdns4/e96b8e2/ : SUCCESS in 59m 42s\n- openstack-tox-cover http://logs.openstack.org/82/664182/2/check/openstack-tox-cover/d2b7872/cover/ : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/664182/2/check/openstack-tox-lower-constraints/df5dbb8/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/82/664182/2/check/openstack-tox-pep8/f2e7862/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/82/664182/2/check/openstack-tox-py27/83cb3ef/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/82/664182/2/check/openstack-tox-py36/a613db2/ : SUCCESS in 5m 21s\n- openstack-tox-py37 http://logs.openstack.org/82/664182/2/check/openstack-tox-py37/3c34802/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/82/664182/2/check/openstack-tox-docs/aed2991/html/ : SUCCESS in 4m 36s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/82/664182/2/check/neutron-tempest-plugin-designate-scenario/3e2af8c/ : SUCCESS in 36m 40s","accounts_in_message":[],"_revision_number":2},{"id":"8e2b749b2bfd7a902a38764a2a864b30d0185214","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-10 18:18:00.000000000","message":"Topic changed from consistent_config to config_improvements","accounts_in_message":[],"_revision_number":2},{"id":"85e318cc289611fd0ed2ae95f721d038f176a4a4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-11 19:38:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1972959c34aae4ca1f77900627bed0957e7f18d8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-11 19:44:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa0d354b377d99042ac6382537687a9e5963b233","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:47:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/82/664182/4/check/designate-bind9/c6b7887/ : SUCCESS in 38m 43s\n- designate-pdns4 http://logs.openstack.org/82/664182/4/check/designate-pdns4/320bd05/ : SUCCESS in 36m 37s\n- designate-bind9-manager-model http://logs.openstack.org/82/664182/4/check/designate-bind9-manager-model/ce0b531/ : SUCCESS in 42m 31s\n- designate-bind9-py27 http://logs.openstack.org/82/664182/4/check/designate-bind9-py27/8204821/ : SUCCESS in 44m 21s\n- designate-pdns4-py27 http://logs.openstack.org/82/664182/4/check/designate-pdns4-py27/dcc1305/ : SUCCESS in 44m 04s\n- designate-pdns4-postgres http://logs.openstack.org/82/664182/4/check/designate-pdns4-postgres/e922c78/ : SUCCESS in 38m 35s\n- designate-grenade-pdns4 http://logs.openstack.org/82/664182/4/check/designate-grenade-pdns4/8b4ed94/ : SUCCESS in 54m 27s\n- openstack-tox-cover http://logs.openstack.org/82/664182/4/check/openstack-tox-cover/a74e6df/cover/ : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/664182/4/check/openstack-tox-lower-constraints/c96ad46/ : RETRY_LIMIT in 1m 29s\n- openstack-tox-pep8 http://logs.openstack.org/82/664182/4/check/openstack-tox-pep8/31c6dce/ : RETRY_LIMIT in 1m 40s\n- openstack-tox-py27 http://logs.openstack.org/82/664182/4/check/openstack-tox-py27/3af4fee/ : SUCCESS in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/82/664182/4/check/openstack-tox-py36/299797f/ : RETRY_LIMIT in 1m 50s\n- openstack-tox-py37 http://logs.openstack.org/82/664182/4/check/openstack-tox-py37/b7f281c/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/82/664182/4/check/openstack-tox-docs/4022848/html/ : SUCCESS in 4m 52s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/82/664182/4/check/neutron-tempest-plugin-designate-scenario/2ed69b4/ : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":4},{"id":"2f5dee988cd63377e8dfa6c3f5e4a8ca18c0eacb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-11 21:32:50.000000000","message":"Patch Set 4:\n\nrecheck mirror error","accounts_in_message":[],"_revision_number":4},{"id":"80804b76559206f60a56b1a14d4a22bd9dec4818","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:48:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/82/664182/4/check/designate-bind9/fc0b79c/ : SUCCESS in 44m 34s\n- designate-pdns4 http://logs.openstack.org/82/664182/4/check/designate-pdns4/ea29041/ : SUCCESS in 34m 19s\n- designate-bind9-manager-model http://logs.openstack.org/82/664182/4/check/designate-bind9-manager-model/a4a0c63/ : SUCCESS in 38m 52s\n- designate-bind9-py27 http://logs.openstack.org/82/664182/4/check/designate-bind9-py27/176acbd/ : SUCCESS in 41m 17s\n- designate-pdns4-py27 http://logs.openstack.org/82/664182/4/check/designate-pdns4-py27/b7c4dd9/ : SUCCESS in 37m 01s\n- designate-pdns4-postgres http://logs.openstack.org/82/664182/4/check/designate-pdns4-postgres/18b0464/ : SUCCESS in 39m 21s\n- designate-grenade-pdns4 http://logs.openstack.org/82/664182/4/check/designate-grenade-pdns4/867367b/ : SUCCESS in 57m 50s\n- openstack-tox-cover http://logs.openstack.org/82/664182/4/check/openstack-tox-cover/b9315e4/cover/ : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/664182/4/check/openstack-tox-lower-constraints/b25293c/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/82/664182/4/check/openstack-tox-pep8/edd6570/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/82/664182/4/check/openstack-tox-py27/d520f45/ : SUCCESS in 6m 10s\n- openstack-tox-py36 http://logs.openstack.org/82/664182/4/check/openstack-tox-py36/2b1c373/ : RETRY_LIMIT in 1m 32s\n- openstack-tox-py37 http://logs.openstack.org/82/664182/4/check/openstack-tox-py37/429c1b5/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/82/664182/4/check/openstack-tox-docs/31e2ec5/html/ : SUCCESS in 4m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/82/664182/4/check/neutron-tempest-plugin-designate-scenario/16c21b6/ : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":4},{"id":"7003ac428b42247a8a8490dcc68f28e96be4a448","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-11 23:28:33.000000000","message":"Patch Set 4:\n\nrecheck mirror error","accounts_in_message":[],"_revision_number":4},{"id":"f2a9438724a41fe52d2a9ad4d15a2c99a1948a0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:26:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/82/664182/4/check/designate-bind9/5d0e639/ : SUCCESS in 42m 59s\n- designate-pdns4 http://logs.openstack.org/82/664182/4/check/designate-pdns4/7fb0224/ : SUCCESS in 38m 41s\n- designate-bind9-manager-model http://logs.openstack.org/82/664182/4/check/designate-bind9-manager-model/3a12dde/ : SUCCESS in 39m 32s\n- designate-bind9-py27 http://logs.openstack.org/82/664182/4/check/designate-bind9-py27/4d337dc/ : SUCCESS in 47m 40s\n- designate-pdns4-py27 http://logs.openstack.org/82/664182/4/check/designate-pdns4-py27/b67ef44/ : SUCCESS in 36m 50s\n- designate-pdns4-postgres http://logs.openstack.org/82/664182/4/check/designate-pdns4-postgres/19b7b5a/ : SUCCESS in 40m 46s\n- designate-grenade-pdns4 http://logs.openstack.org/82/664182/4/check/designate-grenade-pdns4/d5c6d90/ : SUCCESS in 51m 50s\n- openstack-tox-cover http://logs.openstack.org/82/664182/4/check/openstack-tox-cover/f4528b2/cover/ : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/664182/4/check/openstack-tox-lower-constraints/755127a/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/82/664182/4/check/openstack-tox-pep8/62b12a9/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/82/664182/4/check/openstack-tox-py27/8ab574b/ : SUCCESS in 7m 06s\n- openstack-tox-py36 http://logs.openstack.org/82/664182/4/check/openstack-tox-py36/9c35662/ : SUCCESS in 7m 12s\n- openstack-tox-py37 http://logs.openstack.org/82/664182/4/check/openstack-tox-py37/9945f65/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/82/664182/4/check/openstack-tox-docs/e972234/html/ : SUCCESS in 5m 19s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/82/664182/4/check/neutron-tempest-plugin-designate-scenario/2b9a5eb/ : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":4},{"id":"6e72d6152cff468ef520c3a473769809a5fe12fb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 02:40:14.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"e614e210ddcaaeb72aee40ccd7ad73e41d5b7e9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:51:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/82/664182/5/check/designate-bind9/b9b7b80/ : SUCCESS in 40m 10s\n- designate-pdns4 http://logs.openstack.org/82/664182/5/check/designate-pdns4/462c432/ : SUCCESS in 39m 58s\n- designate-bind9-manager-model http://logs.openstack.org/82/664182/5/check/designate-bind9-manager-model/9edd7a0/ : SUCCESS in 38m 58s\n- designate-bind9-py27 http://logs.openstack.org/82/664182/5/check/designate-bind9-py27/28e989f/ : SUCCESS in 41m 13s\n- designate-pdns4-py27 http://logs.openstack.org/82/664182/5/check/designate-pdns4-py27/387e4f4/ : SUCCESS in 36m 22s\n- designate-pdns4-postgres http://logs.openstack.org/82/664182/5/check/designate-pdns4-postgres/6599812/ : SUCCESS in 42m 37s\n- designate-grenade-pdns4 http://logs.openstack.org/82/664182/5/check/designate-grenade-pdns4/03fbe7d/ : SUCCESS in 54m 56s\n- openstack-tox-cover http://logs.openstack.org/82/664182/5/check/openstack-tox-cover/afc45b4/cover/ : SUCCESS in 8m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/664182/5/check/openstack-tox-lower-constraints/0543190/ : SUCCESS in 8m 03s\n- openstack-tox-pep8 http://logs.openstack.org/82/664182/5/check/openstack-tox-pep8/0df32fd/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/82/664182/5/check/openstack-tox-py27/ec310e7/ : SUCCESS in 6m 48s\n- openstack-tox-py36 http://logs.openstack.org/82/664182/5/check/openstack-tox-py36/6f45df0/ : SUCCESS in 6m 32s\n- openstack-tox-py37 http://logs.openstack.org/82/664182/5/check/openstack-tox-py37/631379f/ : SUCCESS in 9m 00s\n- openstack-tox-docs http://logs.openstack.org/82/664182/5/check/openstack-tox-docs/3ab5c54/html/ : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/82/664182/5/check/neutron-tempest-plugin-designate-scenario/2972231/ : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":5},{"id":"eef0465d832ce927a61ab5e67462f15076b09766","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-06-18 01:05:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3482c915bcba7e0d443b5449e146939edeee864d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 19:51:38.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"fe392d3047911b299dec85eb9fc73450917e8941","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:06:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/82/664182/6/check/designate-bind9/ec0773d/ : SUCCESS in 42m 12s\n- designate-pdns4 http://logs.openstack.org/82/664182/6/check/designate-pdns4/35a9a59/ : SUCCESS in 38m 08s\n- designate-bind9-manager-model http://logs.openstack.org/82/664182/6/check/designate-bind9-manager-model/8a8d5a6/ : SUCCESS in 40m 44s\n- designate-bind9-py27 http://logs.openstack.org/82/664182/6/check/designate-bind9-py27/16efe18/ : SUCCESS in 44m 36s\n- designate-pdns4-py27 http://logs.openstack.org/82/664182/6/check/designate-pdns4-py27/f225959/ : SUCCESS in 42m 46s\n- designate-pdns4-postgres http://logs.openstack.org/82/664182/6/check/designate-pdns4-postgres/1fcc54c/ : SUCCESS in 45m 26s\n- designate-grenade-pdns4 http://logs.openstack.org/82/664182/6/check/designate-grenade-pdns4/c0552b2/ : SUCCESS in 1h 05m 14s\n- openstack-tox-cover http://logs.openstack.org/82/664182/6/check/openstack-tox-cover/0c4d2bb/cover/ : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/664182/6/check/openstack-tox-lower-constraints/5001c0f/ : SUCCESS in 6m 46s\n- openstack-tox-pep8 http://logs.openstack.org/82/664182/6/check/openstack-tox-pep8/6dd853e/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/82/664182/6/check/openstack-tox-py27/4a8db77/ : SUCCESS in 5m 31s\n- openstack-tox-py36 http://logs.openstack.org/82/664182/6/check/openstack-tox-py36/d1c3f8c/ : SUCCESS in 5m 48s\n- openstack-tox-py37 http://logs.openstack.org/82/664182/6/check/openstack-tox-py37/c44f89d/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/82/664182/6/check/openstack-tox-docs/bc63924/html/ : SUCCESS in 5m 26s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/82/664182/6/check/neutron-tempest-plugin-designate-scenario/64fce28/ : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":6},{"id":"5e1b3e42bf433591916ebcc94ba7e4ee1aec8b4c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-24 00:57:04.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"57168eee6cf1ae95387afb36d3514874bac1379f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 02:07:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/82/664182/7/check/designate-bind9/4f0be24/ : SUCCESS in 36m 11s\n- designate-pdns4 http://logs.openstack.org/82/664182/7/check/designate-pdns4/44d19ce/ : SUCCESS in 34m 33s\n- designate-bind9-manager-model http://logs.openstack.org/82/664182/7/check/designate-bind9-manager-model/f2db6b0/ : SUCCESS in 36m 20s\n- designate-bind9-py27 http://logs.openstack.org/82/664182/7/check/designate-bind9-py27/dba2e18/ : SUCCESS in 40m 04s\n- designate-pdns4-py27 http://logs.openstack.org/82/664182/7/check/designate-pdns4-py27/5ebe9cd/ : SUCCESS in 40m 22s\n- designate-pdns4-postgres http://logs.openstack.org/82/664182/7/check/designate-pdns4-postgres/26f4d6a/ : SUCCESS in 36m 18s\n- designate-grenade-pdns4 http://logs.openstack.org/82/664182/7/check/designate-grenade-pdns4/3e779e4/ : SUCCESS in 53m 42s\n- openstack-tox-cover http://logs.openstack.org/82/664182/7/check/openstack-tox-cover/6b00253/cover/ : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/664182/7/check/openstack-tox-lower-constraints/138f774/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/82/664182/7/check/openstack-tox-pep8/735a849/ : SUCCESS in 6m 00s\n- openstack-tox-py27 http://logs.openstack.org/82/664182/7/check/openstack-tox-py27/ed5a78e/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/82/664182/7/check/openstack-tox-py36/2d00b1d/ : SUCCESS in 5m 45s\n- openstack-tox-py37 http://logs.openstack.org/82/664182/7/check/openstack-tox-py37/cfd2f66/ : SUCCESS in 6m 37s\n- openstack-tox-docs http://logs.openstack.org/82/664182/7/check/openstack-tox-docs/f9c89a9/html/ : SUCCESS in 5m 24s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/82/664182/7/check/neutron-tempest-plugin-designate-scenario/8244f37/ : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":7},{"id":"c180b41c7dbe3a5b03f4235d94dc770c79f86e10","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-27 18:14:01.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"a07c0744b481db6c817763a779c24f66f971a173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 19:29:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/82/664182/8/check/designate-bind9/f3e3a74/ : SUCCESS in 37m 14s\n- designate-pdns4 http://logs.openstack.org/82/664182/8/check/designate-pdns4/1a9f2da/ : SUCCESS in 37m 47s\n- designate-bind9-manager-model http://logs.openstack.org/82/664182/8/check/designate-bind9-manager-model/2bb453d/ : SUCCESS in 37m 32s\n- designate-bind9-py27 http://logs.openstack.org/82/664182/8/check/designate-bind9-py27/2f4261e/ : SUCCESS in 41m 55s\n- designate-pdns4-py27 http://logs.openstack.org/82/664182/8/check/designate-pdns4-py27/0ef2b46/ : SUCCESS in 38m 33s\n- designate-pdns4-postgres http://logs.openstack.org/82/664182/8/check/designate-pdns4-postgres/698c6c9/ : SUCCESS in 35m 55s\n- designate-grenade-pdns4 http://logs.openstack.org/82/664182/8/check/designate-grenade-pdns4/d8ecb1f/ : SUCCESS in 1h 07m 42s\n- openstack-tox-cover http://logs.openstack.org/82/664182/8/check/openstack-tox-cover/d6d504f/cover/ : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/664182/8/check/openstack-tox-lower-constraints/cdd2fdf/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/82/664182/8/check/openstack-tox-pep8/74ca0bb/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/82/664182/8/check/openstack-tox-py27/ec3fdea/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/82/664182/8/check/openstack-tox-py36/e673b38/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/82/664182/8/check/openstack-tox-py37/aea4dca/ : SUCCESS in 5m 03s\n- openstack-tox-docs http://logs.openstack.org/82/664182/8/check/openstack-tox-docs/eb77aa8/html/ : SUCCESS in 3m 28s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/82/664182/8/check/neutron-tempest-plugin-designate-scenario/2b35693/ : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":8},{"id":"8a38b97416f834b1e105e83c73c890af97fc45a4","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-07-01 15:26:27.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"3b444fd025b6f823f4044befb0f0e80425868cb4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-01 18:10:42.000000000","message":"Patch Set 8: Workflow+1\n\nMerging this in as it does not actually change any functionality or behavior.","accounts_in_message":[],"_revision_number":8},{"id":"da32a6c5dfe88932c26dcc572ce7518278216d4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 18:10:55.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"68a306ef94037290a9c246ad9376f17ae95d8387","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 19:27:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"47e48b8a1111f9f39d0411dac7f1ba6bde8c3fef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 19:27:06.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/82/664182/8/gate/designate-bind9/207ad42/ : SUCCESS in 37m 47s\n- designate-pdns4 http://logs.openstack.org/82/664182/8/gate/designate-pdns4/62a90af/ : SUCCESS in 35m 06s\n- designate-bind9-manager-model http://logs.openstack.org/82/664182/8/gate/designate-bind9-manager-model/32b10e6/ : SUCCESS in 32m 51s\n- designate-bind9-py27 http://logs.openstack.org/82/664182/8/gate/designate-bind9-py27/4d14d50/ : SUCCESS in 37m 48s\n- designate-pdns4-py27 http://logs.openstack.org/82/664182/8/gate/designate-pdns4-py27/bacfd4a/ : SUCCESS in 36m 11s\n- designate-pdns4-postgres http://logs.openstack.org/82/664182/8/gate/designate-pdns4-postgres/5e614d9/ : SUCCESS in 45m 54s\n- designate-grenade-pdns4 http://logs.openstack.org/82/664182/8/gate/designate-grenade-pdns4/f5a4088/ : SUCCESS in 1h 10m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/664182/8/gate/openstack-tox-lower-constraints/cd3b6bc/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/82/664182/8/gate/openstack-tox-pep8/e5a1457/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/82/664182/8/gate/openstack-tox-py27/f29129e/ : SUCCESS in 6m 08s\n- openstack-tox-py36 http://logs.openstack.org/82/664182/8/gate/openstack-tox-py36/0ba0d4a/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/82/664182/8/gate/openstack-tox-py37/726bb8f/ : SUCCESS in 4m 21s\n- openstack-tox-docs http://logs.openstack.org/82/664182/8/gate/openstack-tox-docs/7039ed5/html/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"c5bc8947814b57de9b9bae9c67a201bb59d30ae8","revisions":{"c83fde577766e375f000bbff727dcaed2b260c97":{"kind":"REWORK","_number":1,"created":"2019-06-10 02:03:53.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/664182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/664182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/664182/1"}}},"commit":{"parents":[{"commit":"0a5e2164ef5f14f7a616cc33e627498b9c2b6cab","subject":"Removing unused and duplicate config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a5e2164ef5f14f7a616cc33e627498b9c2b6cab"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 02:02:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 02:02:37.000000000","tz":-420},"subject":"Change configuration to be consitent with underlines","message":"Change configuration to be consitent with underlines\n\nThis patch simply changes all configuration names to\nconsitently use underlines over dashes to make it\neasier to search the code.\n\nChange-Id: Id802d981de5577a54ff1b5b71e9e134d1175bbf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c83fde577766e375f000bbff727dcaed2b260c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c83fde577766e375f000bbff727dcaed2b260c97"}]},"branch":"refs/heads/master"},"d555327a2064bd48570719dff4a533b36eb550b1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-10 02:09:46.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/664182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/664182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/664182/2"}}},"commit":{"parents":[{"commit":"0a5e2164ef5f14f7a616cc33e627498b9c2b6cab","subject":"Removing unused and duplicate config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a5e2164ef5f14f7a616cc33e627498b9c2b6cab"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 02:02:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 02:09:06.000000000","tz":-420},"subject":"Change configuration to be consistent with underlines","message":"Change configuration to be consistent with underlines\n\nThis patch simply changes all configuration names to\nconsitently use underlines over dashes to make it\neasier to search the code.\n\nChange-Id: Id802d981de5577a54ff1b5b71e9e134d1175bbf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d555327a2064bd48570719dff4a533b36eb550b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d555327a2064bd48570719dff4a533b36eb550b1"}]},"branch":"refs/heads/master"},"7fe219d34c84c066649ca0f0738d9fdabff11b29":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-11 19:38:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/664182/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/664182/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/664182/3"}}},"commit":{"parents":[{"commit":"4f3c1fe197824325d0a55b0e2acd6b6c4d5c0b33","subject":"Removing unused and duplicate config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f3c1fe197824325d0a55b0e2acd6b6c4d5c0b33"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 02:02:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 19:38:14.000000000","tz":0},"subject":"Change configuration to be consistent with underlines","message":"Change configuration to be consistent with underlines\n\nThis patch simply changes all configuration names to\nconsitently use underlines over dashes to make it\neasier to search the code.\n\nChange-Id: Id802d981de5577a54ff1b5b71e9e134d1175bbf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe219d34c84c066649ca0f0738d9fdabff11b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe219d34c84c066649ca0f0738d9fdabff11b29"}]},"branch":"refs/heads/master"},"4fe71d1f9d58ecdc1f7d432edab4137f41defd49":{"kind":"REWORK","_number":4,"created":"2019-06-11 19:44:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/664182/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/664182/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/664182/4"}}},"commit":{"parents":[{"commit":"4f3c1fe197824325d0a55b0e2acd6b6c4d5c0b33","subject":"Removing unused and duplicate config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f3c1fe197824325d0a55b0e2acd6b6c4d5c0b33"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 02:02:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 19:44:48.000000000","tz":-420},"subject":"Change configuration to be consistent with underlines","message":"Change configuration to be consistent with underlines\n\nThis patch simply changes all configuration names to\nconsitently use underlines over dashes to make it\neasier to search the code.\n\nChange-Id: Id802d981de5577a54ff1b5b71e9e134d1175bbf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fe71d1f9d58ecdc1f7d432edab4137f41defd49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fe71d1f9d58ecdc1f7d432edab4137f41defd49"}]},"branch":"refs/heads/master"},"8c8c92a20cc18552d13ecae6cd7f1ddc08343306":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-14 02:40:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/664182/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/664182/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/664182/5"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 02:02:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 02:40:24.000000000","tz":-420},"subject":"Change configuration to be consistent with underlines","message":"Change configuration to be consistent with underlines\n\nThis patch simply changes all configuration names to\nconsitently use underlines over dashes to make it\neasier to search the code.\n\nChange-Id: Id802d981de5577a54ff1b5b71e9e134d1175bbf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8c8c92a20cc18552d13ecae6cd7f1ddc08343306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8c8c92a20cc18552d13ecae6cd7f1ddc08343306"}]},"branch":"refs/heads/master"},"edacd3cfa0e90ccb7217ad6791f8e470b55bacdc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-18 19:51:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/664182/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/664182/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/664182/6"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 02:02:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 19:51:38.000000000","tz":0},"subject":"Change configuration to be consistent with underlines","message":"Change configuration to be consistent with underlines\n\nThis patch simply changes all configuration names to\nconsitently use underlines over dashes to make it\neasier to search the code.\n\nChange-Id: Id802d981de5577a54ff1b5b71e9e134d1175bbf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/edacd3cfa0e90ccb7217ad6791f8e470b55bacdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/edacd3cfa0e90ccb7217ad6791f8e470b55bacdc"}]},"branch":"refs/heads/master"},"2bf9832f320871175847c126d33a4cefbc0412f8":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-06-24 00:57:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/664182/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/664182/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/664182/7"}}},"commit":{"parents":[{"commit":"a1d2f371093a90d0bbbbf1b3bcba04ebeb1a92a4","subject":"Merge \"Only register dynamic pool options when running designate-manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a1d2f371093a90d0bbbbf1b3bcba04ebeb1a92a4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 02:02:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-24 00:57:04.000000000","tz":0},"subject":"Change configuration to be consistent with underlines","message":"Change configuration to be consistent with underlines\n\nThis patch simply changes all configuration names to\nconsitently use underlines over dashes to make it\neasier to search the code.\n\nChange-Id: Id802d981de5577a54ff1b5b71e9e134d1175bbf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2bf9832f320871175847c126d33a4cefbc0412f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2bf9832f320871175847c126d33a4cefbc0412f8"}]},"branch":"refs/heads/master"},"c5bc8947814b57de9b9bae9c67a201bb59d30ae8":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-06-27 18:14:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/664182/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/664182/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/664182/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/664182/8"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-10 02:02:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-27 18:14:01.000000000","tz":0},"subject":"Change configuration to be consistent with underlines","message":"Change configuration to be consistent with underlines\n\nThis patch simply changes all configuration names to\nconsitently use underlines over dashes to make it\neasier to search the code.\n\nChange-Id: Id802d981de5577a54ff1b5b71e9e134d1175bbf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c5bc8947814b57de9b9bae9c67a201bb59d30ae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c5bc8947814b57de9b9bae9c67a201bb59d30ae8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
