)]}'
{"id":"openstack%2Fdesignate~664743","triplet_id":"openstack%2Fdesignate~master~Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7","project":"openstack/designate","branch":"master","topic":"improved_tests","hashtags":[],"change_id":"Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7","subject":"Improved worker test coverage","status":"MERGED","created":"2019-06-11 23:04:04.000000000","updated":"2019-06-21 20:35:22.000000000","submitted":"2019-06-21 20:35:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664743-1561149322671-b95397b7","meta_rev_id":"d1164488f82e3cfd6f89459279c63f585f9dbace","_number":664743,"virtual_id_number":664743,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":2,"date":"2019-06-21 20:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-19 13:42:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-06-21 18:57:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-21 19:29:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 13:42:04.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-21 18:57:45.000000000","updated_by":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"reviewer":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"state":"REVIEWER"},{"updated":"2019-06-21 20:35:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ccab94ed0f21235662d6f9e3497c6de861f35d2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-11 23:04:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de6ad7d319ed5f863120307e0ccfc103ecf6f230","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-11 23:04:18.000000000","message":"Topic changed from zone_tests to improved_tests","accounts_in_message":[],"_revision_number":1},{"id":"a434a061e0f05abd8db88c9c230feba123c9506c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-11 23:13:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c2eb7d553ea23b56199432f555c2a3127173116f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-11 23:14:34.000000000","message":"Topic changed from zone_tests to improved_tests","accounts_in_message":[],"_revision_number":2},{"id":"1b6935ab018db9e95ef6f93b4ad6ca4ab00e9a60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:13:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/43/664743/2/check/designate-bind9/f921c4f/ : SUCCESS in 45m 07s\n- designate-pdns4 http://logs.openstack.org/43/664743/2/check/designate-pdns4/e64df4b/ : SUCCESS in 38m 38s\n- designate-bind9-manager-model http://logs.openstack.org/43/664743/2/check/designate-bind9-manager-model/2a3e900/ : SUCCESS in 40m 43s\n- designate-bind9-py27 http://logs.openstack.org/43/664743/2/check/designate-bind9-py27/863a1fb/ : SUCCESS in 44m 07s\n- designate-pdns4-py27 http://logs.openstack.org/43/664743/2/check/designate-pdns4-py27/b02f578/ : SUCCESS in 39m 21s\n- designate-pdns4-postgres http://logs.openstack.org/43/664743/2/check/designate-pdns4-postgres/c98afda/ : SUCCESS in 39m 10s\n- designate-grenade-pdns4 http://logs.openstack.org/43/664743/2/check/designate-grenade-pdns4/2629da1/ : SUCCESS in 57m 05s\n- openstack-tox-cover http://logs.openstack.org/43/664743/2/check/openstack-tox-cover/b3a7454/cover/ : SUCCESS in 7m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/43/664743/2/check/openstack-tox-lower-constraints/2a9ed5c/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/43/664743/2/check/openstack-tox-pep8/0b0a804/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/43/664743/2/check/openstack-tox-py27/76db01b/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/43/664743/2/check/openstack-tox-py36/1ddbcd5/ : SUCCESS in 5m 28s\n- openstack-tox-py37 http://logs.openstack.org/43/664743/2/check/openstack-tox-py37/3893b47/ : SUCCESS in 9m 43s\n- openstack-tox-docs http://logs.openstack.org/43/664743/2/check/openstack-tox-docs/8a9afa4/html/ : SUCCESS in 5m 03s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/43/664743/2/check/neutron-tempest-plugin-designate-scenario/795c8fe/ : SUCCESS in 37m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ad489069fa02b4996732f641be4295d9046b213a","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-06-18 00:40:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3d1f9f01606ea99a0edea49dc97478e961d83b7f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 18:41:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0e38b61b2efc913c6d6c56daab7cc5699b51ea7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:46:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/43/664743/3/check/designate-bind9/f3f7f65/ : SUCCESS in 38m 28s\n- designate-pdns4 http://logs.openstack.org/43/664743/3/check/designate-pdns4/80cec8f/ : SUCCESS in 40m 09s\n- designate-bind9-manager-model http://logs.openstack.org/43/664743/3/check/designate-bind9-manager-model/c4f235b/ : SUCCESS in 41m 54s\n- designate-bind9-py27 http://logs.openstack.org/43/664743/3/check/designate-bind9-py27/c8690f5/ : SUCCESS in 40m 29s\n- designate-pdns4-py27 http://logs.openstack.org/43/664743/3/check/designate-pdns4-py27/ba73f2d/ : SUCCESS in 44m 28s\n- designate-pdns4-postgres http://logs.openstack.org/43/664743/3/check/designate-pdns4-postgres/0e46a36/ : SUCCESS in 38m 52s\n- designate-grenade-pdns4 http://logs.openstack.org/43/664743/3/check/designate-grenade-pdns4/0c4fdd7/ : SUCCESS in 49m 45s\n- openstack-tox-cover http://logs.openstack.org/43/664743/3/check/openstack-tox-cover/148b908/cover/ : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/43/664743/3/check/openstack-tox-lower-constraints/0de3081/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/43/664743/3/check/openstack-tox-pep8/cdff150/ : SUCCESS in 7m 22s\n- openstack-tox-py27 http://logs.openstack.org/43/664743/3/check/openstack-tox-py27/2f12f55/ : SUCCESS in 6m 35s\n- openstack-tox-py36 http://logs.openstack.org/43/664743/3/check/openstack-tox-py36/53fd084/ : SUCCESS in 5m 38s\n- openstack-tox-py37 http://logs.openstack.org/43/664743/3/check/openstack-tox-py37/c36ee29/ : SUCCESS in 6m 13s\n- openstack-tox-docs http://logs.openstack.org/43/664743/3/check/openstack-tox-docs/279631c/html/ : SUCCESS in 5m 07s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/43/664743/3/check/neutron-tempest-plugin-designate-scenario/57731a3/ : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":3},{"id":"957d0fa54e265c03f9c295b128826c4cdc746298","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 22:56:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"83296c4b102012d59fc57b4830c64272ad17a606","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 23:04:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"899ffaf2382748ccc6ebf35a590bc2fbea24c86f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 23:37:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7cb884e4e9ef4e5aff5ec7a68db1049a7bc713df","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 23:40:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bbbb7be6d229dafa2fed08701f1caae7e2893a34","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-19 00:10:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3aa9b644d0d8b3b3894b4e496755c091fcae26a5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-19 00:34:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d4d9640de4b71107c607d57387d64144ddff572e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-19 01:17:36.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"52d409658881177dd3fc89ed38842d7dd67a352d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-19 01:19:14.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"804c2596f27c46f6b66298a1ad68926de0dac119","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-19 01:22:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"342f728953dc4e08ee70c9274fd44ca653d4a84e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 02:37:14.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/43/664743/12/check/designate-bind9/9108060/ : SUCCESS in 39m 01s\n- designate-pdns4 http://logs.openstack.org/43/664743/12/check/designate-pdns4/38180ef/ : SUCCESS in 39m 22s\n- designate-bind9-manager-model http://logs.openstack.org/43/664743/12/check/designate-bind9-manager-model/35081ec/ : SUCCESS in 37m 09s\n- designate-bind9-py27 http://logs.openstack.org/43/664743/12/check/designate-bind9-py27/77f1ecf/ : SUCCESS in 41m 00s\n- designate-pdns4-py27 http://logs.openstack.org/43/664743/12/check/designate-pdns4-py27/e797048/ : SUCCESS in 39m 40s\n- designate-pdns4-postgres http://logs.openstack.org/43/664743/12/check/designate-pdns4-postgres/9b78e4e/ : SUCCESS in 38m 51s\n- designate-grenade-pdns4 http://logs.openstack.org/43/664743/12/check/designate-grenade-pdns4/1ece3d7/ : SUCCESS in 1h 02m 19s\n- openstack-tox-cover http://logs.openstack.org/43/664743/12/check/openstack-tox-cover/8af8162/cover/ : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/43/664743/12/check/openstack-tox-lower-constraints/26b64e9/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/43/664743/12/check/openstack-tox-pep8/b835048/ : SUCCESS in 6m 52s\n- openstack-tox-py27 http://logs.openstack.org/43/664743/12/check/openstack-tox-py27/95211f2/ : SUCCESS in 5m 56s\n- openstack-tox-py36 http://logs.openstack.org/43/664743/12/check/openstack-tox-py36/6b983d0/ : SUCCESS in 6m 11s\n- openstack-tox-py37 http://logs.openstack.org/43/664743/12/check/openstack-tox-py37/d0b5c8d/ : SUCCESS in 7m 52s\n- openstack-tox-docs http://logs.openstack.org/43/664743/12/check/openstack-tox-docs/135d872/html/ : SUCCESS in 6m 55s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/43/664743/12/check/neutron-tempest-plugin-designate-scenario/fe0d2d6/ : SUCCESS in 37m 13s","accounts_in_message":[],"_revision_number":12},{"id":"bf85f9bc8168046c0756fd06f9c24c2ecb651506","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-19 13:42:04.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"c67524e28470b1044dc559481a3d9a9fc0d6f72a","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-06-21 18:57:45.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"2d94fd88a8f0f43cf95127903c72d9297e28cc72","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-21 19:29:20.000000000","message":"Patch Set 12: Workflow+1\n\nThanks Scott and Graham. Gonna merge this in.","accounts_in_message":[],"_revision_number":12},{"id":"cb26a1686ff76390ef1ae0a1010ee2f1a4291ebb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 19:29:40.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"0dffe77a81363d75afe97b21f8d7e4de4a59c482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 20:35:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"d1164488f82e3cfd6f89459279c63f585f9dbace","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 20:35:22.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/43/664743/12/gate/designate-bind9/e7b23a7/ : SUCCESS in 40m 20s\n- designate-pdns4 http://logs.openstack.org/43/664743/12/gate/designate-pdns4/aa29e35/ : SUCCESS in 37m 35s\n- designate-bind9-manager-model http://logs.openstack.org/43/664743/12/gate/designate-bind9-manager-model/40d35de/ : SUCCESS in 37m 41s\n- designate-bind9-py27 http://logs.openstack.org/43/664743/12/gate/designate-bind9-py27/7fd6a19/ : SUCCESS in 43m 57s\n- designate-pdns4-py27 http://logs.openstack.org/43/664743/12/gate/designate-pdns4-py27/6be4ba3/ : SUCCESS in 38m 48s\n- designate-pdns4-postgres http://logs.openstack.org/43/664743/12/gate/designate-pdns4-postgres/894cbe4/ : SUCCESS in 43m 08s\n- designate-grenade-pdns4 http://logs.openstack.org/43/664743/12/gate/designate-grenade-pdns4/6e8557e/ : SUCCESS in 59m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/43/664743/12/gate/openstack-tox-lower-constraints/e70b464/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/43/664743/12/gate/openstack-tox-pep8/4143029/ : SUCCESS in 7m 09s\n- openstack-tox-py27 http://logs.openstack.org/43/664743/12/gate/openstack-tox-py27/3c21efa/ : SUCCESS in 10m 32s\n- openstack-tox-py36 http://logs.openstack.org/43/664743/12/gate/openstack-tox-py36/bcfa7f2/ : SUCCESS in 5m 21s\n- openstack-tox-py37 http://logs.openstack.org/43/664743/12/gate/openstack-tox-py37/8002dfd/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/43/664743/12/gate/openstack-tox-docs/03becd0/html/ : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"511c08fb76a9a7ddd4ecc66b904d48ab7240b692","revisions":{"77271ebb8f655b2136dce004738745033150493c":{"kind":"REWORK","_number":1,"created":"2019-06-11 23:04:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/1"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 23:04:04.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Minor cleanups.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77271ebb8f655b2136dce004738745033150493c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77271ebb8f655b2136dce004738745033150493c"}]},"branch":"refs/heads/master"},"1857def013f94156fb4dd35cf2389f25f42c8587":{"kind":"REWORK","_number":2,"created":"2019-06-11 23:13:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/2"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 23:13:33.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1857def013f94156fb4dd35cf2389f25f42c8587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1857def013f94156fb4dd35cf2389f25f42c8587"}]},"branch":"refs/heads/master"},"fbeaf13a958a87ddc6c59f71111a0ef5ebe0b3d6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-18 18:41:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/3"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 18:41:03.000000000","tz":0},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fbeaf13a958a87ddc6c59f71111a0ef5ebe0b3d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fbeaf13a958a87ddc6c59f71111a0ef5ebe0b3d6"}]},"branch":"refs/heads/master"},"2c3286458f9cb6622fa074b01d69082e80ef21cb":{"kind":"REWORK","_number":4,"created":"2019-06-18 22:56:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/4"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 22:56:02.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c3286458f9cb6622fa074b01d69082e80ef21cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c3286458f9cb6622fa074b01d69082e80ef21cb"}]},"branch":"refs/heads/master"},"805f98be1e13d45f62c1d3b20838617f651aeb03":{"kind":"REWORK","_number":5,"created":"2019-06-18 23:04:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/5"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 23:04:27.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/805f98be1e13d45f62c1d3b20838617f651aeb03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/805f98be1e13d45f62c1d3b20838617f651aeb03"}]},"branch":"refs/heads/master"},"2be041087d8b92b4819f4b88fb0d17bc5337de8a":{"kind":"REWORK","_number":6,"created":"2019-06-18 23:37:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/6"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 23:37:17.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2be041087d8b92b4819f4b88fb0d17bc5337de8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2be041087d8b92b4819f4b88fb0d17bc5337de8a"}]},"branch":"refs/heads/master"},"d5530b22e0d05807fd18c6d55892e25746b0a0d7":{"kind":"REWORK","_number":7,"created":"2019-06-18 23:40:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/7"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 23:40:48.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5530b22e0d05807fd18c6d55892e25746b0a0d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5530b22e0d05807fd18c6d55892e25746b0a0d7"}]},"branch":"refs/heads/master"},"e44ede5d648f6216a9d8cc03cd2279e85a2f5518":{"kind":"REWORK","_number":8,"created":"2019-06-19 00:10:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/8"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-19 00:10:10.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e44ede5d648f6216a9d8cc03cd2279e85a2f5518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e44ede5d648f6216a9d8cc03cd2279e85a2f5518"}]},"branch":"refs/heads/master"},"6f63691f0c224034b341d5c4605ca82b2312d06d":{"kind":"REWORK","_number":9,"created":"2019-06-19 00:34:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/9"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-19 00:34:28.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f63691f0c224034b341d5c4605ca82b2312d06d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f63691f0c224034b341d5c4605ca82b2312d06d"}]},"branch":"refs/heads/master"},"0f8ec3c42e1ee906648434962e54ae2cbc9c1482":{"kind":"REWORK","_number":10,"created":"2019-06-19 01:17:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/10"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-19 01:17:32.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f8ec3c42e1ee906648434962e54ae2cbc9c1482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f8ec3c42e1ee906648434962e54ae2cbc9c1482"}]},"branch":"refs/heads/master"},"88cf57f91d920ad7ffd3a23570ec53b0e78fa455":{"kind":"REWORK","_number":11,"created":"2019-06-19 01:19:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/11"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-19 01:19:01.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88cf57f91d920ad7ffd3a23570ec53b0e78fa455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88cf57f91d920ad7ffd3a23570ec53b0e78fa455"}]},"branch":"refs/heads/master"},"511c08fb76a9a7ddd4ecc66b904d48ab7240b692":{"kind":"REWORK","_number":12,"created":"2019-06-19 01:22:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/664743/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/664743/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/664743/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/664743/12"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-11 22:53:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-19 01:22:13.000000000","tz":-420},"subject":"Improved worker test coverage","message":"Improved worker test coverage\n\n* Added additional test coverage.\n* Cleaned up to match test convention.\n\nChange-Id: Ie76238b18ab3bd6cbe89456fc0abfc18a35a7ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/511c08fb76a9a7ddd4ecc66b904d48ab7240b692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/511c08fb76a9a7ddd4ecc66b904d48ab7240b692"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
