)]}'
{"id":"openstack%2Fdesignate~665268","triplet_id":"openstack%2Fdesignate~master~I16385361ec25025aa533df80793dfc4d22151cbb","project":"openstack/designate","branch":"master","topic":"fix_topic","hashtags":[],"change_id":"I16385361ec25025aa533df80793dfc4d22151cbb","subject":"Fixed renaming topic breaking service","status":"MERGED","created":"2019-06-14 00:29:43.000000000","updated":"2019-06-17 14:11:44.000000000","submitted":"2019-06-17 14:11:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665268-1560780704257-e4049d2a","meta_rev_id":"c5f7cfb4fe30843b195c72ab564e20d8d8c06ed3","_number":665268,"virtual_id_number":665268,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-06-17 14:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-17 07:03:57.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-17 12:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-06-17 08:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-06-17 12:44:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 08:25:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-06-17 12:44:33.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-17 14:11:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72af9e0cce51d2178a74c7fa9fb93e9247154b04","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 00:29:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efee3aed4a62f76fca0ecd4501251dd0a4a17249","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 00:31:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"16953aed68f6b12afe7237f659000fb6ea2af388","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 01:18:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/68/665268/2/check/designate-bind9/1a8ad77/ : FAILURE in 31m 51s\n- designate-pdns4 http://logs.openstack.org/68/665268/2/check/designate-pdns4/c711571/ : FAILURE in 30m 14s\n- designate-bind9-manager-model http://logs.openstack.org/68/665268/2/check/designate-bind9-manager-model/8ef2949/ : FAILURE in 34m 41s\n- designate-bind9-py27 http://logs.openstack.org/68/665268/2/check/designate-bind9-py27/8cf16d1/ : FAILURE in 38m 49s\n- designate-pdns4-py27 http://logs.openstack.org/68/665268/2/check/designate-pdns4-py27/9acc630/ : FAILURE in 29m 08s\n- designate-pdns4-postgres http://logs.openstack.org/68/665268/2/check/designate-pdns4-postgres/60a9681/ : FAILURE in 34m 11s\n- designate-grenade-pdns4 http://logs.openstack.org/68/665268/2/check/designate-grenade-pdns4/ada027b/ : FAILURE in 43m 38s\n- openstack-tox-cover http://logs.openstack.org/68/665268/2/check/openstack-tox-cover/b880c55/ : FAILURE in 6m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/665268/2/check/openstack-tox-lower-constraints/6273406/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/68/665268/2/check/openstack-tox-pep8/831cc87/ : FAILURE in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/68/665268/2/check/openstack-tox-py27/f56b14a/ : FAILURE in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/68/665268/2/check/openstack-tox-py36/097bb14/ : FAILURE in 5m 05s\n- openstack-tox-py37 http://logs.openstack.org/68/665268/2/check/openstack-tox-py37/55b70ed/ : FAILURE in 6m 09s\n- openstack-tox-docs http://logs.openstack.org/68/665268/2/check/openstack-tox-docs/c1b3a39/html/ : SUCCESS in 4m 54s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/68/665268/2/check/neutron-tempest-plugin-designate-scenario/ef1d998/ : FAILURE in 30m 25s","accounts_in_message":[],"_revision_number":2},{"id":"0e4a9a2236d31836a48b4ed4444bb089b4a7e79c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 01:25:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f972762ecd0088b78e1181ed8da01d4da08af6ac","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 01:40:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"483378ee7f97c6e38bf708572e5b908ba82f524d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 01:45:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0354ffd86e644454e5da20522491d7def9d15d5e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 02:39:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"63a2e2b1cd244f8dc1cc1e1e74be5f477ee142f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:45:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/68/665268/6/check/designate-bind9/b4fee5b/ : SUCCESS in 42m 26s\n- designate-pdns4 http://logs.openstack.org/68/665268/6/check/designate-pdns4/0778be6/ : SUCCESS in 40m 03s\n- designate-bind9-manager-model http://logs.openstack.org/68/665268/6/check/designate-bind9-manager-model/86f5fea/ : SUCCESS in 42m 16s\n- designate-bind9-py27 http://logs.openstack.org/68/665268/6/check/designate-bind9-py27/cb497c5/ : SUCCESS in 43m 32s\n- designate-pdns4-py27 http://logs.openstack.org/68/665268/6/check/designate-pdns4-py27/7e052bf/ : SUCCESS in 36m 51s\n- designate-pdns4-postgres http://logs.openstack.org/68/665268/6/check/designate-pdns4-postgres/6aabfe3/ : SUCCESS in 42m 48s\n- designate-grenade-pdns4 http://logs.openstack.org/68/665268/6/check/designate-grenade-pdns4/1a98218/ : SUCCESS in 55m 40s\n- openstack-tox-cover http://logs.openstack.org/68/665268/6/check/openstack-tox-cover/ea63bf2/cover/ : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/665268/6/check/openstack-tox-lower-constraints/097cbd1/ : SUCCESS in 7m 12s\n- openstack-tox-pep8 http://logs.openstack.org/68/665268/6/check/openstack-tox-pep8/bb97c3e/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/68/665268/6/check/openstack-tox-py27/588cfc4/ : SUCCESS in 8m 27s\n- openstack-tox-py36 http://logs.openstack.org/68/665268/6/check/openstack-tox-py36/883cbc0/ : SUCCESS in 6m 49s\n- openstack-tox-py37 http://logs.openstack.org/68/665268/6/check/openstack-tox-py37/325f101/ : SUCCESS in 6m 37s\n- openstack-tox-docs http://logs.openstack.org/68/665268/6/check/openstack-tox-docs/efdc069/html/ : SUCCESS in 6m 11s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/68/665268/6/check/neutron-tempest-plugin-designate-scenario/0cbc9cf/ : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":6},{"id":"e8c00ce323b20c80ba287f8779377e42bfbf4115","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 04:39:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1d5a58fc753d3d279e85abce7054a516df9988d1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 05:15:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7306ddd9f3e8666a1b3dc6815e144f93c50430c8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 05:19:49.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"14587e92cbe0972299f1400a10725d3ac89b8122","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 06:22:46.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/68/665268/9/check/designate-bind9/5e559ab/ : SUCCESS in 40m 38s\n- designate-pdns4 http://logs.openstack.org/68/665268/9/check/designate-pdns4/b1b1d6d/ : SUCCESS in 38m 58s\n- designate-bind9-manager-model http://logs.openstack.org/68/665268/9/check/designate-bind9-manager-model/13d6005/ : SUCCESS in 45m 25s\n- designate-bind9-py27 http://logs.openstack.org/68/665268/9/check/designate-bind9-py27/0759e45/ : POST_FAILURE in 41m 30s\n- designate-pdns4-py27 http://logs.openstack.org/68/665268/9/check/designate-pdns4-py27/2c0b126/ : SUCCESS in 33m 18s\n- designate-pdns4-postgres http://logs.openstack.org/68/665268/9/check/designate-pdns4-postgres/f939e80/ : SUCCESS in 34m 57s\n- designate-grenade-pdns4 http://logs.openstack.org/68/665268/9/check/designate-grenade-pdns4/8740ba0/ : SUCCESS in 56m 28s\n- openstack-tox-cover http://logs.openstack.org/68/665268/9/check/openstack-tox-cover/d717c95/cover/ : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/665268/9/check/openstack-tox-lower-constraints/ae28995/ : SUCCESS in 6m 41s\n- openstack-tox-pep8 http://logs.openstack.org/68/665268/9/check/openstack-tox-pep8/9d6d962/ : SUCCESS in 6m 55s\n- openstack-tox-py27 http://logs.openstack.org/68/665268/9/check/openstack-tox-py27/ed36c8c/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/68/665268/9/check/openstack-tox-py36/20274b8/ : SUCCESS in 5m 30s\n- openstack-tox-py37 http://logs.openstack.org/68/665268/9/check/openstack-tox-py37/3581ff7/ : SUCCESS in 6m 45s\n- openstack-tox-docs http://logs.openstack.org/68/665268/9/check/openstack-tox-docs/2702816/html/ : SUCCESS in 5m 49s\n- build-openstack-releasenotes http://logs.openstack.org/68/665268/9/check/build-openstack-releasenotes/0fcc18f/html/ : SUCCESS in 6m 18s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/68/665268/9/check/neutron-tempest-plugin-designate-scenario/0096961/ : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":9},{"id":"2c61e3ecac1435eac735beeb14c9478a443d04af","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 06:23:27.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"db0c8c96d6cf0f9bfc19b36b8eae24934fd4a6cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 07:56:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/68/665268/9/check/designate-bind9/0a34adc/ : SUCCESS in 51m 01s\n- designate-pdns4 http://logs.openstack.org/68/665268/9/check/designate-pdns4/9a34bff/ : SUCCESS in 34m 12s\n- designate-bind9-manager-model http://logs.openstack.org/68/665268/9/check/designate-bind9-manager-model/8187008/ : SUCCESS in 40m 33s\n- designate-bind9-py27 http://logs.openstack.org/68/665268/9/check/designate-bind9-py27/16651ef/ : SUCCESS in 39m 56s\n- designate-pdns4-py27 http://logs.openstack.org/68/665268/9/check/designate-pdns4-py27/13cb151/ : SUCCESS in 32m 00s\n- designate-pdns4-postgres http://logs.openstack.org/68/665268/9/check/designate-pdns4-postgres/12df93e/ : SUCCESS in 37m 55s\n- designate-grenade-pdns4 http://logs.openstack.org/68/665268/9/check/designate-grenade-pdns4/322d959/ : SUCCESS in 1h 01m 30s\n- openstack-tox-cover http://logs.openstack.org/68/665268/9/check/openstack-tox-cover/d8edcc3/cover/ : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/665268/9/check/openstack-tox-lower-constraints/bfe1fb2/ : SUCCESS in 6m 45s\n- openstack-tox-pep8 http://logs.openstack.org/68/665268/9/check/openstack-tox-pep8/0aca5e7/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/68/665268/9/check/openstack-tox-py27/bb6c084/ : SUCCESS in 5m 54s\n- openstack-tox-py36 http://logs.openstack.org/68/665268/9/check/openstack-tox-py36/650d8c5/ : SUCCESS in 6m 16s\n- openstack-tox-py37 http://logs.openstack.org/68/665268/9/check/openstack-tox-py37/ca04ca1/ : SUCCESS in 8m 26s\n- openstack-tox-docs http://logs.openstack.org/68/665268/9/check/openstack-tox-docs/ab50fa1/html/ : SUCCESS in 5m 01s\n- build-openstack-releasenotes http://logs.openstack.org/68/665268/9/check/build-openstack-releasenotes/9d259b4/html/ : SUCCESS in 6m 32s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/68/665268/9/check/neutron-tempest-plugin-designate-scenario/9571a4f/ : SUCCESS in 37m 19s","accounts_in_message":[],"_revision_number":9},{"id":"a8410f99c7c525485262c43e8171bc81284e9519","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-06-17 06:44:56.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9f6d230adcdefcb971582a4a75e3f7065966f683","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 07:00:59.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"6cec55d46b45c5bda5cf14ca8500e959817df2ac","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 07:02:03.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"b4cee723e974ddcad2fd1edfefa6ef88125e2aef","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 07:02:25.000000000","message":"Patch Set 11:\n\n\u003e (2 comments)\n\nYep - thanks good catch","accounts_in_message":[],"_revision_number":11},{"id":"3cc7e0907f7c25f84bd51d83ac37f5b041dc989c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 07:03:29.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"1d1cc7d11039708f7ea162f30936bddc18263d5b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 07:03:57.000000000","message":"Patch Set 12:\n\nSorry not sure why I thought this was a good idea to fix on the iPad.","accounts_in_message":[],"_revision_number":12},{"id":"d5b3bb9f17918c87b76f68891e836c5cd3c884fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 08:19:30.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/68/665268/12/check/designate-bind9/c16bf26/ : SUCCESS in 36m 56s\n- designate-pdns4 http://logs.openstack.org/68/665268/12/check/designate-pdns4/f70c903/ : SUCCESS in 39m 11s\n- designate-bind9-manager-model http://logs.openstack.org/68/665268/12/check/designate-bind9-manager-model/3810c2d/ : SUCCESS in 37m 16s\n- designate-bind9-py27 http://logs.openstack.org/68/665268/12/check/designate-bind9-py27/788b85a/ : SUCCESS in 41m 14s\n- designate-pdns4-py27 http://logs.openstack.org/68/665268/12/check/designate-pdns4-py27/30daea8/ : SUCCESS in 36m 30s\n- designate-pdns4-postgres http://logs.openstack.org/68/665268/12/check/designate-pdns4-postgres/34a0af6/ : SUCCESS in 35m 35s\n- designate-grenade-pdns4 http://logs.openstack.org/68/665268/12/check/designate-grenade-pdns4/4ba5707/ : SUCCESS in 55m 05s\n- openstack-tox-cover http://logs.openstack.org/68/665268/12/check/openstack-tox-cover/87e9f3c/cover/ : SUCCESS in 6m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/665268/12/check/openstack-tox-lower-constraints/09d0130/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 http://logs.openstack.org/68/665268/12/check/openstack-tox-pep8/4a5aa59/ : SUCCESS in 5m 41s\n- openstack-tox-py27 http://logs.openstack.org/68/665268/12/check/openstack-tox-py27/27902ae/ : SUCCESS in 5m 57s\n- openstack-tox-py36 http://logs.openstack.org/68/665268/12/check/openstack-tox-py36/0a6e5c9/ : SUCCESS in 5m 27s\n- openstack-tox-py37 http://logs.openstack.org/68/665268/12/check/openstack-tox-py37/49738fb/ : SUCCESS in 6m 06s\n- openstack-tox-docs http://logs.openstack.org/68/665268/12/check/openstack-tox-docs/51f27a9/html/ : SUCCESS in 5m 16s\n- build-openstack-releasenotes http://logs.openstack.org/68/665268/12/check/build-openstack-releasenotes/c094004/html/ : SUCCESS in 6m 33s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/68/665268/12/check/neutron-tempest-plugin-designate-scenario/4a56e07/ : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":12},{"id":"7b6f26cab440f233f778ad5dd1d864a18a6d4445","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-06-17 08:25:20.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"eca7d3c16e92f950ad4045303b3fc42e57584cc4","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-17 12:44:33.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"7a3da44caf487854a6934fbc4d23aa513609312b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 12:44:45.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"baadbec5c5bda19bad9a72c66996c2a955eb162d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:11:43.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/68/665268/12/gate/designate-bind9/6eaea8a/ : SUCCESS in 38m 47s\n- designate-pdns4 http://logs.openstack.org/68/665268/12/gate/designate-pdns4/b341289/ : SUCCESS in 37m 25s\n- designate-bind9-manager-model http://logs.openstack.org/68/665268/12/gate/designate-bind9-manager-model/f48e6fe/ : SUCCESS in 39m 09s\n- designate-bind9-py27 http://logs.openstack.org/68/665268/12/gate/designate-bind9-py27/f88e624/ : SUCCESS in 42m 27s\n- designate-pdns4-py27 http://logs.openstack.org/68/665268/12/gate/designate-pdns4-py27/7a91adb/ : SUCCESS in 44m 50s\n- designate-pdns4-postgres http://logs.openstack.org/68/665268/12/gate/designate-pdns4-postgres/16fed2b/ : SUCCESS in 40m 32s\n- designate-grenade-pdns4 http://logs.openstack.org/68/665268/12/gate/designate-grenade-pdns4/f35fc67/ : SUCCESS in 57m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/665268/12/gate/openstack-tox-lower-constraints/c0fcb7a/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/68/665268/12/gate/openstack-tox-pep8/eb3812a/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/68/665268/12/gate/openstack-tox-py27/5914382/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/68/665268/12/gate/openstack-tox-py36/82e018e/ : SUCCESS in 5m 18s\n- openstack-tox-py37 http://logs.openstack.org/68/665268/12/gate/openstack-tox-py37/7c68888/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/68/665268/12/gate/openstack-tox-docs/e02a288/html/ : SUCCESS in 5m 22s\n- build-openstack-releasenotes http://logs.openstack.org/68/665268/12/gate/build-openstack-releasenotes/a16a19e/html/ : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":12},{"id":"7941d79ddb9147ce584bd94dc520940869aaa946","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:11:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"fcb3290be7caf04b6b065f8a25bc4b8a3fd9113a","revisions":{"db8e4c8863ff4ede7f84ae617f87a7719e7ef246":{"kind":"REWORK","_number":1,"created":"2019-06-14 00:29:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/1"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:29:54.000000000","tz":-420},"subject":"[WIP] Fixing renaming topic breaking multiple services","message":"[WIP] Fixing renaming topic breaking multiple services\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n* Renames cfg option [service:\u003cname\u003e}\u003cname\u003e_topic to [service:\u003cname\u003e]topic\n* Adds test coverage.\n\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/db8e4c8863ff4ede7f84ae617f87a7719e7ef246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/db8e4c8863ff4ede7f84ae617f87a7719e7ef246"}]},"branch":"refs/heads/master"},"affa915556bf70c749ecc14fb3dc8d9a8db9d5b8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-14 00:31:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/2"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:31:12.000000000","tz":-420},"subject":"[WIP] Fixing renaming topic breaking multiple services","message":"[WIP] Fixing renaming topic breaking multiple services\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e}\u003cname\u003e_topic to [service:\u003cname\u003e]topic\n* Adds test coverage.\n\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/affa915556bf70c749ecc14fb3dc8d9a8db9d5b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/affa915556bf70c749ecc14fb3dc8d9a8db9d5b8"}]},"branch":"refs/heads/master"},"ec1da5682b8acc48412d0c57574a85b24a200bf3":{"kind":"REWORK","_number":3,"created":"2019-06-14 01:25:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/3"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 01:25:29.000000000","tz":-420},"subject":"[WIP] Fixing renaming topic breaking multiple services","message":"[WIP] Fixing renaming topic breaking multiple services\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e}\u003cname\u003e_topic to [service:\u003cname\u003e]topic\n* Added topic override for producer.\n* Adds test coverage.\n\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec1da5682b8acc48412d0c57574a85b24a200bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec1da5682b8acc48412d0c57574a85b24a200bf3"}]},"branch":"refs/heads/master"},"c04c6f850f0b81abf9dfe0157bd40bd2b4feb4b5":{"kind":"REWORK","_number":4,"created":"2019-06-14 01:40:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/4"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 01:40:53.000000000","tz":-420},"subject":"[WIP] Fixing renaming topic breaking multiple services","message":"[WIP] Fixing renaming topic breaking multiple services\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e}\u003cname\u003e_topic to [service:\u003cname\u003e]topic\n* Added topic override for producer.\n* Adds test coverage.\n\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c04c6f850f0b81abf9dfe0157bd40bd2b4feb4b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c04c6f850f0b81abf9dfe0157bd40bd2b4feb4b5"}]},"branch":"refs/heads/master"},"00e0cf967bb8ebee749687f7d765cf92a6e0b49d":{"kind":"REWORK","_number":5,"created":"2019-06-14 01:45:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/5"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 01:45:33.000000000","tz":-420},"subject":"[WIP] Fixing renaming topic breaking multiple services","message":"[WIP] Fixing renaming topic breaking multiple services\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e}\u003cname\u003e_topic to [service:\u003cname\u003e]topic\n* Added topic override for producer.\n* Adds test coverage.\n\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/00e0cf967bb8ebee749687f7d765cf92a6e0b49d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/00e0cf967bb8ebee749687f7d765cf92a6e0b49d"}]},"branch":"refs/heads/master"},"c38c534571b3bb1c39720b41cfcf1a7dd3d39a96":{"kind":"REWORK","_number":6,"created":"2019-06-14 02:39:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/6"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 02:38:47.000000000","tz":-420},"subject":"[WIP] Fixed renaming topic breaking service","message":"[WIP] Fixed renaming topic breaking service\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e].\u003cname\u003e_topic to [service:\u003cname\u003e].topic\n* Removed previously unused configs related to topics.\n* Added topic override for producer.\n* Adds test coverage.\n\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c38c534571b3bb1c39720b41cfcf1a7dd3d39a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c38c534571b3bb1c39720b41cfcf1a7dd3d39a96"}]},"branch":"refs/heads/master"},"2ddee071a4e40e504ecc30760583061171194bf2":{"kind":"REWORK","_number":7,"created":"2019-06-14 04:39:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/7"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 04:38:59.000000000","tz":-420},"subject":"Fixed renaming topic breaking service","message":"Fixed renaming topic breaking service\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e].\u003cname\u003e_topic to [service:\u003cname\u003e].topic\n* Removed previously unused configs related to topics.\n* Added topic override for producer.\n* Added test coverage.\n\nCloses-Bug: #1832799\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ddee071a4e40e504ecc30760583061171194bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ddee071a4e40e504ecc30760583061171194bf2"}]},"branch":"refs/heads/master"},"b022e85ceb18afa787032bca307f79bb93811427":{"kind":"REWORK","_number":8,"created":"2019-06-14 05:15:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/8"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 05:15:45.000000000","tz":-420},"subject":"Fixed renaming topic breaking service","message":"Fixed renaming topic breaking service\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e].\u003cname\u003e_topic to [service:\u003cname\u003e].topic\n* Removed previously unused configs related to topics.\n* Added topic override for producer.\n* Added test coverage.\n\nCloses-Bug: #1832799\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b022e85ceb18afa787032bca307f79bb93811427"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b022e85ceb18afa787032bca307f79bb93811427"}]},"branch":"refs/heads/master"},"e2a61e67909cfa823521c1999378c4deefa49b3c":{"kind":"REWORK","_number":9,"created":"2019-06-14 05:19:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/9"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 05:18:56.000000000","tz":-420},"subject":"Fixed renaming topic breaking service","message":"Fixed renaming topic breaking service\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e].\u003cname\u003e_topic to [service:\u003cname\u003e].topic\n* Removed previously unused configs related to topics.\n* Added topic override for producer.\n* Added test coverage.\n\nCloses-Bug: #1832799\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2a61e67909cfa823521c1999378c4deefa49b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2a61e67909cfa823521c1999378c4deefa49b3c"}]},"branch":"refs/heads/master"},"72704c0fb90004b3ba1843b3c9861e93f9f04f30":{"kind":"REWORK","_number":10,"created":"2019-06-17 07:00:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/10"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-17 07:00:46.000000000","tz":0},"subject":"Fixed renaming topic breaking service","message":"Fixed renaming topic breaking service\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e].\u003cname\u003e_topic to [service:\u003cname\u003e].topic\n* Removed previously unused configs related to topics.\n* Added topic override for producer.\n* Added test coverage.\n\nCloses-Bug: #1832799\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72704c0fb90004b3ba1843b3c9861e93f9f04f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72704c0fb90004b3ba1843b3c9861e93f9f04f30"}]},"branch":"refs/heads/master"},"18ced099d285eae8d55124e491dd6ae539d8eb55":{"kind":"REWORK","_number":11,"created":"2019-06-17 07:02:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/11"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-17 07:01:57.000000000","tz":0},"subject":"Fixed renaming topic breaking service","message":"Fixed renaming topic breaking service\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e].\u003cname\u003e_topic to [service:\u003cname\u003e].topic\n* Removed previously unused configs related to topics.\n* Added topic override for producer.\n* Added test coverage.\n\nCloses-Bug: #1832799\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/18ced099d285eae8d55124e491dd6ae539d8eb55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/18ced099d285eae8d55124e491dd6ae539d8eb55"}]},"branch":"refs/heads/master"},"fcb3290be7caf04b6b065f8a25bc4b8a3fd9113a":{"kind":"REWORK","_number":12,"created":"2019-06-17 07:03:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/68/665268/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/665268/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/665268/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/665268/12"}}},"commit":{"parents":[{"commit":"fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949","subject":"Merge \"Added sink unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc5700bd61da6ddeacd4ffa7f3ac15b3d05db949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 00:24:50.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-17 07:03:24.000000000","tz":0},"subject":"Fixed renaming topic breaking service","message":"Fixed renaming topic breaking service\n\nChanging the topic currently changes the service_name. This\nbreaks the service, as the service_name is also used to\nload the appropriate configuration seection for the service.\n\n\u003e self._service_config \u003d CONF[\u0027service:%s\u0027 % self.service_name]\n\n* Renames cfg option [service:\u003cname\u003e].\u003cname\u003e_topic to [service:\u003cname\u003e].topic\n* Removed previously unused configs related to topics.\n* Added topic override for producer.\n* Added test coverage.\n\nCloses-Bug: #1832799\nChange-Id: I16385361ec25025aa533df80793dfc4d22151cbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fcb3290be7caf04b6b065f8a25bc4b8a3fd9113a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fcb3290be7caf04b6b065f8a25bc4b8a3fd9113a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
