)]}'
{"id":"openstack%2Fdesignate~665311","triplet_id":"openstack%2Fdesignate~master~I007fc004fc533950c5eda3974f41f05c6db5b43f","project":"openstack/designate","branch":"master","topic":"central_cfg","hashtags":[],"change_id":"I007fc004fc533950c5eda3974f41f05c6db5b43f","subject":"Finalized centralization of configuration","status":"MERGED","created":"2019-06-14 06:35:46.000000000","updated":"2019-06-23 20:14:33.000000000","submitted":"2019-06-23 20:14:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665311-1561320873299-18733654","meta_rev_id":"a72eff75ee506d9064ccc416c4383756b447f8f7","_number":665311,"virtual_id_number":665311,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":2,"date":"2019-06-23 20:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-19 13:40:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-23 19:14:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 07:17:23.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2019-06-19 13:40:05.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-23 20:14:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b680aa1b899c6564f02b011b5e6e31d29d97e61e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 06:35:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c50240d91004c2ab6335a266c7a0bdd4cb6b923c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 06:47:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7159af3cd77e77860a9100a4a54d4b6dfbc67219","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 06:48:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba27b7721324f63b1a87536146c21f87b9987b25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 08:43:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/11/665311/2/check/designate-bind9/f6c17d7/ : SUCCESS in 40m 26s\n- designate-pdns4 http://logs.openstack.org/11/665311/2/check/designate-pdns4/ab98e79/ : SUCCESS in 37m 26s\n- designate-bind9-manager-model http://logs.openstack.org/11/665311/2/check/designate-bind9-manager-model/4902ea7/ : SUCCESS in 44m 34s\n- designate-bind9-py27 http://logs.openstack.org/11/665311/2/check/designate-bind9-py27/433e15f/ : SUCCESS in 1h 00m 39s\n- designate-pdns4-py27 http://logs.openstack.org/11/665311/2/check/designate-pdns4-py27/8da7bfd/ : SUCCESS in 45m 29s\n- designate-pdns4-postgres http://logs.openstack.org/11/665311/2/check/designate-pdns4-postgres/863b7c7/ : SUCCESS in 54m 24s\n- designate-grenade-pdns4 http://logs.openstack.org/11/665311/2/check/designate-grenade-pdns4/90c42e9/ : SUCCESS in 55m 31s\n- openstack-tox-cover http://logs.openstack.org/11/665311/2/check/openstack-tox-cover/d510117/cover/ : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/665311/2/check/openstack-tox-lower-constraints/dc1feb2/ : SUCCESS in 7m 34s\n- openstack-tox-pep8 http://logs.openstack.org/11/665311/2/check/openstack-tox-pep8/c4837a4/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/11/665311/2/check/openstack-tox-py27/eab1152/ : SUCCESS in 6m 09s\n- openstack-tox-py36 http://logs.openstack.org/11/665311/2/check/openstack-tox-py36/1ae5d2f/ : SUCCESS in 5m 50s\n- openstack-tox-py37 http://logs.openstack.org/11/665311/2/check/openstack-tox-py37/b97c6df/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/11/665311/2/check/openstack-tox-docs/fba4596/html/ : SUCCESS in 5m 08s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/665311/2/check/neutron-tempest-plugin-designate-scenario/0178eb2/ : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":2},{"id":"32fe5eb5f17c97f35534dcdb7e8d4ebc206436e1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 18:11:16.000000000","message":"Patch Set 2:\n\nA diff on the config before and after this shows no changes.","accounts_in_message":[],"_revision_number":2},{"id":"485280438720c2bf9b2448e478248950ecef3ed3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 18:41:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8fb6d05c8e0fc1d63a35869c2f26e32e9eaed0d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:44:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/11/665311/3/check/designate-bind9/d1a68a8/ : SUCCESS in 41m 24s\n- designate-pdns4 http://logs.openstack.org/11/665311/3/check/designate-pdns4/7c6397d/ : SUCCESS in 38m 30s\n- designate-bind9-manager-model http://logs.openstack.org/11/665311/3/check/designate-bind9-manager-model/83c51a8/ : SUCCESS in 58m 11s\n- designate-bind9-py27 http://logs.openstack.org/11/665311/3/check/designate-bind9-py27/a890713/ : SUCCESS in 39m 48s\n- designate-pdns4-py27 http://logs.openstack.org/11/665311/3/check/designate-pdns4-py27/443a2dc/ : SUCCESS in 45m 12s\n- designate-pdns4-postgres http://logs.openstack.org/11/665311/3/check/designate-pdns4-postgres/f796def/ : SUCCESS in 39m 01s\n- designate-grenade-pdns4 http://logs.openstack.org/11/665311/3/check/designate-grenade-pdns4/7ef7932/ : SUCCESS in 56m 52s\n- openstack-tox-cover http://logs.openstack.org/11/665311/3/check/openstack-tox-cover/3fbef65/cover/ : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/665311/3/check/openstack-tox-lower-constraints/3243948/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/11/665311/3/check/openstack-tox-pep8/a67fc3e/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/11/665311/3/check/openstack-tox-py27/60ea21d/ : SUCCESS in 6m 10s\n- openstack-tox-py36 http://logs.openstack.org/11/665311/3/check/openstack-tox-py36/cac9489/ : SUCCESS in 6m 41s\n- openstack-tox-py37 http://logs.openstack.org/11/665311/3/check/openstack-tox-py37/e3b6afa/ : SUCCESS in 6m 24s\n- openstack-tox-docs http://logs.openstack.org/11/665311/3/check/openstack-tox-docs/7b329c7/html/ : SUCCESS in 5m 32s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/665311/3/check/neutron-tempest-plugin-designate-scenario/ef532c6/ : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a432d740af8f567ef40bc42b523cb11c43b97992","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-19 13:40:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1bc420d99c065ec19386202a8c70c3f45ad2adc8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-23 19:14:30.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"de1f246f742da6e9675a094085fcbc48a58cc986","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 19:14:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a65e4d44038271d45963f5fcbb2f78e33f9baef5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 20:14:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/11/665311/3/gate/designate-bind9/7374bc0/ : SUCCESS in 38m 43s\n- designate-pdns4 http://logs.openstack.org/11/665311/3/gate/designate-pdns4/cd869cf/ : SUCCESS in 37m 44s\n- designate-bind9-manager-model http://logs.openstack.org/11/665311/3/gate/designate-bind9-manager-model/8bf9ec3/ : SUCCESS in 39m 22s\n- designate-bind9-py27 http://logs.openstack.org/11/665311/3/gate/designate-bind9-py27/247d9a0/ : SUCCESS in 41m 59s\n- designate-pdns4-py27 http://logs.openstack.org/11/665311/3/gate/designate-pdns4-py27/be71810/ : SUCCESS in 33m 29s\n- designate-pdns4-postgres http://logs.openstack.org/11/665311/3/gate/designate-pdns4-postgres/d1fdca0/ : SUCCESS in 43m 28s\n- designate-grenade-pdns4 http://logs.openstack.org/11/665311/3/gate/designate-grenade-pdns4/9250cb6/ : SUCCESS in 54m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/665311/3/gate/openstack-tox-lower-constraints/dfdd9d9/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/11/665311/3/gate/openstack-tox-pep8/60fee4b/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/11/665311/3/gate/openstack-tox-py27/cb2f42d/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/11/665311/3/gate/openstack-tox-py36/b7d356f/ : SUCCESS in 5m 11s\n- openstack-tox-py37 http://logs.openstack.org/11/665311/3/gate/openstack-tox-py37/5fd2d31/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/11/665311/3/gate/openstack-tox-docs/f5f68f3/html/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"43ddb22d25335d865917344a64762d6930899882","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 20:14:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"097ffec548cea59f725920593f07f205c938a8e7","revisions":{"ba3365d3ed29cf6827cd84f7c9ec396a91e56c03":{"kind":"REWORK","_number":1,"created":"2019-06-14 06:35:46.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/665311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/665311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/665311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/665311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/665311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/665311/1"}}},"commit":{"parents":[{"commit":"3294dd956f9c08f36f50dbff00105f1e8c0151e4","subject":"Merge \"Simplified metrics implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3294dd956f9c08f36f50dbff00105f1e8c0151e4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 06:31:20.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 06:33:38.000000000","tz":-420},"subject":"Finalized centralization of configuration","message":"Finalized centralization of configuration\n\nThis patch removes get_cfg_opts from the plugin class,\nand moves all configuration directly into designate.conf.\n\nThere are no changes in the generated configuration due\nto this patch.\n\nFinally, this complets the blueprint to centralize all\nconfiguration, and is a follow up to this patch\nI3be68bf294157631a9bac128df0841220020de96\n\nblueprint centralize-config-designate\n\nChange-Id: I007fc004fc533950c5eda3974f41f05c6db5b43f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ba3365d3ed29cf6827cd84f7c9ec396a91e56c03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ba3365d3ed29cf6827cd84f7c9ec396a91e56c03"}]},"branch":"refs/heads/master"},"64c17690034b614ea7656a53558e34ead65f929d":{"kind":"REWORK","_number":2,"created":"2019-06-14 06:47:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/665311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/665311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/665311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/665311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/665311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/665311/2"}}},"commit":{"parents":[{"commit":"3294dd956f9c08f36f50dbff00105f1e8c0151e4","subject":"Merge \"Simplified metrics implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3294dd956f9c08f36f50dbff00105f1e8c0151e4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 06:31:20.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 06:47:37.000000000","tz":-420},"subject":"Finalized centralization of configuration","message":"Finalized centralization of configuration\n\nThis patch removes get_cfg_opts from the plugin class,\nand moves all configuration directly into designate.conf.\n\nThere are no changes in the generated configuration due\nto this patch.\n\nFinally, this complets the blueprint to centralize all\nconfiguration, and is a follow up to this patch\nI3be68bf294157631a9bac128df0841220020de96\n\nblueprint centralize-config-designate\n\nChange-Id: I007fc004fc533950c5eda3974f41f05c6db5b43f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64c17690034b614ea7656a53558e34ead65f929d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64c17690034b614ea7656a53558e34ead65f929d"}]},"branch":"refs/heads/master"},"097ffec548cea59f725920593f07f205c938a8e7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-18 18:41:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/665311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/665311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/665311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/665311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/665311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/665311/3"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 06:31:20.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 18:41:10.000000000","tz":0},"subject":"Finalized centralization of configuration","message":"Finalized centralization of configuration\n\nThis patch removes get_cfg_opts from the plugin class,\nand moves all configuration directly into designate.conf.\n\nThere are no changes in the generated configuration due\nto this patch.\n\nFinally, this complets the blueprint to centralize all\nconfiguration, and is a follow up to this patch\nI3be68bf294157631a9bac128df0841220020de96\n\nblueprint centralize-config-designate\n\nChange-Id: I007fc004fc533950c5eda3974f41f05c6db5b43f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/097ffec548cea59f725920593f07f205c938a8e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/097ffec548cea59f725920593f07f205c938a8e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
