)]}'
{"id":"openstack%2Fdesignate~665317","triplet_id":"openstack%2Fdesignate~master~Ia9587880ca2428c0b3af074f0cb2a873859685ec","project":"openstack/designate","branch":"master","topic":"lazy_dynamic_pool","hashtags":[],"change_id":"Ia9587880ca2428c0b3af074f0cb2a873859685ec","subject":"Only register dynamic pool options when running designate-manage","status":"MERGED","created":"2019-06-14 07:07:54.000000000","updated":"2019-06-23 23:03:13.000000000","submitted":"2019-06-23 23:03:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665317-1561330993099-eb0806c3","meta_rev_id":"30dbeac9ca759408e62e77982399f3dd9f603688","_number":665317,"virtual_id_number":665317,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-06-23 23:03:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-19 13:41:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-23 19:14:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 13:41:33.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-23 23:03:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"394845645d85fd2f127867332e4bb84784cc23b0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 07:07:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e4fa3c44a4c4ac1ee8dd02530ed5bb1500370cf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 07:10:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0958febd175befbc2eeba723960df0307789df65","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 07:16:35.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fb1e707885f015f59c6c5a5174f34510719e0c39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 08:44:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/17/665317/3/check/designate-bind9/8bd2113/ : SUCCESS in 39m 20s\n- designate-pdns4 http://logs.openstack.org/17/665317/3/check/designate-pdns4/b0be6a1/ : SUCCESS in 41m 55s\n- designate-bind9-manager-model http://logs.openstack.org/17/665317/3/check/designate-bind9-manager-model/cd79bda/ : SUCCESS in 44m 39s\n- designate-bind9-py27 http://logs.openstack.org/17/665317/3/check/designate-bind9-py27/6492e15/ : SUCCESS in 41m 23s\n- designate-pdns4-py27 http://logs.openstack.org/17/665317/3/check/designate-pdns4-py27/796bbfe/ : SUCCESS in 40m 49s\n- designate-pdns4-postgres http://logs.openstack.org/17/665317/3/check/designate-pdns4-postgres/2b40ab3/ : SUCCESS in 38m 16s\n- designate-grenade-pdns4 http://logs.openstack.org/17/665317/3/check/designate-grenade-pdns4/f9c73f6/ : SUCCESS in 1h 00m 30s\n- openstack-tox-cover http://logs.openstack.org/17/665317/3/check/openstack-tox-cover/e474347/cover/ : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665317/3/check/openstack-tox-lower-constraints/215b8f5/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/17/665317/3/check/openstack-tox-pep8/4500ec0/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/17/665317/3/check/openstack-tox-py27/674c279/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/17/665317/3/check/openstack-tox-py36/11e8e4b/ : SUCCESS in 8m 08s\n- openstack-tox-py37 http://logs.openstack.org/17/665317/3/check/openstack-tox-py37/1b0a5e1/ : SUCCESS in 6m 42s\n- openstack-tox-docs http://logs.openstack.org/17/665317/3/check/openstack-tox-docs/13f3282/html/ : SUCCESS in 5m 03s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/17/665317/3/check/neutron-tempest-plugin-designate-scenario/23ee52c/ : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":3},{"id":"2b4b70404748bbc6684e9a41324d4304b4bb8422","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 19:51:24.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a10a31aff734eeb741c8f2073630fa80c0b4f0a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:01:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/17/665317/4/check/designate-bind9/463db82/ : SUCCESS in 44m 02s\n- designate-pdns4 http://logs.openstack.org/17/665317/4/check/designate-pdns4/e15bddb/ : SUCCESS in 42m 24s\n- designate-bind9-manager-model http://logs.openstack.org/17/665317/4/check/designate-bind9-manager-model/bfbeaaa/ : SUCCESS in 37m 11s\n- designate-bind9-py27 http://logs.openstack.org/17/665317/4/check/designate-bind9-py27/e60032b/ : SUCCESS in 37m 36s\n- designate-pdns4-py27 http://logs.openstack.org/17/665317/4/check/designate-pdns4-py27/ebe3736/ : SUCCESS in 41m 03s\n- designate-pdns4-postgres http://logs.openstack.org/17/665317/4/check/designate-pdns4-postgres/35d4cea/ : SUCCESS in 37m 49s\n- designate-grenade-pdns4 http://logs.openstack.org/17/665317/4/check/designate-grenade-pdns4/ff1709f/ : SUCCESS in 56m 08s\n- openstack-tox-cover http://logs.openstack.org/17/665317/4/check/openstack-tox-cover/28b5ca5/cover/ : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665317/4/check/openstack-tox-lower-constraints/0ee12ab/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/17/665317/4/check/openstack-tox-pep8/a3e299b/ : SUCCESS in 6m 21s\n- openstack-tox-py27 http://logs.openstack.org/17/665317/4/check/openstack-tox-py27/2be0360/ : SUCCESS in 5m 34s\n- openstack-tox-py36 http://logs.openstack.org/17/665317/4/check/openstack-tox-py36/312fa3a/ : SUCCESS in 6m 03s\n- openstack-tox-py37 http://logs.openstack.org/17/665317/4/check/openstack-tox-py37/3f762e3/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/17/665317/4/check/openstack-tox-docs/9fe224d/html/ : SUCCESS in 5m 08s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/17/665317/4/check/neutron-tempest-plugin-designate-scenario/2ee93f7/ : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":4},{"id":"226affe1cea0c17ef161bed2289c4d3c404616f1","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-19 13:41:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"91047181fea73df428caa57b445b6a22b6384bd2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-23 19:14:42.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9b30bcb9427e29246e7eff66a00acb986136bcc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 19:14:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6e3ff9aff53bca1ce75766aaab1a721ce02336aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 20:16:06.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/17/665317/4/gate/designate-bind9/d375310/ : SUCCESS in 36m 28s\n- designate-pdns4 http://logs.openstack.org/17/665317/4/gate/designate-pdns4/28747a1/ : SUCCESS in 35m 43s\n- designate-bind9-manager-model http://logs.openstack.org/17/665317/4/gate/designate-bind9-manager-model/f9e9dd9/ : SUCCESS in 39m 05s\n- designate-bind9-py27 http://logs.openstack.org/17/665317/4/gate/designate-bind9-py27/48b5aa3/ : SUCCESS in 40m 41s\n- designate-pdns4-py27 http://logs.openstack.org/17/665317/4/gate/designate-pdns4-py27/3e81527/ : SUCCESS in 34m 34s\n- designate-pdns4-postgres http://logs.openstack.org/17/665317/4/gate/designate-pdns4-postgres/a9306d2/ : SUCCESS in 36m 12s\n- designate-grenade-pdns4 http://logs.openstack.org/17/665317/4/gate/designate-grenade-pdns4/072ffa7/ : SUCCESS in 56m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665317/4/gate/openstack-tox-lower-constraints/05034b0/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/17/665317/4/gate/openstack-tox-pep8/4536d85/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/17/665317/4/gate/openstack-tox-py27/8e7d86a/ : FAILURE in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/17/665317/4/gate/openstack-tox-py36/952b502/ : SUCCESS in 5m 04s\n- openstack-tox-py37 http://logs.openstack.org/17/665317/4/gate/openstack-tox-py37/51889d8/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/17/665317/4/gate/openstack-tox-docs/36223dc/html/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":4},{"id":"fc516535145dfa9c0b34a8adc2890d60579bb7d3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-23 20:53:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7ec7fcbd5bb9dcc46fefa0ebdbc7996d2fc71a49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 22:00:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/17/665317/4/check/designate-bind9/2f10f18/ : SUCCESS in 33m 52s\n- designate-pdns4 http://logs.openstack.org/17/665317/4/check/designate-pdns4/4759e6f/ : SUCCESS in 36m 04s\n- designate-bind9-manager-model http://logs.openstack.org/17/665317/4/check/designate-bind9-manager-model/d021363/ : SUCCESS in 37m 40s\n- designate-bind9-py27 http://logs.openstack.org/17/665317/4/check/designate-bind9-py27/8da0eb1/ : SUCCESS in 36m 11s\n- designate-pdns4-py27 http://logs.openstack.org/17/665317/4/check/designate-pdns4-py27/1d76770/ : SUCCESS in 38m 00s\n- designate-pdns4-postgres http://logs.openstack.org/17/665317/4/check/designate-pdns4-postgres/1721cf1/ : SUCCESS in 38m 35s\n- designate-grenade-pdns4 http://logs.openstack.org/17/665317/4/check/designate-grenade-pdns4/2d55201/ : SUCCESS in 1h 00m 15s\n- openstack-tox-cover http://logs.openstack.org/17/665317/4/check/openstack-tox-cover/162fa7d/cover/ : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665317/4/check/openstack-tox-lower-constraints/1b04be5/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/17/665317/4/check/openstack-tox-pep8/42cb937/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/17/665317/4/check/openstack-tox-py27/e33c64c/ : SUCCESS in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/17/665317/4/check/openstack-tox-py36/1fc0bb1/ : SUCCESS in 5m 14s\n- openstack-tox-py37 http://logs.openstack.org/17/665317/4/check/openstack-tox-py37/637727b/ : SUCCESS in 8m 03s\n- openstack-tox-docs http://logs.openstack.org/17/665317/4/check/openstack-tox-docs/38a54a6/html/ : SUCCESS in 5m 50s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/17/665317/4/check/neutron-tempest-plugin-designate-scenario/ccba9e7/ : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":4},{"id":"79376a43ae8943f26b2db7b753124f948f981659","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 22:00:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d1b6e9e4455359138a46d395824cd6e1299abb9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 23:03:12.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/17/665317/4/gate/designate-bind9/232d758/ : SUCCESS in 38m 10s\n- designate-pdns4 http://logs.openstack.org/17/665317/4/gate/designate-pdns4/a0ddd09/ : SUCCESS in 36m 15s\n- designate-bind9-manager-model http://logs.openstack.org/17/665317/4/gate/designate-bind9-manager-model/dcfcf93/ : SUCCESS in 38m 58s\n- designate-bind9-py27 http://logs.openstack.org/17/665317/4/gate/designate-bind9-py27/d4a7701/ : SUCCESS in 39m 40s\n- designate-pdns4-py27 http://logs.openstack.org/17/665317/4/gate/designate-pdns4-py27/5af70f3/ : SUCCESS in 40m 08s\n- designate-pdns4-postgres http://logs.openstack.org/17/665317/4/gate/designate-pdns4-postgres/566bb27/ : SUCCESS in 37m 24s\n- designate-grenade-pdns4 http://logs.openstack.org/17/665317/4/gate/designate-grenade-pdns4/a676479/ : SUCCESS in 57m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/665317/4/gate/openstack-tox-lower-constraints/346fc44/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/17/665317/4/gate/openstack-tox-pep8/0561b21/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/17/665317/4/gate/openstack-tox-py27/38470f7/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/17/665317/4/gate/openstack-tox-py36/5d72eec/ : SUCCESS in 6m 53s\n- openstack-tox-py37 http://logs.openstack.org/17/665317/4/gate/openstack-tox-py37/9fcc518/ : SUCCESS in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/17/665317/4/gate/openstack-tox-docs/d871fba/html/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":4},{"id":"93413a899e6151f2a1637ae185912ba357f65a8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 23:03:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d787ee45c0f395d8666c978f499b5d0539b984b0","revisions":{"8264c19adfbba7999876ce5d54e6ddc2170e9027":{"kind":"REWORK","_number":1,"created":"2019-06-14 07:07:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/665317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/665317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/665317/1"}}},"commit":{"parents":[{"commit":"3294dd956f9c08f36f50dbff00105f1e8c0151e4","subject":"Merge \"Simplified metrics implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3294dd956f9c08f36f50dbff00105f1e8c0151e4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 07:05:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 07:05:11.000000000","tz":-420},"subject":"Only register dynamic pool options when running designate-manage","message":"Only register dynamic pool options when running designate-manage\n\nWe only need ot load this when convert old designate\nconfiguration (pre-mitaka) into a pools.yaml file.\n\nChange-Id: Ia9587880ca2428c0b3af074f0cb2a873859685ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8264c19adfbba7999876ce5d54e6ddc2170e9027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8264c19adfbba7999876ce5d54e6ddc2170e9027"}]},"branch":"refs/heads/master"},"3156fd454c46cf228a98d82574731c6468853337":{"kind":"REWORK","_number":2,"created":"2019-06-14 07:10:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/665317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/665317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/665317/2"}}},"commit":{"parents":[{"commit":"64c17690034b614ea7656a53558e34ead65f929d","subject":"Finalized centralization of configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64c17690034b614ea7656a53558e34ead65f929d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 07:05:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 07:10:33.000000000","tz":-420},"subject":"Only register dynamic pool options when running designate-manage","message":"Only register dynamic pool options when running designate-manage\n\nWe only need ot load this when convert old designate\nconfiguration (pre-mitaka) into a pools.yaml file.\n\nChange-Id: Ia9587880ca2428c0b3af074f0cb2a873859685ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3156fd454c46cf228a98d82574731c6468853337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3156fd454c46cf228a98d82574731c6468853337"}]},"branch":"refs/heads/master"},"c854ec83ad27e8eb7eda86c7f49a20aa1def619b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-14 07:16:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/665317/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/665317/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/665317/3"}}},"commit":{"parents":[{"commit":"64c17690034b614ea7656a53558e34ead65f929d","subject":"Finalized centralization of configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64c17690034b614ea7656a53558e34ead65f929d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 07:05:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 07:16:33.000000000","tz":0},"subject":"Only register dynamic pool options when running designate-manage","message":"Only register dynamic pool options when running designate-manage\n\nWe only need to load this code path when converting an old designate\nconfiguration (pre-mitaka) into a pools.yaml file.\n\nChange-Id: Ia9587880ca2428c0b3af074f0cb2a873859685ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c854ec83ad27e8eb7eda86c7f49a20aa1def619b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c854ec83ad27e8eb7eda86c7f49a20aa1def619b"}]},"branch":"refs/heads/master"},"d787ee45c0f395d8666c978f499b5d0539b984b0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-18 19:51:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/665317/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/665317/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/665317/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/665317/4"}}},"commit":{"parents":[{"commit":"097ffec548cea59f725920593f07f205c938a8e7","subject":"Finalized centralization of configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/097ffec548cea59f725920593f07f205c938a8e7"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 07:05:11.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 19:51:24.000000000","tz":0},"subject":"Only register dynamic pool options when running designate-manage","message":"Only register dynamic pool options when running designate-manage\n\nWe only need to load this code path when converting an old designate\nconfiguration (pre-mitaka) into a pools.yaml file.\n\nChange-Id: Ia9587880ca2428c0b3af074f0cb2a873859685ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d787ee45c0f395d8666c978f499b5d0539b984b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d787ee45c0f395d8666c978f499b5d0539b984b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
