)]}'
{"id":"openstack%2Fdesignate~665476","triplet_id":"openstack%2Fdesignate~master~Ib258043d492bb25e1c061d765b40885317c73804","project":"openstack/designate","branch":"master","topic":"remove_pool_manager","hashtags":[],"change_id":"Ib258043d492bb25e1c061d765b40885317c73804","subject":"Removed deprecated pool-manager implementation","status":"MERGED","created":"2019-06-14 21:29:12.000000000","updated":"2019-09-25 18:33:52.000000000","submitted":"2019-09-25 18:29:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665476-1569436187381-b42f53c0","meta_rev_id":"42afa63771d6d984db8ad1a91ee68ec9d9d182b9","_number":665476,"virtual_id_number":665476,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 18:29:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-09-24 11:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-09-24 11:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-09-25 17:02:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-09-24 11:10:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 11:16:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-09-25 17:02:35.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-09-25 18:29:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7592adad1e576d08ba55732f03634d11e9bd26be","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 21:29:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5ca4d9e4472aeff741faa82696355e6533c91ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 22:31:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/76/665476/1/check/test-release-openstack/c8c435c/ : SUCCESS in 3m 44s\n- designate-bind9 http://logs.openstack.org/76/665476/1/check/designate-bind9/7dc345d/ : FAILURE in 29m 17s\n- designate-pdns4 http://logs.openstack.org/76/665476/1/check/designate-pdns4/71802a1/ : FAILURE in 36m 42s\n- designate-bind9-manager-model http://logs.openstack.org/76/665476/1/check/designate-bind9-manager-model/f536ade/ : FAILURE in 26m 04s\n- designate-bind9-py27 http://logs.openstack.org/76/665476/1/check/designate-bind9-py27/9cfcc3d/ : FAILURE in 31m 19s\n- designate-pdns4-py27 http://logs.openstack.org/76/665476/1/check/designate-pdns4-py27/ddf24c5/ : FAILURE in 29m 26s\n- designate-pdns4-postgres http://logs.openstack.org/76/665476/1/check/designate-pdns4-postgres/6ec9e08/ : FAILURE in 35m 30s\n- designate-grenade-pdns4 http://logs.openstack.org/76/665476/1/check/designate-grenade-pdns4/7d9ccd7/ : FAILURE in 49m 34s\n- openstack-tox-cover http://logs.openstack.org/76/665476/1/check/openstack-tox-cover/5d01f13/cover/ : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/665476/1/check/openstack-tox-lower-constraints/9195c1c/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/76/665476/1/check/openstack-tox-pep8/ca33e20/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/76/665476/1/check/openstack-tox-py27/07e066d/ : SUCCESS in 8m 33s\n- openstack-tox-py36 http://logs.openstack.org/76/665476/1/check/openstack-tox-py36/7ea3786/ : SUCCESS in 5m 46s\n- openstack-tox-py37 http://logs.openstack.org/76/665476/1/check/openstack-tox-py37/4b6463d/ : SUCCESS in 7m 39s\n- openstack-tox-docs http://logs.openstack.org/76/665476/1/check/openstack-tox-docs/7c415b3/html/ : SUCCESS in 5m 00s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/665476/1/check/neutron-tempest-plugin-designate-scenario/b5a33fc/ : FAILURE in 29m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5bf24134cb52d50511bd00c5ec2f6ed7e7de3c54","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 22:44:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a3b90a09cbe858680f052eaa0682428dfd85c877","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-14 23:48:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b6a547cc65cff6955ce74071e03c641abeea218e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 01:36:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/76/665476/3/check/test-release-openstack/19bccb9/ : SUCCESS in 3m 47s\n- designate-bind9 http://logs.openstack.org/76/665476/3/check/designate-bind9/2fc3cc7/ : SUCCESS in 42m 13s\n- designate-pdns4 http://logs.openstack.org/76/665476/3/check/designate-pdns4/ab77b04/ : SUCCESS in 39m 05s\n- designate-bind9-manager-model http://logs.openstack.org/76/665476/3/check/designate-bind9-manager-model/6aa915a/ : TIMED_OUT in 1h 13m 44s\n- designate-bind9-py27 http://logs.openstack.org/76/665476/3/check/designate-bind9-py27/0853812/ : SUCCESS in 46m 39s\n- designate-pdns4-py27 http://logs.openstack.org/76/665476/3/check/designate-pdns4-py27/1bd51e2/ : SUCCESS in 38m 32s\n- designate-pdns4-postgres http://logs.openstack.org/76/665476/3/check/designate-pdns4-postgres/2cde331/ : SUCCESS in 39m 51s\n- designate-grenade-pdns4 http://logs.openstack.org/76/665476/3/check/designate-grenade-pdns4/10198b6/ : SUCCESS in 1h 34m 16s\n- openstack-tox-cover http://logs.openstack.org/76/665476/3/check/openstack-tox-cover/f3633d9/cover/ : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/665476/3/check/openstack-tox-lower-constraints/d381635/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/76/665476/3/check/openstack-tox-pep8/decc2a2/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/76/665476/3/check/openstack-tox-py27/8d44596/ : SUCCESS in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/76/665476/3/check/openstack-tox-py36/fe29865/ : SUCCESS in 7m 59s\n- openstack-tox-py37 http://logs.openstack.org/76/665476/3/check/openstack-tox-py37/9572212/ : SUCCESS in 8m 48s\n- openstack-tox-docs http://logs.openstack.org/76/665476/3/check/openstack-tox-docs/bdfa83c/html/ : SUCCESS in 5m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/665476/3/check/neutron-tempest-plugin-designate-scenario/f3904c3/ : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":3},{"id":"5f2a75f4cce2b4f364c0279066a1c97e72f0d28b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-15 04:40:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9d18300ccca3b041a95807e7f624bb81879eb9d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 06:02:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/76/665476/4/check/test-release-openstack/9c3c853/ : SUCCESS in 3m 51s\n- designate-bind9 http://logs.openstack.org/76/665476/4/check/designate-bind9/9fc1ebe/ : SUCCESS in 40m 28s\n- designate-pdns4 http://logs.openstack.org/76/665476/4/check/designate-pdns4/1715afe/ : SUCCESS in 40m 50s\n- designate-bind9-py27 http://logs.openstack.org/76/665476/4/check/designate-bind9-py27/0402ae6/ : SUCCESS in 41m 17s\n- designate-pdns4-py27 http://logs.openstack.org/76/665476/4/check/designate-pdns4-py27/6d78d94/ : SUCCESS in 34m 18s\n- designate-pdns4-postgres http://logs.openstack.org/76/665476/4/check/designate-pdns4-postgres/035c0d4/ : SUCCESS in 52m 04s\n- designate-grenade-pdns4 http://logs.openstack.org/76/665476/4/check/designate-grenade-pdns4/9664e6e/ : SUCCESS in 1h 00m 40s\n- openstack-tox-cover http://logs.openstack.org/76/665476/4/check/openstack-tox-cover/b4ba239/cover/ : SUCCESS in 6m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/665476/4/check/openstack-tox-lower-constraints/c710a13/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/76/665476/4/check/openstack-tox-pep8/a51cecd/ : SUCCESS in 6m 51s\n- openstack-tox-py27 http://logs.openstack.org/76/665476/4/check/openstack-tox-py27/b0e9cc1/ : SUCCESS in 6m 08s\n- openstack-tox-py36 http://logs.openstack.org/76/665476/4/check/openstack-tox-py36/698996d/ : SUCCESS in 5m 55s\n- openstack-tox-py37 http://logs.openstack.org/76/665476/4/check/openstack-tox-py37/cac59f2/ : SUCCESS in 6m 13s\n- openstack-tox-docs http://logs.openstack.org/76/665476/4/check/openstack-tox-docs/a924994/html/ : SUCCESS in 6m 24s\n- build-openstack-releasenotes http://logs.openstack.org/76/665476/4/check/build-openstack-releasenotes/d1cd0db/html/ : SUCCESS in 6m 35s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/665476/4/check/neutron-tempest-plugin-designate-scenario/a50b518/ : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":4},{"id":"9a38fcf1a238cf76ffa4b9825f5efd4f5d6992c4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-06-17 10:19:52.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis seems to have some merge conflict, but I agree that now is a good time to get rid of this.","accounts_in_message":[],"_revision_number":4},{"id":"a500927a2d2412ffbe5a16057ff643f28acd0b44","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 17:04:34.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n \u003e \n \u003e This seems to have some merge conflict, but I agree that now is a\n \u003e good time to get rid of this.\n\nYea - would probably be worth adding more documentation. Was thinking that we could follow up with a guide to this patch. \n\nI actually kinda liked it at the time, as it really highlights how long we have been wanting to remove this code.","accounts_in_message":[],"_revision_number":4},{"id":"79a5531650f73201a28c2a83fe2509cee1c90832","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 17:06:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"029d3a24dd2dc6639c0894ae63af62f10d183d0f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 17:18:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e19afc187d6d51a8a92b643b61f2d844ffbb47cc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 17:25:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"75daa4893e1612221d21b77163e96a2565a170b6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 18:04:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"20ecf9b0cb2d9f69340755818e9e05ead6c12a9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:34:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/76/665476/7/check/test-release-openstack/49aa322/ : SUCCESS in 3m 55s\n- designate-bind9 http://logs.openstack.org/76/665476/7/check/designate-bind9/526e7aa/ : SUCCESS in 41m 00s\n- designate-pdns4 http://logs.openstack.org/76/665476/7/check/designate-pdns4/3bd6121/ : SUCCESS in 41m 19s\n- designate-bind9-py27 http://logs.openstack.org/76/665476/7/check/designate-bind9-py27/eb6dd33/ : SUCCESS in 45m 14s\n- designate-pdns4-py27 http://logs.openstack.org/76/665476/7/check/designate-pdns4-py27/ee2f7f4/ : SUCCESS in 35m 43s\n- designate-pdns4-postgres http://logs.openstack.org/76/665476/7/check/designate-pdns4-postgres/d73da25/ : SUCCESS in 40m 36s\n- designate-grenade-pdns4 http://logs.openstack.org/76/665476/7/check/designate-grenade-pdns4/1470d0a/ : SUCCESS in 1h 01m 02s\n- openstack-tox-cover http://logs.openstack.org/76/665476/7/check/openstack-tox-cover/7457ada/cover/ : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/665476/7/check/openstack-tox-lower-constraints/e913468/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/76/665476/7/check/openstack-tox-pep8/1796978/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/76/665476/7/check/openstack-tox-py27/280a6b2/ : SUCCESS in 7m 03s\n- openstack-tox-py36 http://logs.openstack.org/76/665476/7/check/openstack-tox-py36/2b13127/ : SUCCESS in 6m 18s\n- openstack-tox-py37 http://logs.openstack.org/76/665476/7/check/openstack-tox-py37/7b36748/ : SUCCESS in 10m 13s\n- openstack-tox-docs http://logs.openstack.org/76/665476/7/check/openstack-tox-docs/4b9ccd8/html/ : SUCCESS in 5m 28s\n- build-openstack-releasenotes http://logs.openstack.org/76/665476/7/check/build-openstack-releasenotes/5a8e411/html/ : SUCCESS in 8m 00s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/665476/7/check/neutron-tempest-plugin-designate-scenario/67a5e54/ : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":7},{"id":"84e3d7549f20cfc64552c67f0349e30a85268c29","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 19:32:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c6d2fa7317e7f3e9fe55fe2d5e5ef61a17d08013","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 19:34:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7a02086c84c092eb2ffc43ffb555ad1509bc8794","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 19:34:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cb3875fd4f5f8384d7846fcee0a52aee83d24895","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 20:38:26.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/76/665476/8/check/test-release-openstack/6a614a9/ : SUCCESS in 6m 20s\n- designate-bind9 http://logs.openstack.org/76/665476/8/check/designate-bind9/f15b5b6/ : SUCCESS in 48m 10s\n- designate-pdns4 http://logs.openstack.org/76/665476/8/check/designate-pdns4/7e6e694/ : SUCCESS in 36m 55s\n- designate-bind9-py27 http://logs.openstack.org/76/665476/8/check/designate-bind9-py27/5edd6b7/ : SUCCESS in 44m 14s\n- designate-pdns4-py27 http://logs.openstack.org/76/665476/8/check/designate-pdns4-py27/3e32e7a/ : SUCCESS in 41m 08s\n- designate-pdns4-postgres http://logs.openstack.org/76/665476/8/check/designate-pdns4-postgres/a415858/ : SUCCESS in 39m 38s\n- designate-grenade-pdns4 http://logs.openstack.org/76/665476/8/check/designate-grenade-pdns4/4c1e296/ : SUCCESS in 58m 40s\n- openstack-tox-cover http://logs.openstack.org/76/665476/8/check/openstack-tox-cover/f7736ff/cover/ : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/665476/8/check/openstack-tox-lower-constraints/8d3316d/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/76/665476/8/check/openstack-tox-pep8/1315794/ : FAILURE in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/76/665476/8/check/openstack-tox-py27/f705376/ : SUCCESS in 6m 12s\n- openstack-tox-py36 http://logs.openstack.org/76/665476/8/check/openstack-tox-py36/71a3153/ : SUCCESS in 5m 47s\n- openstack-tox-py37 http://logs.openstack.org/76/665476/8/check/openstack-tox-py37/0898fc6/ : SUCCESS in 6m 37s\n- openstack-tox-docs http://logs.openstack.org/76/665476/8/check/openstack-tox-docs/76a55ab/html/ : SUCCESS in 5m 11s\n- build-openstack-releasenotes http://logs.openstack.org/76/665476/8/check/build-openstack-releasenotes/4cee37a/html/ : SUCCESS in 6m 49s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/665476/8/check/neutron-tempest-plugin-designate-scenario/5b6f835/ : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":8},{"id":"34f4d46c69302d4f2c1bccbfb75dfb390988c0ff","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-17 20:48:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f5fd4f7eec5b3d76a24c783c7a81cb18d1e79270","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 22:05:20.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/76/665476/9/check/test-release-openstack/38a470b/ : SUCCESS in 5m 17s\n- designate-bind9 http://logs.openstack.org/76/665476/9/check/designate-bind9/d3f5441/ : SUCCESS in 41m 50s\n- designate-pdns4 http://logs.openstack.org/76/665476/9/check/designate-pdns4/b8e5273/ : SUCCESS in 42m 22s\n- designate-bind9-py27 http://logs.openstack.org/76/665476/9/check/designate-bind9-py27/2d50321/ : SUCCESS in 48m 45s\n- designate-pdns4-py27 http://logs.openstack.org/76/665476/9/check/designate-pdns4-py27/fb68984/ : SUCCESS in 37m 59s\n- designate-pdns4-postgres http://logs.openstack.org/76/665476/9/check/designate-pdns4-postgres/be6d721/ : SUCCESS in 40m 21s\n- designate-grenade-pdns4 http://logs.openstack.org/76/665476/9/check/designate-grenade-pdns4/22a03c0/ : SUCCESS in 55m 24s\n- openstack-tox-cover http://logs.openstack.org/76/665476/9/check/openstack-tox-cover/61e6726/cover/ : SUCCESS in 7m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/665476/9/check/openstack-tox-lower-constraints/360244b/ : SUCCESS in 6m 34s\n- openstack-tox-pep8 http://logs.openstack.org/76/665476/9/check/openstack-tox-pep8/7f531f8/ : SUCCESS in 8m 29s\n- openstack-tox-py27 http://logs.openstack.org/76/665476/9/check/openstack-tox-py27/d2e98d1/ : SUCCESS in 6m 09s\n- openstack-tox-py36 http://logs.openstack.org/76/665476/9/check/openstack-tox-py36/7f64d4b/ : SUCCESS in 7m 20s\n- openstack-tox-py37 http://logs.openstack.org/76/665476/9/check/openstack-tox-py37/1354170/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/76/665476/9/check/openstack-tox-docs/8f784cf/html/ : SUCCESS in 7m 30s\n- build-openstack-releasenotes http://logs.openstack.org/76/665476/9/check/build-openstack-releasenotes/67092e4/html/ : SUCCESS in 8m 46s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/665476/9/check/neutron-tempest-plugin-designate-scenario/8e31c3a/ : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":9},{"id":"46cd2a09070fa4259eb9ffb40b20590bea6c1c14","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 19:30:03.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"322e0336627e77ad9a4824ddf5679701895115fd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 20:27:59.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"93365948ea034bfe0b3136c3a46a33e57dbd38b3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 20:32:03.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"04406c3640aaa26a7645672dc43385b11390e39f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 20:34:07.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"9655ea2f620f96de6efa68ddf06c30f6ad2abe39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:51:50.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/76/665476/13/check/test-release-openstack/e50d020/ : SUCCESS in 3m 23s\n- designate-bind9 http://logs.openstack.org/76/665476/13/check/designate-bind9/a948a37/ : SUCCESS in 41m 57s\n- designate-pdns4 http://logs.openstack.org/76/665476/13/check/designate-pdns4/3255b2e/ : SUCCESS in 39m 57s\n- designate-bind9-py27 http://logs.openstack.org/76/665476/13/check/designate-bind9-py27/4b464d9/ : SUCCESS in 42m 13s\n- designate-pdns4-py27 http://logs.openstack.org/76/665476/13/check/designate-pdns4-py27/c165a6d/ : SUCCESS in 44m 59s\n- designate-pdns4-postgres http://logs.openstack.org/76/665476/13/check/designate-pdns4-postgres/2bba223/ : SUCCESS in 42m 14s\n- designate-grenade-pdns4 http://logs.openstack.org/76/665476/13/check/designate-grenade-pdns4/e6a952e/ : SUCCESS in 55m 22s\n- openstack-tox-cover http://logs.openstack.org/76/665476/13/check/openstack-tox-cover/46c0ad9/cover/ : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/665476/13/check/openstack-tox-lower-constraints/0fc2367/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/76/665476/13/check/openstack-tox-pep8/55b6fc0/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/76/665476/13/check/openstack-tox-py27/a42754a/ : SUCCESS in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/76/665476/13/check/openstack-tox-py36/29a3692/ : SUCCESS in 5m 43s\n- openstack-tox-py37 http://logs.openstack.org/76/665476/13/check/openstack-tox-py37/9f59ce5/ : SUCCESS in 6m 36s\n- openstack-tox-docs http://logs.openstack.org/76/665476/13/check/openstack-tox-docs/fc590e8/html/ : SUCCESS in 4m 54s\n- build-openstack-releasenotes http://logs.openstack.org/76/665476/13/check/build-openstack-releasenotes/35e9bc1/html/ : SUCCESS in 7m 04s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/665476/13/check/neutron-tempest-plugin-designate-scenario/5049b05/ : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":13},{"id":"df98d4da6f0248b75e6b2c597ed596fbfa06eb67","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-15 06:24:11.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ae5d265f34d386bbd5eecb41da516ce5eda428d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 07:29:05.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/76/665476/14/check/test-release-openstack/87a2616/ : SUCCESS in 2m 00s\n- designate-bind9 http://logs.openstack.org/76/665476/14/check/designate-bind9/b0a47d0/ : SUCCESS in 37m 55s\n- designate-pdns4 http://logs.openstack.org/76/665476/14/check/designate-pdns4/e4df49c/ : SUCCESS in 42m 35s\n- designate-bind9-py27 http://logs.openstack.org/76/665476/14/check/designate-bind9-py27/924be27/ : SUCCESS in 39m 43s\n- designate-pdns4-py27 http://logs.openstack.org/76/665476/14/check/designate-pdns4-py27/cad62c2/ : SUCCESS in 38m 49s\n- designate-pdns4-postgres http://logs.openstack.org/76/665476/14/check/designate-pdns4-postgres/2b5553a/ : FAILURE in 30m 48s (non-voting)\n- designate-grenade-pdns4 http://logs.openstack.org/76/665476/14/check/designate-grenade-pdns4/646b068/ : SUCCESS in 58m 28s\n- openstack-tox-cover http://logs.openstack.org/76/665476/14/check/openstack-tox-cover/293f9c2/cover/ : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/665476/14/check/openstack-tox-lower-constraints/c4a0c87/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/76/665476/14/check/openstack-tox-pep8/ad4d53a/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/76/665476/14/check/openstack-tox-py27/5f74fe4/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/76/665476/14/check/openstack-tox-py36/43d3cdf/ : SUCCESS in 3m 39s\n- openstack-tox-py37 http://logs.openstack.org/76/665476/14/check/openstack-tox-py37/9aef688/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/76/665476/14/check/openstack-tox-docs/67398e6/html/ : SUCCESS in 3m 13s\n- build-openstack-releasenotes http://logs.openstack.org/76/665476/14/check/build-openstack-releasenotes/136a5d7/html/ : SUCCESS in 5m 28s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/665476/14/check/neutron-tempest-plugin-designate-scenario/151dd36/ : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":14},{"id":"b359b93bf869d41708dda23cfabc0f8c9ee1ffd8","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-07-23 10:04:27.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"932d191f4bf448cceb35e688bd698c9513d45b40","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-24 18:20:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"cc947f4e342db1b59eab4c0436747076145e5794","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-29 15:35:45.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"cac1957eba48f94da36be878f3a144238c206233","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 16:42:14.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/76/665476/15/check/test-release-openstack/5e3a081/ : SUCCESS in 4m 16s\n- designate-bind9 http://logs.openstack.org/76/665476/15/check/designate-bind9/705e91d/ : SUCCESS in 50m 19s\n- designate-pdns4 http://logs.openstack.org/76/665476/15/check/designate-pdns4/9cac805/ : SUCCESS in 43m 29s\n- designate-bind9-py27 http://logs.openstack.org/76/665476/15/check/designate-bind9-py27/81f5e68/ : SUCCESS in 52m 08s\n- designate-pdns4-py27 http://logs.openstack.org/76/665476/15/check/designate-pdns4-py27/152283d/ : SUCCESS in 35m 58s\n- designate-pdns4-postgres http://logs.openstack.org/76/665476/15/check/designate-pdns4-postgres/4b4277c/ : FAILURE in 34m 09s (non-voting)\n- designate-grenade-pdns4 http://logs.openstack.org/76/665476/15/check/designate-grenade-pdns4/99088ca/ : SUCCESS in 58m 17s\n- openstack-tox-cover http://logs.openstack.org/76/665476/15/check/openstack-tox-cover/26e87d3/cover/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/665476/15/check/openstack-tox-lower-constraints/2bb37f1/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/76/665476/15/check/openstack-tox-pep8/c72fe0f/ : SUCCESS in 7m 27s\n- openstack-tox-py27 http://logs.openstack.org/76/665476/15/check/openstack-tox-py27/04146e7/ : SUCCESS in 6m 18s\n- openstack-tox-py36 http://logs.openstack.org/76/665476/15/check/openstack-tox-py36/01abc40/ : SUCCESS in 5m 09s\n- openstack-tox-py37 http://logs.openstack.org/76/665476/15/check/openstack-tox-py37/cc2f3a2/ : SUCCESS in 8m 09s\n- openstack-tox-docs http://logs.openstack.org/76/665476/15/check/openstack-tox-docs/d9dfabc/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/76/665476/15/check/build-openstack-releasenotes/5242e10/html/ : SUCCESS in 6m 15s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/665476/15/check/neutron-tempest-plugin-designate-scenario/eb6fdfd/ : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":15},{"id":"26a82a93c47bf8b04234774edee0fe7530699c16","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-29 21:05:11.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"291648651ca947448577fcf4ec23acbb8123e7bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:10:54.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/76/665476/16/check/test-release-openstack/972b113/ : SUCCESS in 3m 16s\n- designate-bind9 http://logs.openstack.org/76/665476/16/check/designate-bind9/4a4adb2/ : SUCCESS in 43m 15s\n- designate-pdns4 http://logs.openstack.org/76/665476/16/check/designate-pdns4/79aa8f1/ : SUCCESS in 47m 43s\n- designate-bind9-py27 http://logs.openstack.org/76/665476/16/check/designate-bind9-py27/3ce9213/ : SUCCESS in 47m 21s\n- designate-pdns4-py27 http://logs.openstack.org/76/665476/16/check/designate-pdns4-py27/0731f28/ : SUCCESS in 42m 37s\n- designate-pdns4-postgres http://logs.openstack.org/76/665476/16/check/designate-pdns4-postgres/3009a6e/ : FAILURE in 36m 13s (non-voting)\n- designate-grenade-pdns4 http://logs.openstack.org/76/665476/16/check/designate-grenade-pdns4/d63e7d3/ : SUCCESS in 58m 09s\n- openstack-tox-cover http://logs.openstack.org/76/665476/16/check/openstack-tox-cover/d51c2bf/cover/ : SUCCESS in 7m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/76/665476/16/check/openstack-tox-lower-constraints/18cd2fc/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/76/665476/16/check/openstack-tox-pep8/5bb6a78/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/76/665476/16/check/openstack-tox-py27/07cee18/ : SUCCESS in 5m 01s\n- openstack-tox-py36 http://logs.openstack.org/76/665476/16/check/openstack-tox-py36/f52de35/ : SUCCESS in 5m 43s\n- openstack-tox-py37 http://logs.openstack.org/76/665476/16/check/openstack-tox-py37/b0e491b/ : SUCCESS in 7m 12s\n- openstack-tox-docs http://logs.openstack.org/76/665476/16/check/openstack-tox-docs/6fd5294/html/ : SUCCESS in 4m 00s\n- build-openstack-releasenotes http://logs.openstack.org/76/665476/16/check/build-openstack-releasenotes/40f7add/html/ : SUCCESS in 5m 44s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/76/665476/16/check/neutron-tempest-plugin-designate-scenario/cfc85a9/ : SUCCESS in 40m 20s","accounts_in_message":[],"_revision_number":16},{"id":"fbb44f5626bf911c0bddfd279839a173b7b320b7","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-06 16:34:53.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"9dc25f182a67d0842545193f646023f1b53fc8b3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-08-09 11:46:30.000000000","message":"Patch Set 16:\n\n(5 comments)","accounts_in_message":[],"_revision_number":16},{"id":"c4011794ce4b8375f1ca131563706da2b08fdd1d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-08-12 08:48:51.000000000","message":"Patch Set 16: Code-Review-1\n\nWhat about docs that mention the pool-manager like http://logs.openstack.org/76/665476/16/check/openstack-tox-docs/6fd5294/html/admin/backends/agent.html ?","accounts_in_message":[],"_revision_number":16},{"id":"e12f6c9c4b65a7364700a8cb0a50f1e177919dac","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 02:45:07.000000000","message":"Patch Set 16:\n\nGonna hold off with this until after Train is released. As I don\u0027t think it will have enough time to bake before the next release.","accounts_in_message":[],"_revision_number":16},{"id":"cd15bbeb2f5892b506ffc24ed3db48da86e51ff3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-18 03:00:21.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"d86957ac3e31b3d1398c79e7b46a43a79039996c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-18 03:02:14.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"2d525b78ab733a7887e46d0e159d3fec13ad5baf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-18 03:03:19.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"eed769deddebb4f8b36aac693617d731ea0b576d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-18 03:22:24.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"393c596909c33d5cc4f234faa3dbf9ecc1f1abb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 04:33:20.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4a0edc41025d488889495bc228e93a67 : SUCCESS in 2m 13s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/42ec9584598c4221a11d04a242eb8cf4 : SUCCESS in 38m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a25bf9b794844044a7992810f4d16165 : SUCCESS in 39m 41s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/432285b78897470ea6e21e09ac7d34c9 : SUCCESS in 40m 48s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/37d915f27a8f458f83b5e74ddf4f0366 : SUCCESS in 44m 40s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/cd621085fce04326b01ad4cadb7de240 : FAILURE in 28m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b45570ffdb4342e8b9bcc4e2c21a48c1 : SUCCESS in 1h 04m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a504510348b54717a43af49c96c90e7d : SUCCESS in 41m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2ccd1ed7f13b4ee69e46daeaa6dff7f1 : SUCCESS in 38m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/928639f251c14a918e0ffdcb05bbcb6b : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b89a543b31b4f7fa93f56d96a26bc6e : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab3644b4b1ea44f5a66ced13da53291e : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e91429ce4b78460291e3692d415e9d98 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd4216553e404cdc82b51f80273f6992 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf15ec2f78ee41509de04ba9d5253152 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df08a0762e7443e0adcaa796e77a89b8 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89595060e11c493aa034468df6e9a2de : SUCCESS in 5m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8b1107d87624691bd328439694b0af1 : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":19},{"id":"5f54277f94f7242fde59c89ae74fb5e8d8a5f5dc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-19 01:32:53.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"7bbf57bb2e32049765a87924ea027ea170cd3463","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-19 01:40:56.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"5ea3b9f2d201eb29973ff68c4b19d0a8cbf00d2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 02:47:41.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2d99fa6c51984435b22a9a4699ae13ef : SUCCESS in 1m 58s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bab2669aa0b447a3ac9442d171c7a245 : SUCCESS in 40m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e89e46d28ca1488488289290767ba2e0 : SUCCESS in 50m 45s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/ef1964e4e17f4dd9aea25feeda142497 : SUCCESS in 47m 20s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/3d5b04c42f6349ad8191b5857f5fd03f : SUCCESS in 38m 55s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/bf7805b34a604976a4d2e9a7321d1337 : FAILURE in 28m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ae6136417cf94c7ba0b548f41605da57 : SUCCESS in 58m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/79d46a3f68164d9ca9cf5c1ffafa8380 : SUCCESS in 37m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/803d6c1831604ec9bacc6b8f96c5ffeb : SUCCESS in 42m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fc334cbddb44972b92d1d13237d8e84 : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f99a9420fd8a440d96a1df72036175a0 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b982ea46e4334f86b788a397cdc57941 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/37eb54522eee4ba4a2f96d96d6aee7dd : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/308c648e09bf465fb7a9cbab4e17571f : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2131304ab4604cacac797fa2aa2fa11a : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3732ff395ca64f6f8a8f646e10d217f4 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53c0529925e24efc9a69b223f2780e35 : SUCCESS in 5m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02b7ab03dcaf4393875b112b112c78c8 : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":21},{"id":"1eeb447ec5a90f75b38e74829164357d44bde9fb","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-09-24 10:32:05.000000000","message":"Patch Set 21: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":21},{"id":"d6ab56127ebbebe4cb67aec6b1ae61322fab471b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-24 11:10:54.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"cfec3febde647e0b8f5a8cc4d891461816a08f7c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-09-24 11:16:59.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"b842c69af5a5b26ff1d04349efba6037e02b056a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 12:14:51.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7cdff9d1124e48dda1852877aefb2970 : SUCCESS in 1m 39s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/222126c464804be095394365c16ff7bd : SUCCESS in 47m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2776fcf4f66a469285cbf568939e3e98 : SUCCESS in 36m 44s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/e02180cb12ee4756847a352e584a7f77 : SUCCESS in 41m 31s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/9b1f8b0961f3478bb18cf848773576e6 : SUCCESS in 42m 41s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/add7ea1b703f461ca223ed31600a6f2e : FAILURE in 33m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/668a9d0f914e401087e9c6884953f691 : SUCCESS in 56m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/db32332fd0b24b47aef518ef9950d139 : SUCCESS in 41m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0a8673c1520a49e98c9ec02035d175d8 : SUCCESS in 46m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38b938f7285d4a3eb14ccacd14da9055 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f9c30196ed94acfa329fdff95f0e325 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b307fab70f242e89e8e80f6e65ad008 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad74c1680e734d558d517471ae837229 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d0b4d0e441a4b758b1faf4437226447 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e752c9d4dfc14f2a89c154c394ba9aad : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae1a039b6a448689e97c5bf08c8c144 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71c1db4859ef4324964659094edd5b59 : SUCCESS in 5m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5cc3f5a3a09e45d9941c7a1eb31a3cc6 : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":22},{"id":"ca8f19bc5760e21bdc356e3317903a3cc769e0a4","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-09-25 17:02:35.000000000","message":"Patch Set 22: Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"1c9f0d03f9b2d7da0d370b6cc293eccf97c7470e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 17:04:16.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"2cf08c347ad624cc3e7cdad40ed60bedf61b14a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:29:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22},{"id":"790660551f1708d33bcfaed25f9fa16d2a35cff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:29:47.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cbe4106853674d74b27a584a899739de : SUCCESS in 2m 04s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9178995e8feb418082bb0b89f7bb9499 : SUCCESS in 41m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/898ff4d3f9a346d7a1d557c3871b62d2 : SUCCESS in 51m 28s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/daa3bcefea444c7580ecf72a4cdd1c28 : SUCCESS in 42m 59s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/82b123e98a4841118e5db43032cfcacd : SUCCESS in 54m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3aef1b85e6d1489388bfad406f7ee81f : SUCCESS in 1h 05m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2216a4616c2a46a4a7337bc7a30fd69a : SUCCESS in 41m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b6e46529221d47bfbf10c4912ed70e22 : SUCCESS in 48m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0bed4c9db944040a839d6cdd027a305 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41e96a07a7b4f8b9461e14764d1ff53 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8263611e3ee14e9fab05a736fb24a92e : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/543d762262a0427db4fcc525151c6cd1 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d30efc10b29c480a8fc2b1e852f05d72 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94391f5527444a91ad0cb28c7086cb83 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05714638062640cbb5c1b61904c56f61 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":22},{"id":"b64c3b2a07d7c0579d88b786bf897366ec82ab7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:33:52.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/493dd49df12b4982acc31d855c217c87 : SUCCESS in 2m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8b42dc2c9454aecb3ed902381085bfe : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"52ad1c4c13d1fead646435634e4b07bb32c68d28","revisions":{"1fed66bc522f20fff1447acfce49a3895e53077a":{"kind":"REWORK","_number":1,"created":"2019-06-14 21:29:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/1"}}},"commit":{"parents":[{"commit":"3294dd956f9c08f36f50dbff00105f1e8c0151e4","subject":"Merge \"Simplified metrics implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3294dd956f9c08f36f50dbff00105f1e8c0151e4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"subject":"[WIP] Remove Pool-Manager","message":"[WIP] Remove Pool-Manager\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fed66bc522f20fff1447acfce49a3895e53077a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fed66bc522f20fff1447acfce49a3895e53077a"}]},"branch":"refs/heads/master"},"780d6029b6d107821d35d35dbfeef9a49320d5fc":{"kind":"REWORK","_number":2,"created":"2019-06-14 22:44:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/2"}}},"commit":{"parents":[{"commit":"3294dd956f9c08f36f50dbff00105f1e8c0151e4","subject":"Merge \"Simplified metrics implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3294dd956f9c08f36f50dbff00105f1e8c0151e4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 22:44:32.000000000","tz":-420},"subject":"[WIP] Remove Pool-Manager","message":"[WIP] Remove Pool-Manager\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/780d6029b6d107821d35d35dbfeef9a49320d5fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/780d6029b6d107821d35d35dbfeef9a49320d5fc"}]},"branch":"refs/heads/master"},"0993a90ddfa3dee994418ca20922d65f84cb6c34":{"kind":"REWORK","_number":3,"created":"2019-06-14 23:48:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/3"}}},"commit":{"parents":[{"commit":"3294dd956f9c08f36f50dbff00105f1e8c0151e4","subject":"Merge \"Simplified metrics implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3294dd956f9c08f36f50dbff00105f1e8c0151e4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 23:47:58.000000000","tz":-420},"subject":"[WIP] Remove Pool-Manager","message":"[WIP] Remove Pool-Manager\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0993a90ddfa3dee994418ca20922d65f84cb6c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0993a90ddfa3dee994418ca20922d65f84cb6c34"}]},"branch":"refs/heads/master"},"8953f5400dbcec1bc9a01d1ca37ade6a3965269a":{"kind":"REWORK","_number":4,"created":"2019-06-15 04:40:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/4"}}},"commit":{"parents":[{"commit":"3294dd956f9c08f36f50dbff00105f1e8c0151e4","subject":"Merge \"Simplified metrics implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3294dd956f9c08f36f50dbff00105f1e8c0151e4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-15 04:40:14.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8953f5400dbcec1bc9a01d1ca37ade6a3965269a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8953f5400dbcec1bc9a01d1ca37ade6a3965269a"}]},"branch":"refs/heads/master"},"b94bacea33cad709db3b6fead95e3aec08691c3d":{"kind":"REWORK","_number":5,"created":"2019-06-17 17:06:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/5"}}},"commit":{"parents":[{"commit":"ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6","subject":"Merge \"Fixed renaming topic breaking service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-17 17:06:11.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b94bacea33cad709db3b6fead95e3aec08691c3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b94bacea33cad709db3b6fead95e3aec08691c3d"}]},"branch":"refs/heads/master"},"f55493adbefb94c2bf44c4c1b27fd9d8c473e0fd":{"kind":"REWORK","_number":6,"created":"2019-06-17 17:18:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/6"}}},"commit":{"parents":[{"commit":"ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6","subject":"Merge \"Fixed renaming topic breaking service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-17 17:17:50.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f55493adbefb94c2bf44c4c1b27fd9d8c473e0fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f55493adbefb94c2bf44c4c1b27fd9d8c473e0fd"}]},"branch":"refs/heads/master"},"c487c9c80b0e1a9a80e7c7534121b51d2d8ecf00":{"kind":"REWORK","_number":7,"created":"2019-06-17 17:25:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/7"}}},"commit":{"parents":[{"commit":"ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6","subject":"Merge \"Fixed renaming topic breaking service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-17 17:25:39.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c487c9c80b0e1a9a80e7c7534121b51d2d8ecf00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c487c9c80b0e1a9a80e7c7534121b51d2d8ecf00"}]},"branch":"refs/heads/master"},"5d9f46882367643d85fa198481258933dc31677d":{"kind":"REWORK","_number":8,"created":"2019-06-17 19:32:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/8"}}},"commit":{"parents":[{"commit":"ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6","subject":"Merge \"Fixed renaming topic breaking service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-17 19:32:07.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d9f46882367643d85fa198481258933dc31677d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d9f46882367643d85fa198481258933dc31677d"}]},"branch":"refs/heads/master"},"7963bbebaacb5b26da88ff96c42ea182eb398809":{"kind":"REWORK","_number":9,"created":"2019-06-17 20:48:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/9"}}},"commit":{"parents":[{"commit":"ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6","subject":"Merge \"Fixed renaming topic breaking service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab50fcebc2be5b2fab3ba5bef2b3d6a9c3dd62a6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-17 20:48:44.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7963bbebaacb5b26da88ff96c42ea182eb398809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7963bbebaacb5b26da88ff96c42ea182eb398809"}]},"branch":"refs/heads/master"},"2b647a6644287c372d6a638e65c7902a01af3827":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-06-18 19:30:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/10"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 19:30:03.000000000","tz":0},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b647a6644287c372d6a638e65c7902a01af3827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b647a6644287c372d6a638e65c7902a01af3827"}]},"branch":"refs/heads/master"},"1dd782e207c9f23aeb3d3820e0b4d4f7ac10aabc":{"kind":"REWORK","_number":11,"created":"2019-06-18 20:27:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/11"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 20:27:44.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1dd782e207c9f23aeb3d3820e0b4d4f7ac10aabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1dd782e207c9f23aeb3d3820e0b4d4f7ac10aabc"}]},"branch":"refs/heads/master"},"e4b1b3e79e94541471e6b4f8ca7ec106e212e3ef":{"kind":"REWORK","_number":12,"created":"2019-06-18 20:32:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/12"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 20:31:48.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e4b1b3e79e94541471e6b4f8ca7ec106e212e3ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e4b1b3e79e94541471e6b4f8ca7ec106e212e3ef"}]},"branch":"refs/heads/master"},"5e557af4b44108a591c2b9ff7280248010cdaafa":{"kind":"REWORK","_number":13,"created":"2019-06-18 20:34:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/13"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 20:33:58.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e557af4b44108a591c2b9ff7280248010cdaafa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e557af4b44108a591c2b9ff7280248010cdaafa"}]},"branch":"refs/heads/master"},"092b04a98190e3ac19023bb3382129ed205dd81a":{"kind":"REWORK","_number":14,"created":"2019-07-15 06:24:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/14"}}},"commit":{"parents":[{"commit":"762daa5507152be96956af4dd01c5fa060faa5f9","subject":"Merge \"Fixed akamai default wsdl path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/762daa5507152be96956af4dd01c5fa060faa5f9"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-07-15 06:23:52.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/092b04a98190e3ac19023bb3382129ed205dd81a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/092b04a98190e3ac19023bb3382129ed205dd81a"}]},"branch":"refs/heads/master"},"9884194b17645afeb6afd7690d8bf6f8339c7070":{"kind":"REWORK","_number":15,"created":"2019-07-29 15:35:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/15"}}},"commit":{"parents":[{"commit":"762daa5507152be96956af4dd01c5fa060faa5f9","subject":"Merge \"Fixed akamai default wsdl path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/762daa5507152be96956af4dd01c5fa060faa5f9"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-07-29 15:35:56.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9884194b17645afeb6afd7690d8bf6f8339c7070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9884194b17645afeb6afd7690d8bf6f8339c7070"}]},"branch":"refs/heads/master"},"37c26d1ba19df5db867a7afb6a57787929972300":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-07-29 21:05:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/16"}}},"commit":{"parents":[{"commit":"076f9fce0ac95b4b628f0d95cfe6e04b7b3988c1","subject":"Merge \"Improve log message for better understanding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/076f9fce0ac95b4b628f0d95cfe6e04b7b3988c1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-07-29 21:05:11.000000000","tz":0},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nThe Newton Upgrade guide contains steps on how\nto move from the pool-manager implementation\nto the new worker/producer implementation.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/37c26d1ba19df5db867a7afb6a57787929972300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/37c26d1ba19df5db867a7afb6a57787929972300"}]},"branch":"refs/heads/master"},"418858ef0ff7b9c06a5ca33cd2de0c6dd2effc53":{"kind":"REWORK","_number":17,"created":"2019-09-18 03:00:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/17"}}},"commit":{"parents":[{"commit":"6d8cd21580a5d5edaf0958851683baa09a175289","subject":"Refactored service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d8cd21580a5d5edaf0958851683baa09a175289"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-18 02:59:55.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/418858ef0ff7b9c06a5ca33cd2de0c6dd2effc53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/418858ef0ff7b9c06a5ca33cd2de0c6dd2effc53"}]},"branch":"refs/heads/master"},"f1be763ec2bb50591917ddca1df2a678edc96fd2":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-09-18 03:02:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/18"}}},"commit":{"parents":[{"commit":"6da97079513a69802e08fa192dd8907cda9b2a3d","subject":"Refactored service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6da97079513a69802e08fa192dd8907cda9b2a3d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-18 03:01:42.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f1be763ec2bb50591917ddca1df2a678edc96fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f1be763ec2bb50591917ddca1df2a678edc96fd2"}]},"branch":"refs/heads/master"},"837fac317033c15f4d455f6de0409fcdcd473d4f":{"kind":"REWORK","_number":19,"created":"2019-09-18 03:22:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/19"}}},"commit":{"parents":[{"commit":"6da97079513a69802e08fa192dd8907cda9b2a3d","subject":"Refactored service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6da97079513a69802e08fa192dd8907cda9b2a3d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-18 03:22:12.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/837fac317033c15f4d455f6de0409fcdcd473d4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/837fac317033c15f4d455f6de0409fcdcd473d4f"}]},"branch":"refs/heads/master"},"81ad96cab74b1a715e036627ad20640cd6674711":{"kind":"REWORK","_number":20,"created":"2019-09-19 01:32:53.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/20"}}},"commit":{"parents":[{"commit":"6da97079513a69802e08fa192dd8907cda9b2a3d","subject":"Refactored service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6da97079513a69802e08fa192dd8907cda9b2a3d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-19 01:32:40.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81ad96cab74b1a715e036627ad20640cd6674711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81ad96cab74b1a715e036627ad20640cd6674711"}]},"branch":"refs/heads/master"},"12af1b971eb51d446af58de78c2b211ee43b95f9":{"kind":"REWORK","_number":21,"created":"2019-09-19 01:40:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/21"}}},"commit":{"parents":[{"commit":"6da97079513a69802e08fa192dd8907cda9b2a3d","subject":"Refactored service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6da97079513a69802e08fa192dd8907cda9b2a3d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-19 01:40:48.000000000","tz":-420},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12af1b971eb51d446af58de78c2b211ee43b95f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12af1b971eb51d446af58de78c2b211ee43b95f9"}]},"branch":"refs/heads/master"},"52ad1c4c13d1fead646435634e4b07bb32c68d28":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-09-24 11:10:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/76/665476/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/665476/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/665476/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/665476/22"}}},"commit":{"parents":[{"commit":"a09064a5d15859703b97d61a1f014681a17799c6","subject":"Refactored service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a09064a5d15859703b97d61a1f014681a17799c6"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-14 21:29:10.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-24 11:10:54.000000000","tz":0},"subject":"Removed deprecated pool-manager implementation","message":"Removed deprecated pool-manager implementation\n\nThe pool-manager has been deprecated since Newton and\nis now being removed with the Train release.\n\nChange-Id: Ib258043d492bb25e1c061d765b40885317c73804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/52ad1c4c13d1fead646435634e4b07bb32c68d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/52ad1c4c13d1fead646435634e4b07bb32c68d28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
