)]}'
{"id":"openstack%2Fdesignate~665865","triplet_id":"openstack%2Fdesignate~master~I20c297bd3f802779b53613f1b705d835cfdedfa0","project":"openstack/designate","branch":"master","topic":"improved_tests","hashtags":[],"change_id":"I20c297bd3f802779b53613f1b705d835cfdedfa0","subject":"Rewrote bind9 backend unit tests","status":"MERGED","created":"2019-06-18 05:46:58.000000000","updated":"2019-06-27 17:51:58.000000000","submitted":"2019-06-27 17:51:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665865-1561657918702-6b32c878","meta_rev_id":"1cfc2a27ec501d534572cb817a1783ab0476459e","_number":665865,"virtual_id_number":665865,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":2,"date":"2019-06-27 17:51:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-27 13:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-06-25 20:28:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-27 16:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 23:50:09.000000000","updated_by":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"reviewer":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"state":"REVIEWER"},{"updated":"2019-06-27 13:38:55.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-27 17:51:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba4ba2f1b5de607d831dbdba178132cca43132c3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 05:46:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a03b4bb956c2b603f49066f0dffff780d88ec67d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 05:57:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"88ce46ec816ce589eb77067a038b461b95f7ffb3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 06:02:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"db5fe0344a599a7a2c39b498797885fcbde56c74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 07:18:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/65/665865/3/check/designate-bind9/d2da2ff/ : SUCCESS in 39m 59s\n- designate-pdns4 http://logs.openstack.org/65/665865/3/check/designate-pdns4/dc80885/ : SUCCESS in 37m 46s\n- designate-bind9-manager-model http://logs.openstack.org/65/665865/3/check/designate-bind9-manager-model/c7f9eb2/ : SUCCESS in 37m 38s\n- designate-bind9-py27 http://logs.openstack.org/65/665865/3/check/designate-bind9-py27/ab1d6c9/ : SUCCESS in 37m 08s\n- designate-pdns4-py27 http://logs.openstack.org/65/665865/3/check/designate-pdns4-py27/32168ee/ : SUCCESS in 38m 54s\n- designate-pdns4-postgres http://logs.openstack.org/65/665865/3/check/designate-pdns4-postgres/c3c687d/ : SUCCESS in 39m 40s\n- designate-grenade-pdns4 http://logs.openstack.org/65/665865/3/check/designate-grenade-pdns4/f0ac590/ : SUCCESS in 1h 00m 30s\n- openstack-tox-cover http://logs.openstack.org/65/665865/3/check/openstack-tox-cover/d1b3c12/cover/ : SUCCESS in 7m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/665865/3/check/openstack-tox-lower-constraints/80410cb/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/65/665865/3/check/openstack-tox-pep8/6136cac/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/65/665865/3/check/openstack-tox-py27/cfbef5e/ : SUCCESS in 5m 46s\n- openstack-tox-py36 http://logs.openstack.org/65/665865/3/check/openstack-tox-py36/26c21d7/ : SUCCESS in 5m 36s\n- openstack-tox-py37 http://logs.openstack.org/65/665865/3/check/openstack-tox-py37/905ab2d/ : SUCCESS in 6m 13s\n- openstack-tox-docs http://logs.openstack.org/65/665865/3/check/openstack-tox-docs/a53e865/html/ : SUCCESS in 5m 13s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/65/665865/3/check/neutron-tempest-plugin-designate-scenario/b89059a/ : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":3},{"id":"fbda88b5a51412529011f5fbdbcbed28ebaee0e7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-18 18:41:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"18bd7f4fc0eeba52bc746367f87ca7174eb4edb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:08:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/65/665865/4/check/designate-bind9/cb2653c/ : SUCCESS in 44m 45s\n- designate-pdns4 http://logs.openstack.org/65/665865/4/check/designate-pdns4/06bfed0/ : SUCCESS in 47m 48s\n- designate-bind9-manager-model http://logs.openstack.org/65/665865/4/check/designate-bind9-manager-model/1f327f0/ : SUCCESS in 42m 32s\n- designate-bind9-py27 http://logs.openstack.org/65/665865/4/check/designate-bind9-py27/91e0779/ : SUCCESS in 36m 50s\n- designate-pdns4-py27 http://logs.openstack.org/65/665865/4/check/designate-pdns4-py27/091f5f6/ : SUCCESS in 46m 37s\n- designate-pdns4-postgres http://logs.openstack.org/65/665865/4/check/designate-pdns4-postgres/c531f23/ : SUCCESS in 43m 20s\n- designate-grenade-pdns4 http://logs.openstack.org/65/665865/4/check/designate-grenade-pdns4/f10808a/ : SUCCESS in 1h 03m 30s\n- openstack-tox-cover http://logs.openstack.org/65/665865/4/check/openstack-tox-cover/bc55bc2/cover/ : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/665865/4/check/openstack-tox-lower-constraints/6017b5e/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/65/665865/4/check/openstack-tox-pep8/497c632/ : SUCCESS in 7m 40s\n- openstack-tox-py27 http://logs.openstack.org/65/665865/4/check/openstack-tox-py27/edb453a/ : SUCCESS in 7m 06s\n- openstack-tox-py36 http://logs.openstack.org/65/665865/4/check/openstack-tox-py36/180a999/ : SUCCESS in 7m 47s\n- openstack-tox-py37 http://logs.openstack.org/65/665865/4/check/openstack-tox-py37/eda9b70/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/65/665865/4/check/openstack-tox-docs/9fcf425/html/ : SUCCESS in 6m 46s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/65/665865/4/check/neutron-tempest-plugin-designate-scenario/c544537/ : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":4},{"id":"fd62e9a7a55f0b8f87b86d09f5ebcd3689e36075","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-06-18 23:50:09.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8efd5266b43d9653d46dcf7089e16c6cc7edf20a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-24 00:56:33.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b627a62344187f51523d21985ad6ab372915f4a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 02:01:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/65/665865/5/check/designate-bind9/6fdffcd/ : SUCCESS in 40m 15s\n- designate-pdns4 http://logs.openstack.org/65/665865/5/check/designate-pdns4/1a7668d/ : SUCCESS in 37m 41s\n- designate-bind9-manager-model http://logs.openstack.org/65/665865/5/check/designate-bind9-manager-model/e6e14a1/ : SUCCESS in 40m 06s\n- designate-bind9-py27 http://logs.openstack.org/65/665865/5/check/designate-bind9-py27/53202e1/ : SUCCESS in 39m 57s\n- designate-pdns4-py27 http://logs.openstack.org/65/665865/5/check/designate-pdns4-py27/24426c7/ : SUCCESS in 37m 48s\n- designate-pdns4-postgres http://logs.openstack.org/65/665865/5/check/designate-pdns4-postgres/070f62f/ : SUCCESS in 36m 15s\n- designate-grenade-pdns4 http://logs.openstack.org/65/665865/5/check/designate-grenade-pdns4/8fb6b3c/ : SUCCESS in 52m 47s\n- openstack-tox-cover http://logs.openstack.org/65/665865/5/check/openstack-tox-cover/188a296/cover/ : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/665865/5/check/openstack-tox-lower-constraints/6e2ba94/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/65/665865/5/check/openstack-tox-pep8/d20f048/ : SUCCESS in 6m 32s\n- openstack-tox-py27 http://logs.openstack.org/65/665865/5/check/openstack-tox-py27/a1116c0/ : SUCCESS in 6m 11s\n- openstack-tox-py36 http://logs.openstack.org/65/665865/5/check/openstack-tox-py36/2da1a09/ : SUCCESS in 6m 33s\n- openstack-tox-py37 http://logs.openstack.org/65/665865/5/check/openstack-tox-py37/71b84d7/ : SUCCESS in 7m 47s\n- openstack-tox-docs http://logs.openstack.org/65/665865/5/check/openstack-tox-docs/ff05698/html/ : SUCCESS in 6m 09s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/65/665865/5/check/neutron-tempest-plugin-designate-scenario/6aefb23/ : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":5},{"id":"c038b2e3285b473f954803c262c6cd7282bc7cfe","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-25 20:28:01.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1646016c9c7744b728a2d045cfa24b885b9b89cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 21:34:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/65/665865/6/check/designate-bind9/5a62846/ : SUCCESS in 38m 11s\n- designate-pdns4 http://logs.openstack.org/65/665865/6/check/designate-pdns4/28923ae/ : SUCCESS in 43m 24s\n- designate-bind9-manager-model http://logs.openstack.org/65/665865/6/check/designate-bind9-manager-model/b6ba314/ : SUCCESS in 41m 16s\n- designate-bind9-py27 http://logs.openstack.org/65/665865/6/check/designate-bind9-py27/d7c94df/ : SUCCESS in 44m 37s\n- designate-pdns4-py27 http://logs.openstack.org/65/665865/6/check/designate-pdns4-py27/e8ff8e3/ : SUCCESS in 41m 55s\n- designate-pdns4-postgres http://logs.openstack.org/65/665865/6/check/designate-pdns4-postgres/1e0feec/ : SUCCESS in 37m 51s\n- designate-grenade-pdns4 http://logs.openstack.org/65/665865/6/check/designate-grenade-pdns4/aea9c41/ : SUCCESS in 55m 17s\n- openstack-tox-cover http://logs.openstack.org/65/665865/6/check/openstack-tox-cover/7e0119b/cover/ : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/665865/6/check/openstack-tox-lower-constraints/24fff1b/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/65/665865/6/check/openstack-tox-pep8/2e883d7/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/65/665865/6/check/openstack-tox-py27/eb65e62/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/65/665865/6/check/openstack-tox-py36/6bd9e98/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/65/665865/6/check/openstack-tox-py37/7699d30/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/65/665865/6/check/openstack-tox-docs/007c9a5/html/ : SUCCESS in 3m 18s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/65/665865/6/check/neutron-tempest-plugin-designate-scenario/b8f837e/ : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":6},{"id":"e4b1c11e6fe17ff8b8e9b4cc8d8b16b57f59e2e2","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-27 13:38:55.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"0b8c4642cbb34aaad97fdc22b9b60201cb39a02c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-27 16:43:20.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"459d6d4712c0f12bd5c2d72105b537b73ee244be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 16:43:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4f2d4c18647785d774b6ed659f5d7931ac6e3ad9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 17:51:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"608820d921b786b1329fbd1732194b3b697979ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 17:51:58.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/65/665865/6/gate/designate-bind9/a29ea95/ : SUCCESS in 43m 33s\n- designate-pdns4 http://logs.openstack.org/65/665865/6/gate/designate-pdns4/895aaf7/ : SUCCESS in 40m 07s\n- designate-bind9-manager-model http://logs.openstack.org/65/665865/6/gate/designate-bind9-manager-model/504deb8/ : SUCCESS in 40m 43s\n- designate-bind9-py27 http://logs.openstack.org/65/665865/6/gate/designate-bind9-py27/0dd6ec2/ : SUCCESS in 40m 35s\n- designate-pdns4-py27 http://logs.openstack.org/65/665865/6/gate/designate-pdns4-py27/11b8990/ : SUCCESS in 45m 25s\n- designate-pdns4-postgres http://logs.openstack.org/65/665865/6/gate/designate-pdns4-postgres/ef96b64/ : SUCCESS in 40m 01s\n- designate-grenade-pdns4 http://logs.openstack.org/65/665865/6/gate/designate-grenade-pdns4/e21ee06/ : SUCCESS in 1h 01m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/665865/6/gate/openstack-tox-lower-constraints/d45f457/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/65/665865/6/gate/openstack-tox-pep8/36119eb/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/65/665865/6/gate/openstack-tox-py27/a44ebba/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/65/665865/6/gate/openstack-tox-py36/2d369aa/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/65/665865/6/gate/openstack-tox-py37/e77f3a2/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/65/665865/6/gate/openstack-tox-docs/60b4976/html/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"08684a2c8e1f33362e142ae6e50f5067e9424860","revisions":{"d787945e1fc57c412cb3c088cc9e21d68530adbd":{"kind":"REWORK","_number":1,"created":"2019-06-18 05:46:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/65/665865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/65/665865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/65/665865/1"}}},"commit":{"parents":[{"commit":"eb2443603bbceec1c3a2f2371e66d734da046d9e","subject":"Improving test coverage of designate backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb2443603bbceec1c3a2f2371e66d734da046d9e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 05:39:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 05:45:55.000000000","tz":-420},"subject":"Rewrote bind9 backend unit tests","message":"Rewrote bind9 backend unit tests\n\n* Re-wrote bind9 tests.\n* Removed functional test.\n* Minor updates to other tests for consistency.\n\nChange-Id: I20c297bd3f802779b53613f1b705d835cfdedfa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d787945e1fc57c412cb3c088cc9e21d68530adbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d787945e1fc57c412cb3c088cc9e21d68530adbd"}]},"branch":"refs/heads/master"},"b9fdba8959e0c7824bf5935d65c8877618c680f9":{"kind":"REWORK","_number":2,"created":"2019-06-18 05:57:27.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/65/665865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/65/665865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/65/665865/2"}}},"commit":{"parents":[{"commit":"eb2443603bbceec1c3a2f2371e66d734da046d9e","subject":"Improving test coverage of designate backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb2443603bbceec1c3a2f2371e66d734da046d9e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 05:39:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 05:57:03.000000000","tz":-420},"subject":"Rewrote bind9 backend unit tests","message":"Rewrote bind9 backend unit tests\n\n* Re-wrote bind9 tests.\n* Removed functional test, as this is already covered.\n* Minor updates to other tests for consistency.\n\nChange-Id: I20c297bd3f802779b53613f1b705d835cfdedfa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9fdba8959e0c7824bf5935d65c8877618c680f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9fdba8959e0c7824bf5935d65c8877618c680f9"}]},"branch":"refs/heads/master"},"98c38ddd6c08ae2cdac64d6274bea7aef0b51718":{"kind":"REWORK","_number":3,"created":"2019-06-18 06:02:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/65/665865/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/65/665865/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/65/665865/3"}}},"commit":{"parents":[{"commit":"eb2443603bbceec1c3a2f2371e66d734da046d9e","subject":"Improving test coverage of designate backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb2443603bbceec1c3a2f2371e66d734da046d9e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 05:39:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 06:01:52.000000000","tz":-420},"subject":"Rewrote bind9 backend unit tests","message":"Rewrote bind9 backend unit tests\n\n* Re-wrote bind9 tests.\n* Removed functional test, as this is already covered.\n* Minor updates to other tests for consistency.\n\nChange-Id: I20c297bd3f802779b53613f1b705d835cfdedfa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98c38ddd6c08ae2cdac64d6274bea7aef0b51718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98c38ddd6c08ae2cdac64d6274bea7aef0b51718"}]},"branch":"refs/heads/master"},"cece6d62141510957d1adbd8ace2786a01f160e0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-18 18:41:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/65/665865/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/65/665865/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/65/665865/4"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 05:39:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 18:41:16.000000000","tz":0},"subject":"Rewrote bind9 backend unit tests","message":"Rewrote bind9 backend unit tests\n\n* Re-wrote bind9 tests.\n* Removed functional test, as this is already covered.\n* Minor updates to other tests for consistency.\n\nChange-Id: I20c297bd3f802779b53613f1b705d835cfdedfa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cece6d62141510957d1adbd8ace2786a01f160e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cece6d62141510957d1adbd8ace2786a01f160e0"}]},"branch":"refs/heads/master"},"41b67e69fdc7cc689d5926806bf2a8dfc7116699":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-24 00:56:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/65/665865/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/65/665865/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/65/665865/5"}}},"commit":{"parents":[{"commit":"a1d2f371093a90d0bbbbf1b3bcba04ebeb1a92a4","subject":"Merge \"Only register dynamic pool options when running designate-manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a1d2f371093a90d0bbbbf1b3bcba04ebeb1a92a4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 05:39:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-24 00:56:33.000000000","tz":0},"subject":"Rewrote bind9 backend unit tests","message":"Rewrote bind9 backend unit tests\n\n* Re-wrote bind9 tests.\n* Removed functional test, as this is already covered.\n* Minor updates to other tests for consistency.\n\nChange-Id: I20c297bd3f802779b53613f1b705d835cfdedfa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/41b67e69fdc7cc689d5926806bf2a8dfc7116699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/41b67e69fdc7cc689d5926806bf2a8dfc7116699"}]},"branch":"refs/heads/master"},"08684a2c8e1f33362e142ae6e50f5067e9424860":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-25 20:28:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/65/665865/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/65/665865/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/665865/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/65/665865/6"}}},"commit":{"parents":[{"commit":"fcc453d487ea759a5af4f3d7773334eec560cecf","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fcc453d487ea759a5af4f3d7773334eec560cecf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-18 05:39:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-25 20:28:01.000000000","tz":0},"subject":"Rewrote bind9 backend unit tests","message":"Rewrote bind9 backend unit tests\n\n* Re-wrote bind9 tests.\n* Removed functional test, as this is already covered.\n* Minor updates to other tests for consistency.\n\nChange-Id: I20c297bd3f802779b53613f1b705d835cfdedfa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
