)]}'
{"id":"openstack%2Fdesignate~666219","triplet_id":"openstack%2Fdesignate~master~Ibb90c6821dc5253a4112d53b7bd0965162a9c5e0","project":"openstack/designate","branch":"master","topic":"improved_tests","hashtags":[],"change_id":"Ibb90c6821dc5253a4112d53b7bd0965162a9c5e0","subject":"Improved CAA record test coverage","status":"MERGED","created":"2019-06-19 04:57:32.000000000","updated":"2019-06-21 06:48:08.000000000","submitted":"2019-06-21 06:48:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666219-1561099688750-f1e7b0d6","meta_rev_id":"d10795974c0999ff5b2decd62108481e5ef49f92","_number":666219,"virtual_id_number":666219,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":2,"date":"2019-06-21 06:48:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-19 13:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-06-19 21:27:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-20 15:12:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 13:42:29.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-19 21:27:51.000000000","updated_by":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"reviewer":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"state":"REVIEWER"},{"updated":"2019-06-21 06:48:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a874233c5f9ad0d2a01654dd595ccdaabc30379","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-19 04:57:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff02744dcc13926169a54f1b7d5d16b6cc10a784","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-19 05:19:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e189c363caba82690b8fddb50375a48dfe2f616","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 06:25:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/19/666219/2/check/designate-bind9/a42152e/ : SUCCESS in 41m 43s\n- designate-pdns4 http://logs.openstack.org/19/666219/2/check/designate-pdns4/7de6dfe/ : SUCCESS in 34m 36s\n- designate-bind9-manager-model http://logs.openstack.org/19/666219/2/check/designate-bind9-manager-model/13edc9f/ : SUCCESS in 37m 03s\n- designate-bind9-py27 http://logs.openstack.org/19/666219/2/check/designate-bind9-py27/26c067d/ : SUCCESS in 48m 05s\n- designate-pdns4-py27 http://logs.openstack.org/19/666219/2/check/designate-pdns4-py27/4733d01/ : SUCCESS in 39m 15s\n- designate-pdns4-postgres http://logs.openstack.org/19/666219/2/check/designate-pdns4-postgres/e9bb758/ : SUCCESS in 35m 09s\n- designate-grenade-pdns4 http://logs.openstack.org/19/666219/2/check/designate-grenade-pdns4/11f2337/ : SUCCESS in 52m 55s\n- openstack-tox-cover http://logs.openstack.org/19/666219/2/check/openstack-tox-cover/87ba91b/cover/ : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/666219/2/check/openstack-tox-lower-constraints/7a1fb0c/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/19/666219/2/check/openstack-tox-pep8/e4bd7cd/ : SUCCESS in 5m 47s\n- openstack-tox-py27 http://logs.openstack.org/19/666219/2/check/openstack-tox-py27/8f08ded/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/19/666219/2/check/openstack-tox-py36/78c14e0/ : SUCCESS in 5m 15s\n- openstack-tox-py37 http://logs.openstack.org/19/666219/2/check/openstack-tox-py37/097d3cf/ : SUCCESS in 6m 00s\n- openstack-tox-docs http://logs.openstack.org/19/666219/2/check/openstack-tox-docs/0696287/html/ : SUCCESS in 5m 14s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/19/666219/2/check/neutron-tempest-plugin-designate-scenario/f2204fb/ : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":2},{"id":"e076d6bcc638f3184701be011c39fa2ef6cf5d27","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-19 13:42:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c6ef5ac17bf290c9d5fbeb2c40ed8f8891a003c3","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-06-19 21:27:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5b2c8a03f12359128ba33a4bea31cbe1667f1455","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-20 15:12:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0bfc929b5e252dfd1d61754184be81b5e597dd06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 15:12:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"826a4ed3c89d367feb9173cf9b18893f5dc67db3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-20 15:13:07.000000000","message":"Topic changed from object_test_improvements to improved_tests","accounts_in_message":[],"_revision_number":2},{"id":"bf0d961184775f87569ed57164297fcae0ed43f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 16:32:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/19/666219/2/gate/designate-bind9/db913aa/ : SUCCESS in 42m 46s\n- designate-pdns4 http://logs.openstack.org/19/666219/2/gate/designate-pdns4/327ff29/ : SUCCESS in 39m 10s\n- designate-bind9-manager-model http://logs.openstack.org/19/666219/2/gate/designate-bind9-manager-model/21c4d90/ : SUCCESS in 35m 13s\n- designate-bind9-py27 http://logs.openstack.org/19/666219/2/gate/designate-bind9-py27/c43c2c8/ : SUCCESS in 36m 15s\n- designate-pdns4-py27 http://logs.openstack.org/19/666219/2/gate/designate-pdns4-py27/6d7f296/ : SUCCESS in 32m 19s\n- designate-pdns4-postgres http://logs.openstack.org/19/666219/2/gate/designate-pdns4-postgres/57a59cc/ : TIMED_OUT in 1h 13m 22s\n- designate-grenade-pdns4 http://logs.openstack.org/19/666219/2/gate/designate-grenade-pdns4/d1facc7/ : SUCCESS in 59m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/666219/2/gate/openstack-tox-lower-constraints/a64b819/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/19/666219/2/gate/openstack-tox-pep8/a0ca491/ : SUCCESS in 12m 46s\n- openstack-tox-py27 http://logs.openstack.org/19/666219/2/gate/openstack-tox-py27/2e8b0d8/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/19/666219/2/gate/openstack-tox-py36/abd9573/ : SUCCESS in 6m 05s\n- openstack-tox-py37 http://logs.openstack.org/19/666219/2/gate/openstack-tox-py37/da83ee2/ : SUCCESS in 15m 49s\n- openstack-tox-docs http://logs.openstack.org/19/666219/2/gate/openstack-tox-docs/e7ed7e9/html/ : SUCCESS in 11m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6daf475ce3dc4a6e5e714a00f0406ee4ee1c5f24","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-20 16:35:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6960262818e14917ba46fcdca4f61dd269464be8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 17:50:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/19/666219/2/check/designate-bind9/d45951b/ : SUCCESS in 41m 47s\n- designate-pdns4 http://logs.openstack.org/19/666219/2/check/designate-pdns4/562003d/ : SUCCESS in 45m 34s\n- designate-bind9-manager-model http://logs.openstack.org/19/666219/2/check/designate-bind9-manager-model/0e1b5c6/ : SUCCESS in 44m 59s\n- designate-bind9-py27 http://logs.openstack.org/19/666219/2/check/designate-bind9-py27/e88ebfe/ : SUCCESS in 47m 14s\n- designate-pdns4-py27 http://logs.openstack.org/19/666219/2/check/designate-pdns4-py27/2529b25/ : SUCCESS in 40m 33s\n- designate-pdns4-postgres http://logs.openstack.org/19/666219/2/check/designate-pdns4-postgres/d8dc3a3/ : SUCCESS in 43m 42s\n- designate-grenade-pdns4 http://logs.openstack.org/19/666219/2/check/designate-grenade-pdns4/4db3124/ : SUCCESS in 1h 05m 38s\n- openstack-tox-cover http://logs.openstack.org/19/666219/2/check/openstack-tox-cover/d138f20/cover/ : SUCCESS in 8m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/666219/2/check/openstack-tox-lower-constraints/ddb5528/ : SUCCESS in 6m 41s\n- openstack-tox-pep8 http://logs.openstack.org/19/666219/2/check/openstack-tox-pep8/0573154/ : SUCCESS in 6m 44s\n- openstack-tox-py27 http://logs.openstack.org/19/666219/2/check/openstack-tox-py27/59052aa/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/19/666219/2/check/openstack-tox-py36/56f9ae5/ : SUCCESS in 12m 38s\n- openstack-tox-py37 http://logs.openstack.org/19/666219/2/check/openstack-tox-py37/390f012/ : SUCCESS in 6m 19s\n- openstack-tox-docs http://logs.openstack.org/19/666219/2/check/openstack-tox-docs/42ad07a/html/ : SUCCESS in 5m 12s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/19/666219/2/check/neutron-tempest-plugin-designate-scenario/885d139/ : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":2},{"id":"652ad3d4fa4b37b9e807b8ecb75510a735289a10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 17:50:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1a2019e7aeeddd141f95303319b82c3b4123843","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 19:12:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/19/666219/2/gate/designate-bind9/c4b8461/ : TIMED_OUT in 1h 13m 02s\n- designate-pdns4 http://logs.openstack.org/19/666219/2/gate/designate-pdns4/77224f3/ : TIMED_OUT in 1h 13m 07s\n- designate-bind9-manager-model http://logs.openstack.org/19/666219/2/gate/designate-bind9-manager-model/e4839db/ : SUCCESS in 37m 31s\n- designate-bind9-py27 http://logs.openstack.org/19/666219/2/gate/designate-bind9-py27/4398b1f/ : SUCCESS in 40m 43s\n- designate-pdns4-py27 http://logs.openstack.org/19/666219/2/gate/designate-pdns4-py27/80c9aa2/ : SUCCESS in 37m 39s\n- designate-pdns4-postgres http://logs.openstack.org/19/666219/2/gate/designate-pdns4-postgres/725d377/ : SUCCESS in 38m 01s\n- designate-grenade-pdns4 http://logs.openstack.org/19/666219/2/gate/designate-grenade-pdns4/b23a4c6/ : SUCCESS in 1h 00m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/666219/2/gate/openstack-tox-lower-constraints/15dfbf2/ : SUCCESS in 11m 47s\n- openstack-tox-pep8 http://logs.openstack.org/19/666219/2/gate/openstack-tox-pep8/e9bb8a8/ : SUCCESS in 7m 02s\n- openstack-tox-py27 http://logs.openstack.org/19/666219/2/gate/openstack-tox-py27/456d6ae/ : SUCCESS in 7m 25s\n- openstack-tox-py36 http://logs.openstack.org/19/666219/2/gate/openstack-tox-py36/b82e2f7/ : SUCCESS in 13m 26s\n- openstack-tox-py37 http://logs.openstack.org/19/666219/2/gate/openstack-tox-py37/5d2679c/ : SUCCESS in 6m 48s\n- openstack-tox-docs http://logs.openstack.org/19/666219/2/gate/openstack-tox-docs/de802e8/html/ : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4b88e9da5fd33fa916b3669ad17579cb5650753c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-20 19:35:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"439486b4ada665ec28f7b9e82ed3854271e0f199","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 20:47:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/19/666219/2/check/designate-bind9/4d48f4f/ : SUCCESS in 39m 24s\n- designate-pdns4 http://logs.openstack.org/19/666219/2/check/designate-pdns4/2ab58dd/ : SUCCESS in 48m 52s\n- designate-bind9-manager-model http://logs.openstack.org/19/666219/2/check/designate-bind9-manager-model/75c017c/ : SUCCESS in 39m 34s\n- designate-bind9-py27 http://logs.openstack.org/19/666219/2/check/designate-bind9-py27/5c3a953/ : SUCCESS in 46m 24s\n- designate-pdns4-py27 http://logs.openstack.org/19/666219/2/check/designate-pdns4-py27/9e7ae71/ : SUCCESS in 42m 45s\n- designate-pdns4-postgres http://logs.openstack.org/19/666219/2/check/designate-pdns4-postgres/1774315/ : SUCCESS in 42m 39s\n- designate-grenade-pdns4 http://logs.openstack.org/19/666219/2/check/designate-grenade-pdns4/d36357f/ : SUCCESS in 57m 40s\n- openstack-tox-cover http://logs.openstack.org/19/666219/2/check/openstack-tox-cover/8632c6b/cover/ : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/666219/2/check/openstack-tox-lower-constraints/843ff61/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/19/666219/2/check/openstack-tox-pep8/4250c8f/ : SUCCESS in 12m 27s\n- openstack-tox-py27 http://logs.openstack.org/19/666219/2/check/openstack-tox-py27/87671d1/ : SUCCESS in 5m 42s\n- openstack-tox-py36 http://logs.openstack.org/19/666219/2/check/openstack-tox-py36/fcb0b43/ : SUCCESS in 5m 29s\n- openstack-tox-py37 http://logs.openstack.org/19/666219/2/check/openstack-tox-py37/ea35cb7/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/19/666219/2/check/openstack-tox-docs/2344d5a/html/ : SUCCESS in 5m 17s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/19/666219/2/check/neutron-tempest-plugin-designate-scenario/ce79273/ : TIMED_OUT in 1h 03m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b5f9a35cb6044512d0305305093969a0dc659e29","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-20 22:32:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"aa640046297f26ec77252ebd995ab8db40de21fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 23:52:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/19/666219/2/check/designate-bind9/b739666/ : SUCCESS in 39m 03s\n- designate-pdns4 http://logs.openstack.org/19/666219/2/check/designate-pdns4/de77438/ : SUCCESS in 35m 34s\n- designate-bind9-manager-model http://logs.openstack.org/19/666219/2/check/designate-bind9-manager-model/f7a9672/ : SUCCESS in 40m 12s\n- designate-bind9-py27 http://logs.openstack.org/19/666219/2/check/designate-bind9-py27/27bc9d6/ : SUCCESS in 40m 34s\n- designate-pdns4-py27 http://logs.openstack.org/19/666219/2/check/designate-pdns4-py27/058e20e/ : SUCCESS in 40m 08s\n- designate-pdns4-postgres http://logs.openstack.org/19/666219/2/check/designate-pdns4-postgres/0be87a6/ : TIMED_OUT in 1h 13m 17s\n- designate-grenade-pdns4 http://logs.openstack.org/19/666219/2/check/designate-grenade-pdns4/437eed7/ : SUCCESS in 55m 22s\n- openstack-tox-cover http://logs.openstack.org/19/666219/2/check/openstack-tox-cover/edec815/cover/ : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/666219/2/check/openstack-tox-lower-constraints/ef8b4e3/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/19/666219/2/check/openstack-tox-pep8/ab685e9/ : SUCCESS in 12m 38s\n- openstack-tox-py27 http://logs.openstack.org/19/666219/2/check/openstack-tox-py27/bb1b9ce/ : SUCCESS in 12m 38s\n- openstack-tox-py36 http://logs.openstack.org/19/666219/2/check/openstack-tox-py36/bb45383/ : SUCCESS in 5m 35s\n- openstack-tox-py37 http://logs.openstack.org/19/666219/2/check/openstack-tox-py37/fd2ab37/ : SUCCESS in 6m 08s\n- openstack-tox-docs http://logs.openstack.org/19/666219/2/check/openstack-tox-docs/5c3d7c6/html/ : SUCCESS in 6m 25s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/19/666219/2/check/neutron-tempest-plugin-designate-scenario/846bbe6/ : TIMED_OUT in 1h 02m 41s","accounts_in_message":[],"_revision_number":2},{"id":"3e8955b8a7b6f7fa2a1dc2136b0f0c4e1935eb76","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-21 04:51:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8191f6d2c6e07957bb967c136feec018fb032818","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 05:50:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/19/666219/2/check/designate-bind9/eac3ca9/ : SUCCESS in 40m 26s\n- designate-pdns4 http://logs.openstack.org/19/666219/2/check/designate-pdns4/1fce7d8/ : SUCCESS in 40m 19s\n- designate-bind9-manager-model http://logs.openstack.org/19/666219/2/check/designate-bind9-manager-model/7294209/ : SUCCESS in 57m 31s\n- designate-bind9-py27 http://logs.openstack.org/19/666219/2/check/designate-bind9-py27/988cef8/ : SUCCESS in 34m 01s\n- designate-pdns4-py27 http://logs.openstack.org/19/666219/2/check/designate-pdns4-py27/ba8ff1d/ : SUCCESS in 36m 15s\n- designate-pdns4-postgres http://logs.openstack.org/19/666219/2/check/designate-pdns4-postgres/377af4f/ : SUCCESS in 34m 56s\n- designate-grenade-pdns4 http://logs.openstack.org/19/666219/2/check/designate-grenade-pdns4/9db743c/ : SUCCESS in 52m 41s\n- openstack-tox-cover http://logs.openstack.org/19/666219/2/check/openstack-tox-cover/29c98e4/cover/ : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/666219/2/check/openstack-tox-lower-constraints/921dd75/ : SUCCESS in 6m 17s\n- openstack-tox-pep8 http://logs.openstack.org/19/666219/2/check/openstack-tox-pep8/75dad16/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/19/666219/2/check/openstack-tox-py27/93d600e/ : SUCCESS in 5m 54s\n- openstack-tox-py36 http://logs.openstack.org/19/666219/2/check/openstack-tox-py36/9195816/ : SUCCESS in 6m 55s\n- openstack-tox-py37 http://logs.openstack.org/19/666219/2/check/openstack-tox-py37/fdbf459/ : SUCCESS in 6m 30s\n- openstack-tox-docs http://logs.openstack.org/19/666219/2/check/openstack-tox-docs/d208c9f/html/ : SUCCESS in 5m 02s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/19/666219/2/check/neutron-tempest-plugin-designate-scenario/03758b5/ : SUCCESS in 30m 53s","accounts_in_message":[],"_revision_number":2},{"id":"e0d069c9300e230901a338cc4e45110b951c6b18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 05:50:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6711b89af9394cdac8326367809cb6c14a9f34a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 06:48:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/19/666219/2/gate/designate-bind9/6c80d92/ : SUCCESS in 36m 07s\n- designate-pdns4 http://logs.openstack.org/19/666219/2/gate/designate-pdns4/0c56099/ : SUCCESS in 36m 38s\n- designate-bind9-manager-model http://logs.openstack.org/19/666219/2/gate/designate-bind9-manager-model/2bd9db9/ : SUCCESS in 38m 59s\n- designate-bind9-py27 http://logs.openstack.org/19/666219/2/gate/designate-bind9-py27/9254521/ : SUCCESS in 44m 04s\n- designate-pdns4-py27 http://logs.openstack.org/19/666219/2/gate/designate-pdns4-py27/7968e38/ : SUCCESS in 33m 02s\n- designate-pdns4-postgres http://logs.openstack.org/19/666219/2/gate/designate-pdns4-postgres/5b204f5/ : SUCCESS in 37m 06s\n- designate-grenade-pdns4 http://logs.openstack.org/19/666219/2/gate/designate-grenade-pdns4/80c233d/ : SUCCESS in 50m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/666219/2/gate/openstack-tox-lower-constraints/953eade/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/19/666219/2/gate/openstack-tox-pep8/f604503/ : SUCCESS in 6m 46s\n- openstack-tox-py27 http://logs.openstack.org/19/666219/2/gate/openstack-tox-py27/222b78f/ : SUCCESS in 6m 45s\n- openstack-tox-py36 http://logs.openstack.org/19/666219/2/gate/openstack-tox-py36/e5beaa8/ : SUCCESS in 5m 42s\n- openstack-tox-py37 http://logs.openstack.org/19/666219/2/gate/openstack-tox-py37/6254d97/ : SUCCESS in 9m 40s\n- openstack-tox-docs http://logs.openstack.org/19/666219/2/gate/openstack-tox-docs/5d4486e/html/ : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":2},{"id":"d10795974c0999ff5b2decd62108481e5ef49f92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 06:48:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81acdc7a7fa15717aed4419462d1593ab811ceea","revisions":{"369e0b17ca20e269884ba7c6bcb5febc5cbd8246":{"kind":"REWORK","_number":1,"created":"2019-06-19 04:57:32.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/19/666219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/19/666219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/666219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/666219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/666219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/19/666219/1"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-19 04:56:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-19 04:56:54.000000000","tz":-420},"subject":"Improved CAA record test coverage","message":"Improved CAA record test coverage\n\nChange-Id: Ibb90c6821dc5253a4112d53b7bd0965162a9c5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/369e0b17ca20e269884ba7c6bcb5febc5cbd8246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/369e0b17ca20e269884ba7c6bcb5febc5cbd8246"}]},"branch":"refs/heads/master"},"81acdc7a7fa15717aed4419462d1593ab811ceea":{"kind":"REWORK","_number":2,"created":"2019-06-19 05:19:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/19/666219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/19/666219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/666219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/666219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/666219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/19/666219/2"}}},"commit":{"parents":[{"commit":"14bf795a9de172bcbcf291c50facc85c49c6cd09","subject":"Merge \"Added config opts unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/14bf795a9de172bcbcf291c50facc85c49c6cd09"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-19 04:56:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-06-19 05:19:15.000000000","tz":-420},"subject":"Improved CAA record test coverage","message":"Improved CAA record test coverage\n\nChange-Id: Ibb90c6821dc5253a4112d53b7bd0965162a9c5e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81acdc7a7fa15717aed4419462d1593ab811ceea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81acdc7a7fa15717aed4419462d1593ab811ceea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
