)]}'
{"id":"openstack%2Fdesignate~667077","triplet_id":"openstack%2Fdesignate~master~I344d4253515569ece66eee7ff92fac3d4190a065","project":"openstack/designate","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I344d4253515569ece66eee7ff92fac3d4190a065","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-06-24 13:08:38.000000000","updated":"2019-06-25 20:11:03.000000000","submitted":"2019-06-25 20:11:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667077-1561493463592-de56cdb0","meta_rev_id":"ea9807aeb3906e1b92aa3310c6d8b2370bc8168c","_number":667077,"virtual_id_number":667077,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"date":"2019-06-25 11:32:23.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-06-25 20:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-06-25 13:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-24 16:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-06-25 13:29:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 16:13:56.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-06-25 11:32:23.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-06-25 13:29:57.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-06-25 20:11:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8db83d7455c847d754daca8e083c9abc685ef11","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-06-24 13:08:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"927da16c62f0c8f95480ae2be53d087c357d5b3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:14:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/77/667077/1/check/build-openstack-api-ref/8bcbf98/html/ : SUCCESS in 6m 22s\n- designate-bind9 http://logs.openstack.org/77/667077/1/check/designate-bind9/1dacb67/ : SUCCESS in 40m 23s\n- designate-pdns4 http://logs.openstack.org/77/667077/1/check/designate-pdns4/443686d/ : SUCCESS in 37m 01s\n- designate-bind9-manager-model http://logs.openstack.org/77/667077/1/check/designate-bind9-manager-model/86d22d0/ : SUCCESS in 43m 12s\n- designate-bind9-py27 http://logs.openstack.org/77/667077/1/check/designate-bind9-py27/b0f363a/ : SUCCESS in 42m 54s\n- designate-pdns4-py27 http://logs.openstack.org/77/667077/1/check/designate-pdns4-py27/d2a58e7/ : SUCCESS in 40m 59s\n- designate-pdns4-postgres http://logs.openstack.org/77/667077/1/check/designate-pdns4-postgres/b827cdc/ : SUCCESS in 39m 54s\n- designate-grenade-pdns4 http://logs.openstack.org/77/667077/1/check/designate-grenade-pdns4/40065ea/ : SUCCESS in 58m 25s\n- openstack-tox-cover http://logs.openstack.org/77/667077/1/check/openstack-tox-cover/9eb8488/cover/ : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/667077/1/check/openstack-tox-lower-constraints/ddf0f95/ : SUCCESS in 6m 22s\n- openstack-tox-pep8 http://logs.openstack.org/77/667077/1/check/openstack-tox-pep8/22f535c/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/77/667077/1/check/openstack-tox-py27/13e9f0b/ : SUCCESS in 7m 20s\n- openstack-tox-py36 http://logs.openstack.org/77/667077/1/check/openstack-tox-py36/e4dae69/ : SUCCESS in 5m 06s\n- openstack-tox-py37 http://logs.openstack.org/77/667077/1/check/openstack-tox-py37/be7e9ff/ : SUCCESS in 7m 23s\n- openstack-tox-docs http://logs.openstack.org/77/667077/1/check/openstack-tox-docs/a03450d/html/ : SUCCESS in 4m 37s\n- build-openstack-releasenotes http://logs.openstack.org/77/667077/1/check/build-openstack-releasenotes/2bce543/html/ : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"db4ef0d9f8665bcd2ee2969a9cf6c6d0a6444d7e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-24 16:13:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e63f4cdecd1d8d43528452e3b83efc8a60b5f9e6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-25 11:32:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17d5f53ed4b2159370ac9adbafdaac3c391f317e","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-06-25 13:29:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6817c042fe137bef409e219d9f58c77050b0c080","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 13:30:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9a05a10c08e0e5bf3c6f13c1e3dfeb5241e3851","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 15:04:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/77/667077/1/gate/build-openstack-api-ref/a821ea3/html/ : SUCCESS in 3m 06s\n- designate-bind9 http://logs.openstack.org/77/667077/1/gate/designate-bind9/d7cb519/ : RETRY_LIMIT in 21m 20s\n- designate-pdns4 http://logs.openstack.org/77/667077/1/gate/designate-pdns4/f2952c6/ : RETRY_LIMIT in 19m 45s\n- designate-bind9-manager-model finger://ze02.openstack.org/e7cf7aa674be419f8d48501f29ed0dd9 : RETRY_LIMIT in 28m 23s\n- designate-bind9-py27 http://logs.openstack.org/77/667077/1/gate/designate-bind9-py27/f1a38ab/ : RETRY_LIMIT in 22m 01s\n- designate-pdns4-py27 http://logs.openstack.org/77/667077/1/gate/designate-pdns4-py27/2524f4c/ : RETRY_LIMIT in 22m 07s\n- designate-pdns4-postgres http://logs.openstack.org/77/667077/1/gate/designate-pdns4-postgres/415c127/ : RETRY_LIMIT in 19m 39s\n- designate-grenade-pdns4 finger://ze01.openstack.org/387b0c0b9b8c4221b2e8b160bb43b1b6 : RETRY_LIMIT in 20m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/667077/1/gate/openstack-tox-lower-constraints/bce49be/ : SUCCESS in 4m 14s\n- openstack-tox-pep8 http://logs.openstack.org/77/667077/1/gate/openstack-tox-pep8/b766edf/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/77/667077/1/gate/openstack-tox-py27/22f8b05/ : SUCCESS in 6m 46s\n- openstack-tox-py36 http://logs.openstack.org/77/667077/1/gate/openstack-tox-py36/cd164bc/ : SUCCESS in 4m 34s\n- openstack-tox-py37 http://logs.openstack.org/77/667077/1/gate/openstack-tox-py37/03de91d/ : SUCCESS in 20m 29s\n- openstack-tox-docs http://logs.openstack.org/77/667077/1/gate/openstack-tox-docs/8b9ad16/html/ : SUCCESS in 3m 18s\n- build-openstack-releasenotes http://logs.openstack.org/77/667077/1/gate/build-openstack-releasenotes/4fa191a/html/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2e3a1a735106ee513e45b5bc16bf8e827383f915","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-06-25 17:48:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a85b2d84003b2025935bc33952bf18dc6e1a8fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 19:01:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/77/667077/1/check/build-openstack-api-ref/7c95493/html/ : SUCCESS in 3m 43s\n- designate-bind9 http://logs.openstack.org/77/667077/1/check/designate-bind9/9c554a7/ : SUCCESS in 41m 07s\n- designate-pdns4 http://logs.openstack.org/77/667077/1/check/designate-pdns4/805f3c3/ : SUCCESS in 37m 24s\n- designate-bind9-manager-model http://logs.openstack.org/77/667077/1/check/designate-bind9-manager-model/8f7d4c8/ : SUCCESS in 39m 01s\n- designate-bind9-py27 http://logs.openstack.org/77/667077/1/check/designate-bind9-py27/742c07c/ : SUCCESS in 40m 23s\n- designate-pdns4-py27 http://logs.openstack.org/77/667077/1/check/designate-pdns4-py27/6a90a85/ : SUCCESS in 42m 21s\n- designate-pdns4-postgres http://logs.openstack.org/77/667077/1/check/designate-pdns4-postgres/dfeb60c/ : SUCCESS in 39m 06s\n- designate-grenade-pdns4 http://logs.openstack.org/77/667077/1/check/designate-grenade-pdns4/8db87e6/ : SUCCESS in 1h 05m 37s\n- openstack-tox-cover http://logs.openstack.org/77/667077/1/check/openstack-tox-cover/21b19ab/cover/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/667077/1/check/openstack-tox-lower-constraints/904a0ed/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/77/667077/1/check/openstack-tox-pep8/175a3f9/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/77/667077/1/check/openstack-tox-py27/c83b003/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/77/667077/1/check/openstack-tox-py36/5ac938c/ : SUCCESS in 4m 49s\n- openstack-tox-py37 http://logs.openstack.org/77/667077/1/check/openstack-tox-py37/1cf85e6/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/77/667077/1/check/openstack-tox-docs/ba7ca93/html/ : SUCCESS in 3m 35s\n- build-openstack-releasenotes http://logs.openstack.org/77/667077/1/check/build-openstack-releasenotes/ef8cd89/html/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d2249d3338e4b26094c209121a3cbe775d0abe22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 19:01:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0289bf5c643fcb9ac0508d7102a23746292422ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 20:11:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b3f4b258a7e42fcd681eacacb9b462822b1b87d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 20:11:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/77/667077/1/gate/build-openstack-api-ref/82ca5e5/html/ : SUCCESS in 3m 10s\n- designate-bind9 http://logs.openstack.org/77/667077/1/gate/designate-bind9/e1bcb05/ : SUCCESS in 39m 58s\n- designate-pdns4 http://logs.openstack.org/77/667077/1/gate/designate-pdns4/ae12a2a/ : SUCCESS in 42m 38s\n- designate-bind9-manager-model http://logs.openstack.org/77/667077/1/gate/designate-bind9-manager-model/1e567b4/ : SUCCESS in 36m 18s\n- designate-bind9-py27 http://logs.openstack.org/77/667077/1/gate/designate-bind9-py27/657191d/ : SUCCESS in 41m 23s\n- designate-pdns4-py27 http://logs.openstack.org/77/667077/1/gate/designate-pdns4-py27/758eb39/ : SUCCESS in 37m 47s\n- designate-pdns4-postgres http://logs.openstack.org/77/667077/1/gate/designate-pdns4-postgres/3fca3a3/ : SUCCESS in 41m 12s\n- designate-grenade-pdns4 http://logs.openstack.org/77/667077/1/gate/designate-grenade-pdns4/4b189c5/ : SUCCESS in 1h 03m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/667077/1/gate/openstack-tox-lower-constraints/dd108ee/ : SUCCESS in 3m 15s\n- openstack-tox-pep8 http://logs.openstack.org/77/667077/1/gate/openstack-tox-pep8/21df5ee/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/77/667077/1/gate/openstack-tox-py27/b46fe4e/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/77/667077/1/gate/openstack-tox-py36/20b10b6/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/77/667077/1/gate/openstack-tox-py37/88367a2/ : SUCCESS in 6m 58s\n- openstack-tox-docs http://logs.openstack.org/77/667077/1/gate/openstack-tox-docs/496a1b5/html/ : SUCCESS in 4m 24s\n- build-openstack-releasenotes http://logs.openstack.org/77/667077/1/gate/build-openstack-releasenotes/8e17d2e/html/ : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a15a2da6b2805ea917a262cd1ee706f09c43a06f","revisions":{"a15a2da6b2805ea917a262cd1ee706f09c43a06f":{"kind":"REWORK","_number":1,"created":"2019-06-24 13:08:38.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/77/667077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/667077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/667077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/667077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/667077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/667077/1"}}},"commit":{"parents":[{"commit":"a1d2f371093a90d0bbbbf1b3bcba04ebeb1a92a4","subject":"Merge \"Only register dynamic pool options when running designate-manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a1d2f371093a90d0bbbbf1b3bcba04ebeb1a92a4"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-24 13:07:42.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-24 13:08:37.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I344d4253515569ece66eee7ff92fac3d4190a065\nStory: #2005924\nTask: #34204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a15a2da6b2805ea917a262cd1ee706f09c43a06f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a15a2da6b2805ea917a262cd1ee706f09c43a06f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
