)]}'
{"id":"openstack%2Fdesignate~668493","triplet_id":"openstack%2Fdesignate~master~Id87bd31513517190474f425acc1b6b0d77279108","project":"openstack/designate","branch":"master","topic":"bug/1834993","hashtags":[],"change_id":"Id87bd31513517190474f425acc1b6b0d77279108","subject":"sqlalchemy: Fix migrations for PostgreSQL","status":"ABANDONED","created":"2019-07-01 15:38:18.000000000","updated":"2022-09-24 05:53:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"487c5c61afa3a2dbb17f24d0bff1d78706edb2aa","_number":668493,"virtual_id_number":668493,"owner":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"date":"2019-07-03 10:11:17.000000000","_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2019-07-03 09:51:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-07-04 03:36:29.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"all":[{"value":-1,"date":"2019-07-03 14:02:16.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 08:51:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-07-03 09:51:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 10:11:17.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-07-04 03:36:29.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2021-03-19 19:01:31.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2021-03-19 19:01:31.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2021-03-23 10:38:03.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"state":"CC"}],"messages":[{"id":"35ec3e41291e54706b0e87d2bcc42cfc2a3f57e3","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-07-01 15:38:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"398a91ba41487cb883d82740f169ac9f0b48a124","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 16:43:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/93/668493/1/check/designate-bind9/1399b91/ : SUCCESS in 38m 02s\n- designate-pdns4 http://logs.openstack.org/93/668493/1/check/designate-pdns4/369576c/ : SUCCESS in 38m 42s\n- designate-bind9-manager-model http://logs.openstack.org/93/668493/1/check/designate-bind9-manager-model/75c4f0e/ : SUCCESS in 40m 18s\n- designate-bind9-py27 http://logs.openstack.org/93/668493/1/check/designate-bind9-py27/3be737e/ : SUCCESS in 40m 21s\n- designate-pdns4-py27 http://logs.openstack.org/93/668493/1/check/designate-pdns4-py27/161b9be/ : SUCCESS in 40m 52s\n- designate-pdns4-postgres http://logs.openstack.org/93/668493/1/check/designate-pdns4-postgres/0520c0b/ : SUCCESS in 35m 30s\n- designate-grenade-pdns4 http://logs.openstack.org/93/668493/1/check/designate-grenade-pdns4/8433e63/ : SUCCESS in 57m 56s\n- openstack-tox-cover http://logs.openstack.org/93/668493/1/check/openstack-tox-cover/a4bc718/cover/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/668493/1/check/openstack-tox-lower-constraints/5cefdf2/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/93/668493/1/check/openstack-tox-pep8/642d6d2/ : FAILURE in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/93/668493/1/check/openstack-tox-py27/074ec00/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/93/668493/1/check/openstack-tox-py36/820308b/ : SUCCESS in 4m 43s\n- openstack-tox-py37 http://logs.openstack.org/93/668493/1/check/openstack-tox-py37/572d7c7/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/93/668493/1/check/openstack-tox-docs/87acd59/html/ : SUCCESS in 3m 43s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/93/668493/1/check/neutron-tempest-plugin-designate-scenario/6cb535a/ : FAILURE in 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"cb7a6e6e68b41c5855461dd30bfa533c01fcdfd7","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-07-01 18:34:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f02003ada96341298185504cd8dbaec3df6178ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 19:35:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/93/668493/2/check/designate-bind9/84da866/ : SUCCESS in 38m 38s\n- designate-pdns4 http://logs.openstack.org/93/668493/2/check/designate-pdns4/3258529/ : SUCCESS in 40m 32s\n- designate-bind9-manager-model http://logs.openstack.org/93/668493/2/check/designate-bind9-manager-model/308765f/ : SUCCESS in 39m 32s\n- designate-bind9-py27 http://logs.openstack.org/93/668493/2/check/designate-bind9-py27/d20ef94/ : SUCCESS in 43m 37s\n- designate-pdns4-py27 http://logs.openstack.org/93/668493/2/check/designate-pdns4-py27/8c64dff/ : SUCCESS in 40m 17s\n- designate-pdns4-postgres http://logs.openstack.org/93/668493/2/check/designate-pdns4-postgres/4a3832f/ : SUCCESS in 35m 37s\n- designate-grenade-pdns4 http://logs.openstack.org/93/668493/2/check/designate-grenade-pdns4/10ba1e9/ : SUCCESS in 57m 34s\n- openstack-tox-cover http://logs.openstack.org/93/668493/2/check/openstack-tox-cover/d493f5e/cover/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/668493/2/check/openstack-tox-lower-constraints/d0a91cc/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/93/668493/2/check/openstack-tox-pep8/b2e170e/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/93/668493/2/check/openstack-tox-py27/07b2e13/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/93/668493/2/check/openstack-tox-py36/013df5e/ : SUCCESS in 3m 50s\n- openstack-tox-py37 http://logs.openstack.org/93/668493/2/check/openstack-tox-py37/90ab7b2/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/93/668493/2/check/openstack-tox-docs/373ec6c/html/ : SUCCESS in 3m 12s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/93/668493/2/check/neutron-tempest-plugin-designate-scenario/9c85ec5/ : FAILURE in 37m 46s","accounts_in_message":[],"_revision_number":2},{"id":"0109ad1457393c5bae54c69c60e075377bd5bd3a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-02 02:05:06.000000000","message":"Patch Set 2:\n\nCould you open a bug report for this and associate it with the change?","accounts_in_message":[],"_revision_number":2},{"id":"b21aba227c063eb9d48c29238d9a8555ece8c127","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-02 02:05:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"817bf69af13c86568f201825928d4a800d4f2a1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 03:03:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/93/668493/2/check/designate-bind9/36f44ca/ : SUCCESS in 38m 37s\n- designate-pdns4 http://logs.openstack.org/93/668493/2/check/designate-pdns4/a8e242d/ : SUCCESS in 36m 05s\n- designate-bind9-manager-model http://logs.openstack.org/93/668493/2/check/designate-bind9-manager-model/b3c84be/ : SUCCESS in 38m 03s\n- designate-bind9-py27 http://logs.openstack.org/93/668493/2/check/designate-bind9-py27/eb697f0/ : SUCCESS in 39m 33s\n- designate-pdns4-py27 http://logs.openstack.org/93/668493/2/check/designate-pdns4-py27/9a82f89/ : SUCCESS in 36m 21s\n- designate-pdns4-postgres http://logs.openstack.org/93/668493/2/check/designate-pdns4-postgres/4b81614/ : SUCCESS in 33m 26s\n- designate-grenade-pdns4 http://logs.openstack.org/93/668493/2/check/designate-grenade-pdns4/ff74d7a/ : SUCCESS in 56m 23s\n- openstack-tox-cover http://logs.openstack.org/93/668493/2/check/openstack-tox-cover/3daa367/cover/ : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/668493/2/check/openstack-tox-lower-constraints/6fce0d5/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/93/668493/2/check/openstack-tox-pep8/c9ad93e/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/93/668493/2/check/openstack-tox-py27/544e505/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/93/668493/2/check/openstack-tox-py36/f762df3/ : SUCCESS in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/93/668493/2/check/openstack-tox-py37/a659be6/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/93/668493/2/check/openstack-tox-docs/ebdeb55/html/ : SUCCESS in 3m 19s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/93/668493/2/check/neutron-tempest-plugin-designate-scenario/43e2069/ : FAILURE in 33m 01s","accounts_in_message":[],"_revision_number":2},{"id":"377b8b92581868e69bfb5cf6e3c3dbbadbc5c1fd","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-07-02 06:15:50.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d19493b5ac61bc096831b5ee5744e34b7c51fcf8","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-07-02 06:50:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"73087b04f1cad7bb8e7b6a0af28402ab299035d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 07:53:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/93/668493/4/check/designate-bind9/df5dc39/ : SUCCESS in 44m 41s\n- designate-pdns4 http://logs.openstack.org/93/668493/4/check/designate-pdns4/b86f5f9/ : SUCCESS in 39m 13s\n- designate-bind9-manager-model http://logs.openstack.org/93/668493/4/check/designate-bind9-manager-model/6849146/ : SUCCESS in 45m 18s\n- designate-bind9-py27 http://logs.openstack.org/93/668493/4/check/designate-bind9-py27/defb42f/ : SUCCESS in 48m 12s\n- designate-pdns4-py27 http://logs.openstack.org/93/668493/4/check/designate-pdns4-py27/3c78eb8/ : SUCCESS in 43m 25s\n- designate-pdns4-postgres http://logs.openstack.org/93/668493/4/check/designate-pdns4-postgres/f81fc5a/ : SUCCESS in 43m 57s\n- designate-grenade-pdns4 http://logs.openstack.org/93/668493/4/check/designate-grenade-pdns4/ad9f856/ : SUCCESS in 56m 31s\n- openstack-tox-cover http://logs.openstack.org/93/668493/4/check/openstack-tox-cover/a38696d/cover/ : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/668493/4/check/openstack-tox-lower-constraints/4dae256/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/93/668493/4/check/openstack-tox-pep8/f1b6db2/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/93/668493/4/check/openstack-tox-py27/6218d36/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/93/668493/4/check/openstack-tox-py36/a22d7c6/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/93/668493/4/check/openstack-tox-py37/ad83ac1/ : SUCCESS in 8m 01s\n- openstack-tox-docs http://logs.openstack.org/93/668493/4/check/openstack-tox-docs/71ab85d/html/ : SUCCESS in 5m 09s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/93/668493/4/check/neutron-tempest-plugin-designate-scenario/b05c237/ : SUCCESS in 34m 41s","accounts_in_message":[],"_revision_number":4},{"id":"f8a4063692f5e0378c7fe59bb7d8c638ce5e14eb","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-07-02 08:05:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8724e1cc405cba91aed1e048362b5c5c75f33be3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 09:13:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/93/668493/5/check/designate-bind9/56540a7/ : SUCCESS in 52m 06s\n- designate-pdns4 http://logs.openstack.org/93/668493/5/check/designate-pdns4/6601d36/ : SUCCESS in 51m 15s\n- designate-bind9-manager-model http://logs.openstack.org/93/668493/5/check/designate-bind9-manager-model/b3e3930/ : SUCCESS in 41m 15s\n- designate-bind9-py27 http://logs.openstack.org/93/668493/5/check/designate-bind9-py27/d86ed10/ : SUCCESS in 36m 44s\n- designate-pdns4-py27 http://logs.openstack.org/93/668493/5/check/designate-pdns4-py27/8cffc71/ : SUCCESS in 36m 08s\n- designate-pdns4-postgres http://logs.openstack.org/93/668493/5/check/designate-pdns4-postgres/837c86b/ : SUCCESS in 37m 14s\n- designate-grenade-pdns4 http://logs.openstack.org/93/668493/5/check/designate-grenade-pdns4/618e28f/ : SUCCESS in 1h 01m 08s\n- openstack-tox-cover http://logs.openstack.org/93/668493/5/check/openstack-tox-cover/969049d/cover/ : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/668493/5/check/openstack-tox-lower-constraints/f984e31/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/93/668493/5/check/openstack-tox-pep8/4d2b82d/ : SUCCESS in 7m 07s\n- openstack-tox-py27 http://logs.openstack.org/93/668493/5/check/openstack-tox-py27/0c7ba15/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/93/668493/5/check/openstack-tox-py36/5c14872/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/93/668493/5/check/openstack-tox-py37/023de32/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/93/668493/5/check/openstack-tox-docs/28366c9/html/ : SUCCESS in 6m 33s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/93/668493/5/check/neutron-tempest-plugin-designate-scenario/fadf8b1/ : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":5},{"id":"958927da69c2746e8f59ce37190d64b9b3ff009f","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-07-02 12:30:32.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"7ee48d7cdeb1750e978d610d1ad6564bd14c8963","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 13:40:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/93/668493/6/check/designate-bind9/11d933a/ : SUCCESS in 37m 28s\n- designate-pdns4 http://logs.openstack.org/93/668493/6/check/designate-pdns4/8cc7fb8/ : SUCCESS in 42m 27s\n- designate-bind9-manager-model http://logs.openstack.org/93/668493/6/check/designate-bind9-manager-model/8fda3ef/ : SUCCESS in 36m 36s\n- designate-bind9-py27 http://logs.openstack.org/93/668493/6/check/designate-bind9-py27/41673fd/ : SUCCESS in 37m 50s\n- designate-pdns4-py27 http://logs.openstack.org/93/668493/6/check/designate-pdns4-py27/fd19fce/ : SUCCESS in 33m 45s\n- designate-pdns4-postgres http://logs.openstack.org/93/668493/6/check/designate-pdns4-postgres/f19a0c4/ : FAILURE in 41m 39s\n- designate-grenade-pdns4 http://logs.openstack.org/93/668493/6/check/designate-grenade-pdns4/7a399b3/ : SUCCESS in 1h 04m 05s\n- openstack-tox-cover http://logs.openstack.org/93/668493/6/check/openstack-tox-cover/9fc9f47/cover/ : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/668493/6/check/openstack-tox-lower-constraints/dd5ed25/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/93/668493/6/check/openstack-tox-pep8/767a06f/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/93/668493/6/check/openstack-tox-py27/3fdab55/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/93/668493/6/check/openstack-tox-py36/794272a/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/93/668493/6/check/openstack-tox-py37/2846b0f/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/93/668493/6/check/openstack-tox-docs/f60cf16/html/ : SUCCESS in 3m 48s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/93/668493/6/check/neutron-tempest-plugin-designate-scenario/809c9cb/ : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":6},{"id":"10d6f5458e875a02b7b8c511e9d6b472f79add9f","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-07-03 06:45:59.000000000","message":"Patch Set 6: Workflow-1\n\nI\u0027m putting this on hold, doing the enum migrations like this does not change to enum definition on PostgreSQL.","accounts_in_message":[],"_revision_number":6},{"id":"8e31d1e99b4016801be5beeb0b01083f1368c8f3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-07-03 08:51:29.000000000","message":"Patch Set 6:\n\nIt looks to me like the initial error is fixed by this patch and we just have collected some other issue(s) during the time that the postgres job was broken. So I\u0027d be fine with merging this as a first step and then fixing the other things iteratively.\n\nI\u0027ve proposed to make the postgres job to be non-voting for the time being in order to allow that.","accounts_in_message":[],"_revision_number":6},{"id":"182438d9ba5bfa0630f04d2d470886a2e018fe38","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-07-03 08:52:00.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"23168f3029ce85f5de2b25e0fed2cc9178046d21","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-07-03 09:23:37.000000000","message":"Patch Set 7:\n\nI agree, looking at the CI output the errors come from different parts of the code.\n\nThis commit fixes both errors I mentioned in the commit message, but looking at the final database schema, after running the migrations, the actual enum changes are not being applied on PostgreSQL at all.\nWe need a completely different way for enum migrations on PostgreSQL.\nDo we want to do that in an additional migration file, merging this commit before that, or fix the migrations in place, as they currently cannot be applied at all on PostgreSQL?","accounts_in_message":[],"_revision_number":7},{"id":"c4e21b816ae3fedd46e9a39781f0f263b76e07b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 09:51:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/93/668493/7/check/designate-bind9/61a4a85/ : SUCCESS in 41m 49s\n- designate-pdns4 http://logs.openstack.org/93/668493/7/check/designate-pdns4/bab96cf/ : SUCCESS in 42m 20s\n- designate-bind9-manager-model http://logs.openstack.org/93/668493/7/check/designate-bind9-manager-model/8634ef8/ : SUCCESS in 44m 41s\n- designate-bind9-py27 http://logs.openstack.org/93/668493/7/check/designate-bind9-py27/dd003fd/ : SUCCESS in 45m 39s\n- designate-pdns4-py27 http://logs.openstack.org/93/668493/7/check/designate-pdns4-py27/ab35c41/ : SUCCESS in 45m 28s\n- designate-pdns4-postgres http://logs.openstack.org/93/668493/7/check/designate-pdns4-postgres/d5f24ac/ : FAILURE in 20m 10s (non-voting)\n- designate-grenade-pdns4 http://logs.openstack.org/93/668493/7/check/designate-grenade-pdns4/8e5bf10/ : SUCCESS in 58m 26s\n- openstack-tox-cover http://logs.openstack.org/93/668493/7/check/openstack-tox-cover/8d0c9a7/cover/ : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/668493/7/check/openstack-tox-lower-constraints/ca16180/ : SUCCESS in 3m 07s\n- openstack-tox-pep8 http://logs.openstack.org/93/668493/7/check/openstack-tox-pep8/1f17df0/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/93/668493/7/check/openstack-tox-py27/37c8714/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/93/668493/7/check/openstack-tox-py36/0c9b140/ : SUCCESS in 3m 50s\n- openstack-tox-py37 http://logs.openstack.org/93/668493/7/check/openstack-tox-py37/a8a8466/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/93/668493/7/check/openstack-tox-docs/008396e/html/ : SUCCESS in 4m 54s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/93/668493/7/check/neutron-tempest-plugin-designate-scenario/29af0e7/ : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":7},{"id":"9fcc50abca100d926880ecc2fbee6c06a077a64b","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-07-03 10:11:17.000000000","message":"Patch Set 7:\n\n@sven - if it is these migrations, we need to fix in place.","accounts_in_message":[],"_revision_number":7},{"id":"c34a4bb7036c9120f1fdb527027cb2e977f1aa5d","author":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"date":"2019-07-03 14:02:16.000000000","message":"Patch Set 7: Workflow-1\n\nIt\u0027s these exact migrations that are not being applied on PostgreSQL.","accounts_in_message":[],"_revision_number":7},{"id":"adeef4dfcd4024036358ed45dda5903726e58b96","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-04 03:36:29.000000000","message":"Patch Set 7:\n\nWe probably need something like this\nhttps://github.com/openstack/neutron/blob/master/neutron/db/migration/__init__.py#L119\n\nUnfortunately this is alembic, so can\u0027t be used directly.","accounts_in_message":[],"_revision_number":7},{"id":"eba495f422418c3be0ac0aed10a8d8cf6784b80e","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-19 19:01:31.000000000","message":"Patch Set 7:\n\nI just ran into this bug on when setting up a fresh OS Ussuri to validate my change adding PostgreSQL to puppet-designate (https://review.opendev.org/c/openstack/puppet-designate/+/781756).\nI just applied the patches and they seem to work fine upgrading the database from 100 to 102.\n\nIs there any chance we can get this change merged and the bug fixed?","accounts_in_message":[],"_revision_number":7},{"id":"653c0847ab33aa61039a4f63e66fcd352622e269","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2021-03-23 10:38:03.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\n\u003e I just applied the patches and they seem to work fine upgrading the database from 100 to 102.\n\nI just revisited the change and your previous conversation. Of course the bugfix does not \"work\" as adding entries to a PostgreSQL ENUM seems buggy.\n\n\nWhy not add a switch / condition on the migrate_engine as has been done before, i.e at \nhttps://opendev.org/openstack/designate/src/branch/master/designate/storage/impl_sqlalchemy/migrate_repo/versions/098_fix_service_charset.py#L25\n\nI am thinking about something the likes of ... \n\n--- cut ---\nif migrate_engine.name !\u003d \"postgresql\":\n   ALTER type record_types ADD VALUE \u0027NAPTR\u0027;\n--- cut ---","accounts_in_message":[],"_revision_number":7},{"id":"487c5c61afa3a2dbb17f24d0bff1d78706edb2aa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:53:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"68acfa176550a90f01c1b41fc27533b3b710f574","revisions":{"79536b5e4a4a78249305d62dd084d4b70dae33da":{"kind":"REWORK","_number":1,"created":"2019-07-01 15:38:18.000000000","uploader":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"ref":"refs/changes/93/668493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/668493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/668493/1"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-06-28 12:12:30.000000000","tz":120},"committer":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-07-01 15:37:29.000000000","tz":120},"subject":"sqlalchemy: Update migrations for PostgreSQL","message":"sqlalchemy: Update migrations for PostgreSQL\n\nPostgreSQL requires no name for the column and a name for the enum.\n\nChange-Id: Id87bd31513517190474f425acc1b6b0d77279108\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79536b5e4a4a78249305d62dd084d4b70dae33da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79536b5e4a4a78249305d62dd084d4b70dae33da"}]},"branch":"refs/heads/master"},"63855568bb8ac8193d1e3b2c09a74d4e0e731b22":{"kind":"REWORK","_number":2,"created":"2019-07-01 18:34:54.000000000","uploader":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"ref":"refs/changes/93/668493/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/668493/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/668493/2"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-06-28 12:12:30.000000000","tz":120},"committer":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-07-01 18:33:12.000000000","tz":120},"subject":"sqlalchemy: Update migrations for PostgreSQL","message":"sqlalchemy: Update migrations for PostgreSQL\n\nPostgreSQL requires no name for the column, but does require a name for\nthe enum.\n\nThis fixes:\n\nsqlalchemy.exc.CompileError: PostgreSQL ENUM type requires a name.\n\noslo_db.exception.DBMigrationError: (psycopg2.ProgrammingError) column \"type\" of relation \"recordsets\" already exists\n [SQL: \u0027\\nALTER TABLE recordsets RENAME COLUMN type TO type\u0027] (Background on this error at: http://sqlalche.me/e/f405)\n\nChange-Id: Id87bd31513517190474f425acc1b6b0d77279108\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63855568bb8ac8193d1e3b2c09a74d4e0e731b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63855568bb8ac8193d1e3b2c09a74d4e0e731b22"}]},"branch":"refs/heads/master"},"5fb60a32374109cf81fafb348f04f95d68d74824":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-02 06:15:50.000000000","uploader":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"ref":"refs/changes/93/668493/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/668493/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/668493/3"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-06-28 12:12:30.000000000","tz":120},"committer":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-07-02 06:14:17.000000000","tz":120},"subject":"sqlalchemy: Fix migrations for PostgreSQL","message":"sqlalchemy: Fix migrations for PostgreSQL\n\nPostgreSQL does not support specifying the same name for the column, but\ndoes require a name for the enum.\n\nThis fixes:\n\nsqlalchemy.exc.CompileError: PostgreSQL ENUM type requires a name.\n\noslo_db.exception.DBMigrationError: (psycopg2.ProgrammingError) column \"type\" of relation \"recordsets\" already exists\n [SQL: \u0027\\nALTER TABLE recordsets RENAME COLUMN type TO type\u0027] (Background on this error at: http://sqlalche.me/e/f405)\n\nChange-Id: Id87bd31513517190474f425acc1b6b0d77279108\nCloses-Bug: 1834993\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5fb60a32374109cf81fafb348f04f95d68d74824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5fb60a32374109cf81fafb348f04f95d68d74824"}]},"branch":"refs/heads/master"},"97be6cc3afdf0838fa99ce31f704540f35bc7c4a":{"kind":"REWORK","_number":4,"created":"2019-07-02 06:50:29.000000000","uploader":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"ref":"refs/changes/93/668493/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/668493/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/668493/4"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-06-28 12:12:30.000000000","tz":120},"committer":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-07-02 06:50:17.000000000","tz":120},"subject":"sqlalchemy: Fix migrations for PostgreSQL","message":"sqlalchemy: Fix migrations for PostgreSQL\n\nPostgreSQL does not support specifying the same name for the column, but\ndoes require a name for the enum.\n\nThis fixes:\n\nsqlalchemy.exc.CompileError: PostgreSQL ENUM type requires a name.\n\noslo_db.exception.DBMigrationError: (psycopg2.ProgrammingError) column \"type\" of relation \"recordsets\" already exists\n [SQL: \u0027\\nALTER TABLE recordsets RENAME COLUMN type TO type\u0027] (Background on this error at: http://sqlalche.me/e/f405)\n\nChange-Id: Id87bd31513517190474f425acc1b6b0d77279108\nCloses-Bug: 1834993\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97be6cc3afdf0838fa99ce31f704540f35bc7c4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/97be6cc3afdf0838fa99ce31f704540f35bc7c4a"}]},"branch":"refs/heads/master"},"f5109fdf1d57de7616e2953abd7b34b5a7b67082":{"kind":"REWORK","_number":5,"created":"2019-07-02 08:05:43.000000000","uploader":{"_account_id":27056,"name":"Sven Wegener","email":"sven.wegener@stealer.net","username":"swegener"},"ref":"refs/changes/93/668493/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/668493/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/668493/5"}}},"commit":{"parents":[{"commit":"08684a2c8e1f33362e142ae6e50f5067e9424860","subject":"Rewrote bind9 backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/08684a2c8e1f33362e142ae6e50f5067e9424860"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-06-28 12:12:30.000000000","tz":120},"committer":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-07-02 08:05:27.000000000","tz":120},"subject":"sqlalchemy: Fix migrations for PostgreSQL","message":"sqlalchemy: Fix migrations for PostgreSQL\n\nPostgreSQL does not support specifying the same name for the column, but\ndoes require a name for the enum.\n\nThis fixes:\n\nsqlalchemy.exc.CompileError: PostgreSQL ENUM type requires a name.\n\noslo_db.exception.DBMigrationError: (psycopg2.ProgrammingError) column \"type\" of relation \"recordsets\" already exists\n [SQL: \u0027\\nALTER TABLE recordsets RENAME COLUMN type TO type\u0027] (Background on this error at: http://sqlalche.me/e/f405)\n\nChange-Id: Id87bd31513517190474f425acc1b6b0d77279108\nCloses-Bug: 1834993\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5109fdf1d57de7616e2953abd7b34b5a7b67082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5109fdf1d57de7616e2953abd7b34b5a7b67082"}]},"branch":"refs/heads/master"},"fdffc05e741182fffa949de475d789c04766e68c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-07-02 12:30:32.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/93/668493/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/668493/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/668493/6"}}},"commit":{"parents":[{"commit":"2af3997b0f4298cce75262e3f6d2ded9bff11f6e","subject":"Fix postgres job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2af3997b0f4298cce75262e3f6d2ded9bff11f6e"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-06-28 12:12:30.000000000","tz":120},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-07-02 12:30:32.000000000","tz":0},"subject":"sqlalchemy: Fix migrations for PostgreSQL","message":"sqlalchemy: Fix migrations for PostgreSQL\n\nPostgreSQL does not support specifying the same name for the column, but\ndoes require a name for the enum.\n\nThis fixes:\n\nsqlalchemy.exc.CompileError: PostgreSQL ENUM type requires a name.\n\noslo_db.exception.DBMigrationError: (psycopg2.ProgrammingError) column \"type\" of relation \"recordsets\" already exists\n [SQL: \u0027\\nALTER TABLE recordsets RENAME COLUMN type TO type\u0027] (Background on this error at: http://sqlalche.me/e/f405)\n\nChange-Id: Id87bd31513517190474f425acc1b6b0d77279108\nCloses-Bug: 1834993\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fdffc05e741182fffa949de475d789c04766e68c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fdffc05e741182fffa949de475d789c04766e68c"}]},"branch":"refs/heads/master"},"68acfa176550a90f01c1b41fc27533b3b710f574":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-07-03 08:52:00.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/93/668493/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/668493/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/668493/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/668493/7"}}},"commit":{"parents":[{"commit":"d491cd78ff3ef85d1459977571e527e17c8d08bb","subject":"Fix postgres job and make it non-voting for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d491cd78ff3ef85d1459977571e527e17c8d08bb"}]}],"author":{"name":"Sven Wegener","email":"sven.wegener@inovex.de","date":"2019-06-28 12:12:30.000000000","tz":120},"committer":{"name":"Jens Harbott (frickler)","email":"j.harbott@x-ion.de","date":"2019-07-03 08:52:00.000000000","tz":0},"subject":"sqlalchemy: Fix migrations for PostgreSQL","message":"sqlalchemy: Fix migrations for PostgreSQL\n\nPostgreSQL does not support specifying the same name for the column, but\ndoes require a name for the enum.\n\nThis fixes:\n\nsqlalchemy.exc.CompileError: PostgreSQL ENUM type requires a name.\n\noslo_db.exception.DBMigrationError: (psycopg2.ProgrammingError) column \"type\" of relation \"recordsets\" already exists\n [SQL: \u0027\\nALTER TABLE recordsets RENAME COLUMN type TO type\u0027] (Background on this error at: http://sqlalche.me/e/f405)\n\nChange-Id: Id87bd31513517190474f425acc1b6b0d77279108\nCloses-Bug: 1834993\nSigned-off-by: Sven Wegener \u003csven.wegener@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/68acfa176550a90f01c1b41fc27533b3b710f574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/68acfa176550a90f01c1b41fc27533b3b710f574"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
