)]}'
{"id":"openstack%2Fdesignate~668648","triplet_id":"openstack%2Fdesignate~master~Ia6bb9d3f670f24377da870a1197f90785f249c9c","project":"openstack/designate","branch":"master","topic":"drop-postgres-job","hashtags":[],"change_id":"Ia6bb9d3f670f24377da870a1197f90785f249c9c","subject":"Drop designate-pdns4-postgres job","status":"ABANDONED","created":"2019-07-02 10:19:07.000000000","updated":"2019-07-03 08:44:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bc3df13e2f8fdce101823bb50785d534fcb8f7b4","_number":668648,"virtual_id_number":668648,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2019-07-02 11:19:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2019-07-02 11:32:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 11:19:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"edb8bfa3d9c631bb4566a4e097be46f7c45a1fb0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-07-02 10:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3180757fe8e39e49f35069464b33c621df3f1a4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 11:19:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/48/668648/1/check/designate-bind9/2a856d0/ : SUCCESS in 41m 07s\n- designate-pdns4 http://logs.openstack.org/48/668648/1/check/designate-pdns4/749e61e/ : SUCCESS in 42m 35s\n- designate-bind9-manager-model http://logs.openstack.org/48/668648/1/check/designate-bind9-manager-model/6aa9682/ : SUCCESS in 44m 20s\n- designate-bind9-py27 http://logs.openstack.org/48/668648/1/check/designate-bind9-py27/22440ea/ : SUCCESS in 43m 54s\n- designate-pdns4-py27 http://logs.openstack.org/48/668648/1/check/designate-pdns4-py27/365de62/ : SUCCESS in 40m 10s\n- designate-grenade-pdns4 http://logs.openstack.org/48/668648/1/check/designate-grenade-pdns4/c288497/ : SUCCESS in 55m 27s\n- openstack-tox-cover http://logs.openstack.org/48/668648/1/check/openstack-tox-cover/f3f89c0/cover/ : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/668648/1/check/openstack-tox-lower-constraints/1982a6a/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/48/668648/1/check/openstack-tox-pep8/7ba4373/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/48/668648/1/check/openstack-tox-py27/dbf4e7e/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/48/668648/1/check/openstack-tox-py36/6c4e0d8/ : SUCCESS in 3m 59s\n- openstack-tox-py37 http://logs.openstack.org/48/668648/1/check/openstack-tox-py37/afb5c6b/ : FAILURE in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/48/668648/1/check/openstack-tox-docs/35ad740/html/ : SUCCESS in 3m 31s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/48/668648/1/check/neutron-tempest-plugin-designate-scenario/5f1bc35/ : SUCCESS in 33m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bfd18b76a78f3c69848ef803826e03b419280207","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-07-02 11:32:55.000000000","message":"Patch Set 1: Workflow-1\n\nMaybe there\u0027s an easy fix, see https://review.opendev.org/668657","accounts_in_message":[],"_revision_number":1},{"id":"43ca1b37bec3a9d56422a90e32726a9d7dcf7371","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-07-03 08:44:46.000000000","message":"Abandoned\n\nWe can fix the job instead.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91e00952596d7b7d8817a0b18535c0fbffbb7578","revisions":{"91e00952596d7b7d8817a0b18535c0fbffbb7578":{"kind":"REWORK","_number":1,"created":"2019-07-02 10:19:07.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/48/668648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/48/668648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/668648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/668648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/48/668648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/48/668648/1"}}},"commit":{"parents":[{"commit":"973cafd79b8f4a0866a3474383e74a4e56d0f881","subject":"Merge \"Remove unused code in utils and improved utils testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/973cafd79b8f4a0866a3474383e74a4e56d0f881"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-07-02 10:18:13.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-07-02 10:18:13.000000000","tz":0},"subject":"Drop designate-pdns4-postgres job","message":"Drop designate-pdns4-postgres job\n\nThis didn\u0027t use postgres anymore since we switched from the\ndevstack-gate legacy jobs to the Zuul v3 version.\n\nChange-Id: Ia6bb9d3f670f24377da870a1197f90785f249c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/91e00952596d7b7d8817a0b18535c0fbffbb7578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/91e00952596d7b7d8817a0b18535c0fbffbb7578"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
