)]}'
{"id":"openstack%2Fdesignate~668709","triplet_id":"openstack%2Fdesignate~master~Ie2d41cdd1f536764a3896265eeb7d7ee5208164b","project":"openstack/designate","branch":"master","topic":"dnm_test","hashtags":[],"change_id":"Ie2d41cdd1f536764a3896265eeb7d7ee5208164b","subject":"DNM: Test","status":"ABANDONED","created":"2019-07-02 16:38:08.000000000","updated":"2019-08-24 03:19:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3f741af8dbc645f81d49879741583ab73de18072","_number":668709,"virtual_id_number":668709,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-08-23 23:31:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 23:31:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b011694a908bff33e7ec757652cb62d37664ea40","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-02 16:38:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab2d8610a2fcba34211ea3bd27376f80eaf413e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 17:46:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/09/668709/1/check/designate-bind9/690c841/ : SUCCESS in 38m 46s\n- designate-pdns4 http://logs.openstack.org/09/668709/1/check/designate-pdns4/04caa81/ : SUCCESS in 37m 04s\n- designate-bind9-manager-model http://logs.openstack.org/09/668709/1/check/designate-bind9-manager-model/95858b9/ : SUCCESS in 46m 15s\n- designate-bind9-py27 http://logs.openstack.org/09/668709/1/check/designate-bind9-py27/4268952/ : SUCCESS in 42m 26s\n- designate-pdns4-py27 http://logs.openstack.org/09/668709/1/check/designate-pdns4-py27/3ba7e42/ : SUCCESS in 39m 38s\n- designate-pdns4-postgres http://logs.openstack.org/09/668709/1/check/designate-pdns4-postgres/2b975a6/ : FAILURE in 37m 18s\n- designate-grenade-pdns4 http://logs.openstack.org/09/668709/1/check/designate-grenade-pdns4/45ca59d/ : SUCCESS in 56m 01s\n- openstack-tox-cover http://logs.openstack.org/09/668709/1/check/openstack-tox-cover/f16b8ba/cover/ : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/668709/1/check/openstack-tox-lower-constraints/8326e20/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/09/668709/1/check/openstack-tox-pep8/7b5037f/ : SUCCESS in 6m 33s\n- openstack-tox-py27 http://logs.openstack.org/09/668709/1/check/openstack-tox-py27/ac98de5/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/09/668709/1/check/openstack-tox-py36/4f6c5df/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/09/668709/1/check/openstack-tox-py37/2509d4e/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/09/668709/1/check/openstack-tox-docs/5198fd2/html/ : SUCCESS in 3m 30s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/09/668709/1/check/neutron-tempest-plugin-designate-scenario/600419a/ : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d3a5970562d7797e3437b672909d2a3ae903c0ff","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-29 15:46:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"887e95dc8fd9f67d546b4836ba193be3e575aeba","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-23 22:24:59.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"bd27638fc31bd28fd2e599dad74d321a9220776a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-23 22:25:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ca5fd91e1dd97c7e049e8a8d68bd05abbc0f2f5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 23:31:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d01118ffe29b463dac6c55cf9c3c2556 : SUCCESS in 39m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/783612a26d3742a8b79adf532b47e2d6 : SUCCESS in 37m 38s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/b3c3b0be30394b8baee22bbcc13701ed : SUCCESS in 40m 24s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/2e2c60e29dc6449fbd04968e7b668616 : SUCCESS in 41m 16s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/3a3819ba32e34cc5a4d8d3657c8625ae : SUCCESS in 1h 03m 52s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/52460d92b4664288afcb5c34c0b65888 : FAILURE in 45m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/04faf82bdab34a78a23d27c4c007266c : SUCCESS in 1h 04m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd3abde8fa6342b998c3ef546dbb16c3 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee1fbe429f4f4a9f96db2a3eb62346c9 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd2f5f2efb6460b9e2934b6e018d8e8 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a93d11df0a3f46f4b29e1628a911bd3b : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3191fe82d93248c58f60d83e9b2baf20 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/828d3cb26c754de1aa6952569cc7e274 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/228cba0bb1e14d189d279ed979936d9d : FAILURE in 5m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3f6a9e9077f44056a8dbc228dfd1893d : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":2},{"id":"05133d12a87fbe18ab45a08c4c2bf83e0040b80b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-24 03:19:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7a487591b8e383c0ae9813113d74ddef174827b9","revisions":{"28b7ef2179a3c18975e13964e05479780c8f6240":{"kind":"REWORK","_number":1,"created":"2019-07-02 16:38:08.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/668709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/668709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/668709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/668709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/668709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/668709/1"}}},"commit":{"parents":[{"commit":"fdffc05e741182fffa949de475d789c04766e68c","subject":"sqlalchemy: Fix migrations for PostgreSQL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fdffc05e741182fffa949de475d789c04766e68c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-07-02 16:38:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-07-02 16:38:09.000000000","tz":-420},"subject":"DNM: Test","message":"DNM: Test\n\nChange-Id: Ie2d41cdd1f536764a3896265eeb7d7ee5208164b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28b7ef2179a3c18975e13964e05479780c8f6240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28b7ef2179a3c18975e13964e05479780c8f6240"}]},"branch":"refs/heads/master"},"7a487591b8e383c0ae9813113d74ddef174827b9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-23 22:25:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/668709/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/668709/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/668709/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/668709/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/668709/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/668709/2"}}},"commit":{"parents":[{"commit":"68acfa176550a90f01c1b41fc27533b3b710f574","subject":"sqlalchemy: Fix migrations for PostgreSQL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/68acfa176550a90f01c1b41fc27533b3b710f574"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-07-02 16:38:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-23 22:25:01.000000000","tz":0},"subject":"DNM: Test","message":"DNM: Test\n\nChange-Id: Ie2d41cdd1f536764a3896265eeb7d7ee5208164b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a487591b8e383c0ae9813113d74ddef174827b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a487591b8e383c0ae9813113d74ddef174827b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
