)]}'
{"id":"openstack%2Fdesignate~669154","triplet_id":"openstack%2Fdesignate~master~I26d36c15406436bb3b733d44c490a14e5ece39a6","project":"openstack/designate","branch":"master","topic":"bug/1835407","hashtags":[],"change_id":"I26d36c15406436bb3b733d44c490a14e5ece39a6","subject":"Improve log message for better understanding","status":"MERGED","created":"2019-07-04 11:49:54.000000000","updated":"2019-07-30 03:19:10.000000000","submitted":"2019-07-24 19:28:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669154-1563996495742-8e621402","meta_rev_id":"27d6f1969d126bb4f5c3e1e23382c68d9cf95393","_number":669154,"virtual_id_number":669154,"owner":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-07-24 19:28:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-05 05:36:02.000000000","_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-07-23 10:08:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":2,"date":"2019-07-24 18:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":1,"date":"2019-07-24 18:21:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 11:54:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-07-23 10:08:11.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-07-24 18:21:18.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-07-24 19:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cde25eeef25a0c767fe543eb2d4d0dca69b1b461","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-04 11:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc472ede73723e3df8182907a195c949440a14c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:59:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/54/669154/1/check/designate-bind9/a6d0b6c/ : FAILURE in 31m 01s\n- designate-pdns4 http://logs.openstack.org/54/669154/1/check/designate-pdns4/7abbc48/ : FAILURE in 30m 35s\n- designate-bind9-manager-model http://logs.openstack.org/54/669154/1/check/designate-bind9-manager-model/2d2ebea/ : FAILURE in 30m 28s\n- designate-bind9-py27 http://logs.openstack.org/54/669154/1/check/designate-bind9-py27/578165b/ : FAILURE in 29m 19s\n- designate-pdns4-py27 http://logs.openstack.org/54/669154/1/check/designate-pdns4-py27/de74cfd/ : FAILURE in 26m 25s\n- designate-pdns4-postgres http://logs.openstack.org/54/669154/1/check/designate-pdns4-postgres/13ca193/ : FAILURE in 36m 09s (non-voting)\n- designate-grenade-pdns4 http://logs.openstack.org/54/669154/1/check/designate-grenade-pdns4/2dca127/ : SUCCESS in 1h 02m 48s\n- openstack-tox-cover http://logs.openstack.org/54/669154/1/check/openstack-tox-cover/bc5f3d8/cover/ : SUCCESS in 6m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/669154/1/check/openstack-tox-lower-constraints/a3d9eef/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/54/669154/1/check/openstack-tox-pep8/8a1d444/ : FAILURE in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/54/669154/1/check/openstack-tox-py27/42c3888/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/54/669154/1/check/openstack-tox-py36/63184b1/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/54/669154/1/check/openstack-tox-py37/b85cf7d/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/54/669154/1/check/openstack-tox-docs/05c9296/html/ : SUCCESS in 3m 22s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/54/669154/1/check/neutron-tempest-plugin-designate-scenario/1925b8d/ : FAILURE in 29m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b49eb55a7a2bc95446ec55095cfcb375a6c9a0ae","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-04 15:18:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc8410e9af1abf462d0271e608ccfdcfcacd1fd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 16:37:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/54/669154/2/check/designate-bind9/86cbf72/ : SUCCESS in 43m 05s\n- designate-pdns4 http://logs.openstack.org/54/669154/2/check/designate-pdns4/b1848ed/ : SUCCESS in 49m 31s\n- designate-bind9-manager-model http://logs.openstack.org/54/669154/2/check/designate-bind9-manager-model/1a9bbeb/ : SUCCESS in 48m 39s\n- designate-bind9-py27 http://logs.openstack.org/54/669154/2/check/designate-bind9-py27/9781519/ : SUCCESS in 47m 19s\n- designate-pdns4-py27 http://logs.openstack.org/54/669154/2/check/designate-pdns4-py27/102dbda/ : SUCCESS in 42m 17s\n- designate-pdns4-postgres http://logs.openstack.org/54/669154/2/check/designate-pdns4-postgres/2c8cef9/ : FAILURE in 30m 57s (non-voting)\n- designate-grenade-pdns4 http://logs.openstack.org/54/669154/2/check/designate-grenade-pdns4/f0cf3f0/ : SUCCESS in 1h 06m 55s\n- openstack-tox-cover http://logs.openstack.org/54/669154/2/check/openstack-tox-cover/f7a1d8d/cover/ : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/669154/2/check/openstack-tox-lower-constraints/8bcef89/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/54/669154/2/check/openstack-tox-pep8/c821bcf/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/54/669154/2/check/openstack-tox-py27/0a2de07/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/54/669154/2/check/openstack-tox-py36/6830587/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/54/669154/2/check/openstack-tox-py37/f1015bc/ : SUCCESS in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/54/669154/2/check/openstack-tox-docs/57e2242/html/ : SUCCESS in 3m 02s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/54/669154/2/check/neutron-tempest-plugin-designate-scenario/38008bf/ : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":2},{"id":"4592c4a935524ef4360e2233d8d3eafd99663866","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-04 17:01:57.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ee59a7dcc4c88fd1959496b08107d2628a400273","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-05 00:19:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8726d5c43d2f8ca3202df718093a2e186da13943","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-05 00:21:09.000000000","message":"Patch Set 3:\n\nThank you for reviewing. I fixed your point.","accounts_in_message":[],"_revision_number":3},{"id":"ea9d680a204ac7129a11b0ed0ee3d0dafcb4c432","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-05 00:54:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"20face8a012298e31b39ac1f76f52ec73c6da74d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 01:11:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 http://logs.openstack.org/54/669154/3/check/designate-bind9/ae06b7b/ : SUCCESS in 36m 02s\n- designate-pdns4 http://logs.openstack.org/54/669154/3/check/designate-pdns4/b074fcb/ : SUCCESS in 37m 17s\n- designate-bind9-manager-model http://logs.openstack.org/54/669154/3/check/designate-bind9-manager-model/ecc3ec2/ : FAILURE in 43m 04s\n- designate-bind9-py27 http://logs.openstack.org/54/669154/3/check/designate-bind9-py27/4451e70/ : SUCCESS in 43m 08s\n- designate-pdns4-py27 http://logs.openstack.org/54/669154/3/check/designate-pdns4-py27/9ae1ca1/ : SUCCESS in 35m 49s\n- designate-pdns4-postgres http://logs.openstack.org/54/669154/3/check/designate-pdns4-postgres/7ba3bcc/ : FAILURE in 27m 59s (non-voting)\n- designate-grenade-pdns4 http://logs.openstack.org/54/669154/3/check/designate-grenade-pdns4/bd4e54a/ : SUCCESS in 51m 48s\n- openstack-tox-cover http://logs.openstack.org/54/669154/3/check/openstack-tox-cover/593423e/cover/ : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/669154/3/check/openstack-tox-lower-constraints/98507e3/ : SUCCESS in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/54/669154/3/check/openstack-tox-pep8/9c88f7c/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/54/669154/3/check/openstack-tox-py27/382b189/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/54/669154/3/check/openstack-tox-py36/0b9af42/ : SUCCESS in 3m 52s\n- openstack-tox-py37 http://logs.openstack.org/54/669154/3/check/openstack-tox-py37/9914480/ : SUCCESS in 4m 11s\n- openstack-tox-docs http://logs.openstack.org/54/669154/3/check/openstack-tox-docs/5055e72/html/ : SUCCESS in 2m 58s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/54/669154/3/check/neutron-tempest-plugin-designate-scenario/15ceec5/ : SUCCESS in 34m 32s","accounts_in_message":[],"_revision_number":3},{"id":"ddfb16095c5f70176d99c2545fea023cedf717af","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-05 04:28:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"23a9cb9339074715ef63fdc5bc937ba38d9e04ca","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-05 04:56:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d03ea76d06d2e60b10c83ca130ebdc8dfd80a00e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-05 05:08:58.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"7bf12daca673b8e7f27928111f83c499220f6849","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 05:27:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/54/669154/3/check/designate-bind9/87d3026/ : SUCCESS in 37m 12s\n- designate-pdns4 http://logs.openstack.org/54/669154/3/check/designate-pdns4/ea39bcb/ : SUCCESS in 35m 45s\n- designate-bind9-manager-model http://logs.openstack.org/54/669154/3/check/designate-bind9-manager-model/4a7d9d8/ : SUCCESS in 39m 51s\n- designate-bind9-py27 http://logs.openstack.org/54/669154/3/check/designate-bind9-py27/f120534/ : SUCCESS in 43m 50s\n- designate-pdns4-py27 http://logs.openstack.org/54/669154/3/check/designate-pdns4-py27/d9ab29b/ : SUCCESS in 35m 35s\n- designate-pdns4-postgres http://logs.openstack.org/54/669154/3/check/designate-pdns4-postgres/55d7313/ : FAILURE in 31m 44s (non-voting)\n- designate-grenade-pdns4 http://logs.openstack.org/54/669154/3/check/designate-grenade-pdns4/7877e5a/ : SUCCESS in 53m 31s\n- openstack-tox-cover http://logs.openstack.org/54/669154/3/check/openstack-tox-cover/5db7582/cover/ : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/669154/3/check/openstack-tox-lower-constraints/308f4ce/ : SUCCESS in 2m 55s\n- openstack-tox-pep8 http://logs.openstack.org/54/669154/3/check/openstack-tox-pep8/a3c0d31/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/54/669154/3/check/openstack-tox-py27/c66702b/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/54/669154/3/check/openstack-tox-py36/df701ea/ : SUCCESS in 4m 01s\n- openstack-tox-py37 http://logs.openstack.org/54/669154/3/check/openstack-tox-py37/83537f6/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/54/669154/3/check/openstack-tox-docs/5c8fe1b/html/ : SUCCESS in 3m 12s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/54/669154/3/check/neutron-tempest-plugin-designate-scenario/eb42aa5/ : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":3},{"id":"0c3b409cdd25a98df143acc7d077ee07a82f715d","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-05 05:31:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8797f573062fb264fe0b42d68d0b25e25d6fa675","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-05 05:33:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c659638302031f2dd5ccc455c7361b4f9816891b","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-05 05:36:02.000000000","message":"Patch Set 4:\n\nThank you for reviewing.","accounts_in_message":[],"_revision_number":4},{"id":"c6663bf940cd545c0e05142e326c59c281e70708","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 06:34:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/54/669154/4/check/designate-bind9/7f85593/ : SUCCESS in 37m 26s\n- designate-pdns4 http://logs.openstack.org/54/669154/4/check/designate-pdns4/1bcb837/ : SUCCESS in 39m 26s\n- designate-bind9-manager-model http://logs.openstack.org/54/669154/4/check/designate-bind9-manager-model/031bb69/ : SUCCESS in 37m 24s\n- designate-bind9-py27 http://logs.openstack.org/54/669154/4/check/designate-bind9-py27/fa1d450/ : SUCCESS in 37m 13s\n- designate-pdns4-py27 http://logs.openstack.org/54/669154/4/check/designate-pdns4-py27/946a0c8/ : SUCCESS in 37m 34s\n- designate-pdns4-postgres http://logs.openstack.org/54/669154/4/check/designate-pdns4-postgres/2ba19ff/ : FAILURE in 30m 47s (non-voting)\n- designate-grenade-pdns4 http://logs.openstack.org/54/669154/4/check/designate-grenade-pdns4/9fcae6d/ : SUCCESS in 59m 18s\n- openstack-tox-cover http://logs.openstack.org/54/669154/4/check/openstack-tox-cover/5d1f6c2/cover/ : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/669154/4/check/openstack-tox-lower-constraints/725bf7d/ : SUCCESS in 2m 58s\n- openstack-tox-pep8 http://logs.openstack.org/54/669154/4/check/openstack-tox-pep8/85b221e/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/54/669154/4/check/openstack-tox-py27/49e598f/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/54/669154/4/check/openstack-tox-py36/1b7b958/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/54/669154/4/check/openstack-tox-py37/ff411f6/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/54/669154/4/check/openstack-tox-docs/e36b951/html/ : SUCCESS in 3m 17s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/54/669154/4/check/neutron-tempest-plugin-designate-scenario/c647bd6/ : SUCCESS in 36m 40s","accounts_in_message":[],"_revision_number":4},{"id":"45d5c2eb8da84fa0c2eb745bfb76d1e7bfc9e80d","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-07-23 10:08:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"23110ffe66b5b3e4d65f7e98951b25396dfa0fb7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-24 18:21:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fb85a665c55dca37c7294b111249e45558728b32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 18:21:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"480cfcf8a85dfc4c6c3bbacce3328553b15db4a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 19:28:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 http://logs.openstack.org/54/669154/4/gate/designate-bind9/1718f31/ : SUCCESS in 38m 21s\n- designate-pdns4 http://logs.openstack.org/54/669154/4/gate/designate-pdns4/e5c2460/ : SUCCESS in 43m 19s\n- designate-bind9-manager-model http://logs.openstack.org/54/669154/4/gate/designate-bind9-manager-model/a242b64/ : SUCCESS in 43m 43s\n- designate-bind9-py27 http://logs.openstack.org/54/669154/4/gate/designate-bind9-py27/6ebdd96/ : SUCCESS in 46m 31s\n- designate-pdns4-py27 http://logs.openstack.org/54/669154/4/gate/designate-pdns4-py27/a6ea7ea/ : SUCCESS in 47m 05s\n- designate-grenade-pdns4 http://logs.openstack.org/54/669154/4/gate/designate-grenade-pdns4/53749d2/ : SUCCESS in 1h 04m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/669154/4/gate/openstack-tox-lower-constraints/3f358d8/ : SUCCESS in 5m 52s\n- openstack-tox-pep8 http://logs.openstack.org/54/669154/4/gate/openstack-tox-pep8/011a581/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/54/669154/4/gate/openstack-tox-py27/471ca50/ : SUCCESS in 5m 45s\n- openstack-tox-py36 http://logs.openstack.org/54/669154/4/gate/openstack-tox-py36/f9f2b7d/ : SUCCESS in 6m 54s\n- openstack-tox-py37 http://logs.openstack.org/54/669154/4/gate/openstack-tox-py37/6416a7b/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/54/669154/4/gate/openstack-tox-docs/9cd9361/html/ : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":4},{"id":"b0fd632d997d244d3c533c40ae8893fd67099fde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 19:28:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5bf5317205c8ada3c8085d1864228251a3c116da","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-30 03:18:55.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 756dcbb4afe937d2de0df60b8ddf3219ebf2dc76","accounts_in_message":[],"_revision_number":4},{"id":"0cee1db44d9b7f70102058183441f144e3b551c1","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-30 03:19:10.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit eb6abf7dbd077fa539fab1de6c8532f1729dc725","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e70af61dac1794ecd858c875c44d6b2ab0d59206","revisions":{"b4ec3010b66bfa18348e302cd684182edc6d0031":{"kind":"REWORK","_number":1,"created":"2019-07-04 11:49:54.000000000","uploader":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"ref":"refs/changes/54/669154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/54/669154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/54/669154/1"}}},"commit":{"parents":[{"commit":"d491cd78ff3ef85d1459977571e527e17c8d08bb","subject":"Fix postgres job and make it non-voting for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d491cd78ff3ef85d1459977571e527e17c8d08bb"}]}],"author":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-07-04 11:45:18.000000000","tz":540},"committer":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-07-04 11:45:18.000000000","tz":540},"subject":"Improve log message for better understanding","message":"Improve log message for better understanding\n\nSometimes designate central could not update the Pool configuration.\nVery limited information is logged into the designate central logs\nthrough which we can not find the root cause of failure.\nThis patch adds some more information into the log for better\nunderstanding.\n\nChange-Id: I26d36c15406436bb3b733d44c490a14e5ece39a6\nCloses-Bug: 1835407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4ec3010b66bfa18348e302cd684182edc6d0031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b4ec3010b66bfa18348e302cd684182edc6d0031"}]},"branch":"refs/heads/master"},"d5be2bb790519022864168bc77ba1e1acb61a993":{"kind":"REWORK","_number":2,"created":"2019-07-04 15:18:00.000000000","uploader":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"ref":"refs/changes/54/669154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/54/669154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/54/669154/2"}}},"commit":{"parents":[{"commit":"d491cd78ff3ef85d1459977571e527e17c8d08bb","subject":"Fix postgres job and make it non-voting for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d491cd78ff3ef85d1459977571e527e17c8d08bb"}]}],"author":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-07-04 11:45:18.000000000","tz":540},"committer":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-07-04 15:17:05.000000000","tz":540},"subject":"Improve log message for better understanding","message":"Improve log message for better understanding\n\nSometimes designate central could not update the Pool configuration.\nVery limited information is logged into the designate central logs\nthrough which we can not find the root cause of failure.\nThis patch adds some more information into the log for better\nunderstanding.\n\nChange-Id: I26d36c15406436bb3b733d44c490a14e5ece39a6\nCloses-Bug: 1835407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5be2bb790519022864168bc77ba1e1acb61a993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5be2bb790519022864168bc77ba1e1acb61a993"}]},"branch":"refs/heads/master"},"d494e85647e870fc03b02273d79518a39e0d8974":{"kind":"REWORK","_number":3,"created":"2019-07-05 00:19:17.000000000","uploader":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"ref":"refs/changes/54/669154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/54/669154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/54/669154/3"}}},"commit":{"parents":[{"commit":"d491cd78ff3ef85d1459977571e527e17c8d08bb","subject":"Fix postgres job and make it non-voting for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d491cd78ff3ef85d1459977571e527e17c8d08bb"}]}],"author":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-07-04 11:45:18.000000000","tz":540},"committer":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-07-05 00:17:11.000000000","tz":540},"subject":"Improve log message for better understanding","message":"Improve log message for better understanding\n\nSometimes designate central could not update the Pool configuration.\nVery limited information is logged into the designate central logs\nthrough which we can not find the root cause of failure.\nThis patch adds some more information into the log for better\nunderstanding.\n\nChange-Id: I26d36c15406436bb3b733d44c490a14e5ece39a6\nCloses-Bug: 1835407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d494e85647e870fc03b02273d79518a39e0d8974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d494e85647e870fc03b02273d79518a39e0d8974"}]},"branch":"refs/heads/master"},"e70af61dac1794ecd858c875c44d6b2ab0d59206":{"kind":"REWORK","_number":4,"created":"2019-07-05 05:33:58.000000000","uploader":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"ref":"refs/changes/54/669154/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/54/669154/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/669154/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/54/669154/4"}}},"commit":{"parents":[{"commit":"d491cd78ff3ef85d1459977571e527e17c8d08bb","subject":"Fix postgres job and make it non-voting for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d491cd78ff3ef85d1459977571e527e17c8d08bb"}]}],"author":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-07-04 11:45:18.000000000","tz":540},"committer":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-07-05 05:32:53.000000000","tz":540},"subject":"Improve log message for better understanding","message":"Improve log message for better understanding\n\nSometimes designate central could not update the Pool configuration.\nVery limited information is logged into the designate central logs\nthrough which we can not find the root cause of failure.\nThis patch adds some more information into the log for better\nunderstanding.\n\nChange-Id: I26d36c15406436bb3b733d44c490a14e5ece39a6\nCloses-Bug: 1835407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e70af61dac1794ecd858c875c44d6b2ab0d59206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e70af61dac1794ecd858c875c44d6b2ab0d59206"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
