)]}'
{"id":"openstack%2Fdesignate~673393","triplet_id":"openstack%2Fdesignate~stable%2Fstein~I26d36c15406436bb3b733d44c490a14e5ece39a6","project":"openstack/designate","branch":"stable/stein","topic":"bug/1835407-stable/stein","hashtags":[],"change_id":"I26d36c15406436bb3b733d44c490a14e5ece39a6","subject":"Improve log message for better understanding","status":"MERGED","created":"2019-07-30 03:18:55.000000000","updated":"2019-08-06 12:39:55.000000000","submitted":"2019-08-06 12:39:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673393-1565095195648-f6c17697","meta_rev_id":"b2d70baec590ac99caad70a685776a932274f3f5","_number":673393,"virtual_id_number":673393,"owner":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-08-06 12:39:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2019-08-06 11:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-30 22:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2019-08-06 11:19:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 22:46:36.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-08-06 11:19:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-08-06 12:39:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc854c85325fbf6e66e092f11c8f071a7023e830","author":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"date":"2019-07-30 03:18:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c178684c8a6bcd50ddd10cd94e0796653664d87a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 04:31:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 http://logs.openstack.org/93/673393/1/check/designate-bind9/669991e/ : SUCCESS in 40m 44s\n- designate-bind9-manager-model http://logs.openstack.org/93/673393/1/check/designate-bind9-manager-model/e3d0102/ : SUCCESS in 39m 14s\n- designate-bind9-py36 http://logs.openstack.org/93/673393/1/check/designate-bind9-py36/151e215/ : SUCCESS in 40m 43s\n- designate-devstack-pdns4 http://logs.openstack.org/93/673393/1/check/designate-devstack-pdns4/6841b64/ : SUCCESS in 31m 29s\n- designate-devstack-pdns4-postgres http://logs.openstack.org/93/673393/1/check/designate-devstack-pdns4-postgres/2f58cdd/ : SUCCESS in 36m 35s\n- designate-grenade-pdns4 http://logs.openstack.org/93/673393/1/check/designate-grenade-pdns4/85edddf/ : SUCCESS in 1h 07m 05s\n- openstack-tox-cover http://logs.openstack.org/93/673393/1/check/openstack-tox-cover/582c89b/cover/ : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/673393/1/check/openstack-tox-lower-constraints/bbedc49/ : SUCCESS in 2m 59s\n- openstack-tox-pep8 http://logs.openstack.org/93/673393/1/check/openstack-tox-pep8/a36a54e/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/93/673393/1/check/openstack-tox-py27/ef4d287/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/93/673393/1/check/openstack-tox-py35/a750cdf/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/93/673393/1/check/openstack-tox-py36/f628acd/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/93/673393/1/check/openstack-tox-py37/bcb5a6d/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/93/673393/1/check/openstack-tox-docs/9ca7ae7/html/ : SUCCESS in 3m 10s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/93/673393/1/check/neutron-tempest-plugin-designate-scenario/f9f554d/ : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":1},{"id":"92f6684269553c74100630f21a76ed3dc0ae9bac","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-07-30 22:46:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a09c9ed8db2a898539d52552e0e26f0982c1a88","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-08-06 11:19:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b954e1915061da119fdfc718b8c56151e67ff031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 11:19:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c0f8fec1946d84081e44591f89e29a05e0ffca9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 12:39:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f126edf2b8e74431d8aaad99e52096fe71b2ce16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 12:39:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://logs.opendev.org/93/673393/1/gate/designate-bind9/db4bbe5/ : SUCCESS in 41m 29s\n- designate-bind9-manager-model https://logs.opendev.org/93/673393/1/gate/designate-bind9-manager-model/b664512/ : SUCCESS in 41m 34s\n- designate-bind9-py36 https://logs.opendev.org/93/673393/1/gate/designate-bind9-py36/2d168b8/ : SUCCESS in 44m 25s\n- designate-devstack-pdns4 https://logs.opendev.org/93/673393/1/gate/designate-devstack-pdns4/5cda0d0/ : SUCCESS in 38m 48s\n- designate-devstack-pdns4-postgres https://logs.opendev.org/93/673393/1/gate/designate-devstack-pdns4-postgres/844ec7b/ : SUCCESS in 43m 53s\n- designate-grenade-pdns4 https://logs.opendev.org/93/673393/1/gate/designate-grenade-pdns4/4b013f1/ : SUCCESS in 1h 17m 02s\n- openstack-tox-lower-constraints https://logs.opendev.org/93/673393/1/gate/openstack-tox-lower-constraints/f06db82/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://logs.opendev.org/93/673393/1/gate/openstack-tox-pep8/2a67a7d/ : SUCCESS in 3m 39s\n- openstack-tox-py27 https://logs.opendev.org/93/673393/1/gate/openstack-tox-py27/a66c39c/ : SUCCESS in 7m 00s\n- openstack-tox-py35 https://logs.opendev.org/93/673393/1/gate/openstack-tox-py35/2ae1435/ : SUCCESS in 6m 03s\n- openstack-tox-py36 https://logs.opendev.org/93/673393/1/gate/openstack-tox-py36/125165d/ : SUCCESS in 5m 36s\n- openstack-tox-py37 https://logs.opendev.org/93/673393/1/gate/openstack-tox-py37/2cb4a66/ : SUCCESS in 5m 51s\n- openstack-tox-docs https://logs.opendev.org/93/673393/1/gate/openstack-tox-docs/4fecf08/html/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"756dcbb4afe937d2de0df60b8ddf3219ebf2dc76","revisions":{"756dcbb4afe937d2de0df60b8ddf3219ebf2dc76":{"kind":"REWORK","_number":1,"created":"2019-07-30 03:18:55.000000000","uploader":{"_account_id":28140,"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","username":"thirose"},"ref":"refs/changes/93/673393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/93/673393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/673393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/673393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/93/673393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/93/673393/1"}}},"commit":{"parents":[{"commit":"82e2d0725d4e55b659fb788d6646f6ebd504fc9c","subject":"Clean up zone locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82e2d0725d4e55b659fb788d6646f6ebd504fc9c"}]}],"author":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-07-04 11:45:18.000000000","tz":540},"committer":{"name":"Takahito Hirose","email":"takahito.hirose0518@gmail.com","date":"2019-07-30 03:18:55.000000000","tz":0},"subject":"Improve log message for better understanding","message":"Improve log message for better understanding\n\nSometimes designate central could not update the Pool configuration.\nVery limited information is logged into the designate central logs\nthrough which we can not find the root cause of failure.\nThis patch adds some more information into the log for better\nunderstanding.\n\nChange-Id: I26d36c15406436bb3b733d44c490a14e5ece39a6\nCloses-Bug: 1835407\n(cherry picked from commit e70af61dac1794ecd858c875c44d6b2ab0d59206)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/756dcbb4afe937d2de0df60b8ddf3219ebf2dc76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/756dcbb4afe937d2de0df60b8ddf3219ebf2dc76"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
