)]}'
{"id":"openstack%2Fdesignate~673975","triplet_id":"openstack%2Fdesignate~master~Ife5af7529d6736231d5518832908bca728c8929d","project":"openstack/designate","branch":"master","topic":"openstackdocstheme","hashtags":[],"change_id":"Ife5af7529d6736231d5518832908bca728c8929d","subject":"Bump the openstackdocstheme extension to 1.20","status":"ABANDONED","created":"2019-08-01 09:25:47.000000000","updated":"2021-12-24 04:26:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3258e7c418ee01fa890d19c693d82e88c59cd75b","_number":673975,"virtual_id_number":673975,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":-1,"date":"2019-09-10 02:12:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 16:45:35.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-09-10 02:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"199fdc00e08305ff853ef7c2f5204e9245742c4a","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-01 09:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaa76e0149a5a9fc87a0ad72ca23e0085c3cb3eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 10:37:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/75/673975/1/check/build-openstack-api-ref/76d3a16/html/ : SUCCESS in 3m 07s\n- designate-grenade-pdns4 https://logs.opendev.org/75/673975/1/check/designate-grenade-pdns4/6e645b2/ : SUCCESS in 1h 05m 20s\n- openstack-tox-pep8 https://logs.opendev.org/75/673975/1/check/openstack-tox-pep8/b94d80d/ : SUCCESS in 4m 02s\n- openstack-tox-docs https://logs.opendev.org/75/673975/1/check/openstack-tox-docs/2a96deb/ : POST_FAILURE in 3m 36s\n- requirements-check https://logs.opendev.org/75/673975/1/check/requirements-check/6563210/ : SUCCESS in 2m 03s\n- build-openstack-releasenotes https://logs.opendev.org/75/673975/1/check/build-openstack-releasenotes/bdf4c82/html/ : SUCCESS in 6m 47s\n- neutron-tempest-plugin-designate-scenario https://logs.opendev.org/75/673975/1/check/neutron-tempest-plugin-designate-scenario/69e5333/ : SUCCESS in 31m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5c139f1a9ae574084fe3d1e1766c2a93aa08e992","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-06 16:45:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1e9d1f8004c706f49fe93b2bbe4a679d9273d2a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 17:46:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/75/673975/1/check/build-openstack-api-ref/a4b217d/html/ : SUCCESS in 3m 46s\n- designate-grenade-pdns4 https://logs.opendev.org/75/673975/1/check/designate-grenade-pdns4/30b8201/ : SUCCESS in 55m 04s\n- openstack-tox-pep8 https://logs.opendev.org/75/673975/1/check/openstack-tox-pep8/ab09bca/ : SUCCESS in 4m 23s\n- openstack-tox-docs https://logs.opendev.org/75/673975/1/check/openstack-tox-docs/10672cd/ : POST_FAILURE in 3m 27s\n- requirements-check https://logs.opendev.org/75/673975/1/check/requirements-check/3c37752/ : SUCCESS in 2m 39s\n- build-openstack-releasenotes https://logs.opendev.org/75/673975/1/check/build-openstack-releasenotes/c6ee7d5/html/ : SUCCESS in 5m 32s\n- neutron-tempest-plugin-designate-scenario https://logs.opendev.org/75/673975/1/check/neutron-tempest-plugin-designate-scenario/2dda661/ : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":1},{"id":"182a2efe0d950eccb619d80b1b5dadcbe0379649","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-06 03:08:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fc58024f2c76944f38c874433e4e9f1cbc8d5245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:58:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3817e50f85484ab8b6e0c323c4440fe3 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f25d62c86804cb4992f91ef3f9ee960 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8e1f0198fa4b8ca0f976ad28b4ec30 : FAILURE in 3m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/04edab549e3b4bc28518707d3c4e4c37 : SUCCESS in 2m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1798f7afb417457c97627e32f844a783 : SUCCESS in 6m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f8ff7972ca0407bb6fdeb0f6b02f881 : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"32956460427461d421f22910c6340b840ce548aa","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-10 01:38:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8af6ea21e6f17bd44ea070cce7888a467a1eb838","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 02:12:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4c520769c5f14d1fb1e79b0a37f75442 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e9c926add64c62b65db9199bea1e45 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2108c960c39487e80a3d0e2ebe1ad85 : FAILURE in 4m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cbb299a27bbd4245a50acee9fb5637c3 : SUCCESS in 2m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c91aad750d9f48e991986e1d4ef8e2bd : SUCCESS in 5m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df88ac1cabb146639fcff7e090d1c8d9 : SUCCESS in 32m 06s","accounts_in_message":[],"_revision_number":2},{"id":"3258e7c418ee01fa890d19c693d82e88c59cd75b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:26:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"25963164313b6e4af78691e34301cdb09091aac3","revisions":{"9cfc2ee6be33b05719a969d309908589085b4865":{"kind":"REWORK","_number":1,"created":"2019-08-01 09:25:47.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/673975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/673975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/673975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/673975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/673975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/673975/1"}}},"commit":{"parents":[{"commit":"076f9fce0ac95b4b628f0d95cfe6e04b7b3988c1","subject":"Merge \"Improve log message for better understanding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/076f9fce0ac95b4b628f0d95cfe6e04b7b3988c1"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-01 09:25:08.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-01 09:25:11.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ife5af7529d6736231d5518832908bca728c8929d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9cfc2ee6be33b05719a969d309908589085b4865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9cfc2ee6be33b05719a969d309908589085b4865"}]},"branch":"refs/heads/master"},"25963164313b6e4af78691e34301cdb09091aac3":{"kind":"REWORK","_number":2,"created":"2019-09-10 01:38:48.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/75/673975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/673975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/673975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/673975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/673975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/673975/2"}}},"commit":{"parents":[{"commit":"9fb598b7d1582d6063913cb97f50863e0a0586c8","subject":"Merge \"Fix bug in recordset status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9fb598b7d1582d6063913cb97f50863e0a0586c8"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-01 09:25:08.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-10 01:38:50.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ife5af7529d6736231d5518832908bca728c8929d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/25963164313b6e4af78691e34301cdb09091aac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/25963164313b6e4af78691e34301cdb09091aac3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
