)]}'
{"id":"openstack%2Fdesignate~675175","triplet_id":"openstack%2Fdesignate~master~I7578414cad7064433701defc2f14bde5cfb6007d","project":"openstack/designate","branch":"master","topic":"pdns3_fix","hashtags":[],"change_id":"I7578414cad7064433701defc2f14bde5cfb6007d","subject":"Use zone.tenant_id when context.project_id is None for pdns3","status":"MERGED","created":"2019-08-07 18:35:21.000000000","updated":"2019-08-09 09:37:02.000000000","submitted":"2019-08-09 09:37:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675175-1565343422574-c9b5ff1c","meta_rev_id":"60ccff7acf6338aaa2481470e8f6c490cdec8edf","_number":675175,"virtual_id_number":675175,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-08-09 09:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-08-08 11:37:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-08-09 08:33:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-08-09 08:33:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 11:37:45.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-08-09 08:33:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-08-09 09:37:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5be5344c1d341c790840afa6e7f65d0fd6fb3e29","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-07 18:35:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1eb5e4f4a452cb2422f9a3e69c688ff0c0a0412","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 19:35:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://logs.opendev.org/75/675175/1/check/designate-bind9/3264c57/ : SUCCESS in 47m 16s\n- designate-pdns4 https://logs.opendev.org/75/675175/1/check/designate-pdns4/62c51e3/ : SUCCESS in 39m 09s\n- designate-bind9-manager-model https://logs.opendev.org/75/675175/1/check/designate-bind9-manager-model/1af9eaa/ : SUCCESS in 41m 24s\n- designate-bind9-py27 https://logs.opendev.org/75/675175/1/check/designate-bind9-py27/92c019c/ : SUCCESS in 42m 10s\n- designate-pdns4-py27 https://logs.opendev.org/75/675175/1/check/designate-pdns4-py27/006a2ae/ : SUCCESS in 41m 14s\n- designate-pdns4-postgres https://logs.opendev.org/75/675175/1/check/designate-pdns4-postgres/d8879d8/ : FAILURE in 34m 52s (non-voting)\n- designate-grenade-pdns4 https://logs.opendev.org/75/675175/1/check/designate-grenade-pdns4/4b1d0bc/ : SUCCESS in 55m 08s\n- openstack-tox-cover https://logs.opendev.org/75/675175/1/check/openstack-tox-cover/4f7a6c9/cover/ : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://logs.opendev.org/75/675175/1/check/openstack-tox-lower-constraints/34be571/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://logs.opendev.org/75/675175/1/check/openstack-tox-pep8/1fd6a39/ : SUCCESS in 4m 36s\n- openstack-tox-py27 https://logs.opendev.org/75/675175/1/check/openstack-tox-py27/766494e/ : SUCCESS in 4m 50s\n- openstack-tox-py36 https://logs.opendev.org/75/675175/1/check/openstack-tox-py36/b119c91/ : SUCCESS in 5m 39s\n- openstack-tox-py37 https://logs.opendev.org/75/675175/1/check/openstack-tox-py37/c8f2118/ : SUCCESS in 7m 18s\n- openstack-tox-docs https://logs.opendev.org/75/675175/1/check/openstack-tox-docs/811fa91/html/ : SUCCESS in 5m 22s\n- neutron-tempest-plugin-designate-scenario https://logs.opendev.org/75/675175/1/check/neutron-tempest-plugin-designate-scenario/40f6110/ : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6194a8008bdeb2bbbbd78a5f782d97d9a8ea9444","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-08 11:37:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01a3fa89bcc024d497000e60c7a3214e43b697a4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-08-09 08:33:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f84e48be5da22e8b70fd567d21519654f163283","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 08:33:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a32f7acbec92f1a5c8d2e8b786fe694e7be0456","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:37:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0309266937463d20faf13b98e608007d593776b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:37:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://logs.opendev.org/75/675175/1/gate/designate-bind9/c3588e3/ : SUCCESS in 40m 35s\n- designate-pdns4 https://logs.opendev.org/75/675175/1/gate/designate-pdns4/167e76e/ : SUCCESS in 42m 07s\n- designate-bind9-manager-model https://logs.opendev.org/75/675175/1/gate/designate-bind9-manager-model/ba41e31/ : SUCCESS in 39m 34s\n- designate-bind9-py27 https://logs.opendev.org/75/675175/1/gate/designate-bind9-py27/3ee1dcd/ : SUCCESS in 44m 36s\n- designate-pdns4-py27 https://logs.opendev.org/75/675175/1/gate/designate-pdns4-py27/5aa85bc/ : SUCCESS in 41m 21s\n- designate-grenade-pdns4 https://logs.opendev.org/75/675175/1/gate/designate-grenade-pdns4/348333a/ : SUCCESS in 58m 03s\n- openstack-tox-lower-constraints https://logs.opendev.org/75/675175/1/gate/openstack-tox-lower-constraints/2decc1c/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://logs.opendev.org/75/675175/1/gate/openstack-tox-pep8/122c67c/ : SUCCESS in 3m 41s\n- openstack-tox-py27 https://logs.opendev.org/75/675175/1/gate/openstack-tox-py27/9f6b57c/ : SUCCESS in 4m 54s\n- openstack-tox-py36 https://logs.opendev.org/75/675175/1/gate/openstack-tox-py36/7f64d3a/ : SUCCESS in 4m 40s\n- openstack-tox-py37 https://logs.opendev.org/75/675175/1/gate/openstack-tox-py37/d8c432f/ : SUCCESS in 5m 42s\n- openstack-tox-docs https://logs.opendev.org/75/675175/1/gate/openstack-tox-docs/7428b48/html/ : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed34f9a30777777db72d85920481c5ab191877df","revisions":{"ed34f9a30777777db72d85920481c5ab191877df":{"kind":"REWORK","_number":1,"created":"2019-08-07 18:35:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/75/675175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/675175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/675175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/675175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/675175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/675175/1"}}},"commit":{"parents":[{"commit":"4ca278f8053cec8e9f2144ba9d991be5d9646691","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ca278f8053cec8e9f2144ba9d991be5d9646691"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-07 18:33:12.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-07 18:33:12.000000000","tz":-420},"subject":"Use zone.tenant_id when context.project_id is None for pdns3","message":"Use zone.tenant_id when context.project_id is None for pdns3\n\nSimilar to Bug 1838379, that most likely affects the pdns3\nimplementation as well.\n\nThe periodic recovery job creates a new context that does\nnot contain the project_id. This commit adds a fallback to the\nzone tenant_id.\n\nChange-Id: I7578414cad7064433701defc2f14bde5cfb6007d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ed34f9a30777777db72d85920481c5ab191877df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ed34f9a30777777db72d85920481c5ab191877df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
