)]}'
{"id":"openstack%2Fdesignate~675911","triplet_id":"openstack%2Fdesignate~master~Iaafec07cf9a0196b5da850924b98e3aa94d559f1","project":"openstack/designate","branch":"master","topic":"drop-extra-endpoints","hashtags":[],"change_id":"Iaafec07cf9a0196b5da850924b98e3aa94d559f1","subject":"Drop creating extra endpoints in devstack","status":"MERGED","created":"2019-08-12 14:27:24.000000000","updated":"2019-08-26 04:58:43.000000000","submitted":"2019-08-26 04:56:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675911-1566795387650-30e5c917","meta_rev_id":"1c6e5a12b477cf04d1dcff605f639fd5389534b2","_number":675911,"virtual_id_number":675911,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-08-26 04:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-08-12 17:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-24 22:46:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-24 22:46:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 17:02:07.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-08-24 22:46:47.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-08-26 04:56:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c589e8780e99b73a496d98286d407d75972f525","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-08-12 14:27:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efe1cfe537c6df152967df2fd689d26af1157ff7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:48:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a5b77ed1be754232ac71c19c83acf2e0 : SUCCESS in 45m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7ae9e5ce61e94614a9cd84fd82bbfdaf : SUCCESS in 45m 02s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/d79dc24fafc04fd99eb69a3c20ee8c92 : SUCCESS in 48m 55s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/55f16af6c0594dc7a420bd445c93bee9 : SUCCESS in 46m 05s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/86eddd34458b4a2684677106ef8cb643 : SUCCESS in 44m 24s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/c5b705ce8a094266b3cc1ed142199225 : FAILURE in 26m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1bc4c2d992514a1fab47d64846b50aa1 : SUCCESS in 1h 14m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b089a1459314fea8e48dbbaaf34f64d : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/393ba93df13a494fb8c30d07b60d12fd : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2f786548d3649bd838a75238ce534f3 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40df4c8d7f7a493fa074f3b31a1efe81 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cb7cba00f8346dca66bedf919d4b029 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b56c6cbfb6bc4e1f8805f5dfb9e89bf8 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c78852ce5e0c46cbbb4fea4aff4fb2cd : SUCCESS in 4m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc8a3499fb764bc5ba83b29baf9d95dc : SUCCESS in 35m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4b2c6c919e08eaeeed14dfe62b26e990b31c3ae7","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-12 17:02:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9abf2c2ec627ffb2952aee7a8d58284dbf7fbae5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-24 22:46:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"39e1d605dd7dbde7388fa1ec82e95e0c96932920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 22:47:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44999adab073ec06ef30eba67d5dc910a028220e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 00:06:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/def63627ef29412581cc87a930b723e2 : SUCCESS in 43m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/88c96310e8f041778596c960afabc503 : SUCCESS in 35m 00s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/70ad7fd2f84b47b7986f2715977e20da : SUCCESS in 42m 48s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/571b1f4d289444cab2940999bd4b5f96 : SUCCESS in 41m 01s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/afc5ea4347614269ae256207ea0f8b9e : SUCCESS in 38m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a21ff8a55c7a4ddd82789384ce790c3f : SUCCESS in 1h 03m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/277d19a394534cdda4b48375896db0eb : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/537cec3d94e342fab71b01433cb0bcaa : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f4507fee908435e870879ad787b3e7f : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fec0e21f9067467c9a4bf1a2fefff3cb : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c8e8d5381f6474db8e3d8d4f239062c : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/271e0c1004b54ce5b2060cb1979a3d1f : FAILURE in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e2ab472e1d064c01d1518fadc282916c8a9b71bf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 22:29:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae32ca9d3334b814cb4dd2fb05d5f6560907f4df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 23:36:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b7315d65365346b199225432afab38f1 : SUCCESS in 39m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1fb3e531eed346d483fb183aff943617 : SUCCESS in 36m 50s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/65fb92a4e1a243d887bfcb5675645994 : SUCCESS in 40m 35s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/ddfb6bd9947240169067013430154ac0 : SUCCESS in 38m 59s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/c3a6dd8b287c42deb3275fe8274480d9 : POST_FAILURE in 57m 19s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/35c3336a529b401eb26d2cb60094509f : FAILURE in 23m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/12dd014c879940979ff8a7af1a242e76 : SUCCESS in 1h 06m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f53926bf6924e4a8e70c7b884878843 : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67375b33670045268689f71ba0e57ece : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b73196c1de4343949204dfc91c3679 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbb72d655ba84dfdb22861e2ab735723 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee8f8df1d082416192debc6ada8a00c1 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/487784fe0d04486eb6b9de2895b01dc1 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bffd54ddc83345398df2cf16fa11d011 : SUCCESS in 5m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/78ad610f211747a885d5e538d0b756f4 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":1},{"id":"082dfa5ae454b09099cf016211f166aa199d672c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 02:44:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"31bf03fb2525cfb18c3172ae4825296b3a3b5691","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 03:51:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b86da8dff649446ea7e854f3b01c5880 : SUCCESS in 41m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e259e3dd801a4885b6748306ef1481b5 : SUCCESS in 38m 47s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/1ff6c919501a44e78b11b90d13cfb323 : SUCCESS in 43m 07s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/fd11496944d5436aaf5b2bbb0ae5480b : SUCCESS in 43m 05s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/725e92a3e92545b2bf3284a83c374359 : SUCCESS in 41m 53s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/e922af71bf32465080327a530ae3d864 : FAILURE in 32m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/279461488ed842c5a35bb20ea3711788 : SUCCESS in 1h 01m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/111e64c0f66f466aaba3e41ad7386ccb : FAILURE in 28m 28s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fe37ebbdb6b24333baa492f672a80dca : FAILURE in 26m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99194c1c9ca745ffb26af5b1fda9de9c : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9ffa02ee6b54bfa867bd9d3433b260e : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a033d9165342169584b8307c48dc1d : SUCCESS in 7m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf316c1080704c3cae203ee8303b2301 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca6dcd4aa94f467eb112f34c06d04a5b : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b81ac3b9b6b497784291539ac2127b1 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394ce0be8210462da89112bd0166feb8 : SUCCESS in 4m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc659ad8cc4e452a8d9868e4982ef2b4 : SUCCESS in 40m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b1f7f72b7ee2a4f22102344e29c43b836f49f445","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 03:51:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3efba9016d159d582b5fd3f9f87dad24065d8fcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 04:56:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/922bbc0949644922ad8fdb9f07905877 : SUCCESS in 41m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b00df39f54384132ad3a2427fc371ac9 : SUCCESS in 36m 45s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/563fdb39165a469c958668045a45f6c6 : SUCCESS in 39m 02s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/a69a4e1a334f4b339a59586521db285a : SUCCESS in 44m 13s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/e54e9add0ceb482fb65b19d59acee028 : SUCCESS in 38m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d24b1e6f5fed4127b3cbce1805b68f2f : SUCCESS in 1h 02m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11e5b75cfb574826a1a9e80e44bd4331 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15b8e569b2464195b54d67ac33c77273 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4934fae4557445f9e6425f2726a71c9 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e6b0033dfa4435cad2474dd1d63c0e4 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bac25a54471f4126b24cbb0a0530b2db : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30eb47533f7948eba7a0ca3bdb27fa59 : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"90caa75e6bc8d4ab3b2e56fbf1f34fbe3c48b700","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 04:56:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5eba94073b2350439290dd0092efa95d0a43e2cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 04:58:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/682ee28fbd5f4d93b29f20e552456ab0 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e23f47d04a1abe4f1fe467c3c143eea0c879bd71","revisions":{"e23f47d04a1abe4f1fe467c3c143eea0c879bd71":{"kind":"REWORK","_number":1,"created":"2019-08-12 14:27:24.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/11/675911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/675911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/675911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/675911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/675911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/675911/1"}}},"commit":{"parents":[{"commit":"4ca278f8053cec8e9f2144ba9d991be5d9646691","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ca278f8053cec8e9f2144ba9d991be5d9646691"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-08-12 14:14:22.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-08-12 14:27:15.000000000","tz":0},"subject":"Drop creating extra endpoints in devstack","message":"Drop creating extra endpoints in devstack\n\nWe don\u0027t need admin or internal endpoints in a normal devstack setup,\nother basic projects are also only creating the public endpoint now.\n\nChange-Id: Iaafec07cf9a0196b5da850924b98e3aa94d559f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e23f47d04a1abe4f1fe467c3c143eea0c879bd71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e23f47d04a1abe4f1fe467c3c143eea0c879bd71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
