)]}'
{"id":"openstack%2Fdesignate~676383","triplet_id":"openstack%2Fdesignate~master~I4598cc184aefe9ab2570b8ef76a76aa5e65e1f6c","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I4598cc184aefe9ab2570b8ef76a76aa5e65e1f6c","subject":"Add region name when calling neutron api","status":"ABANDONED","created":"2019-08-14 10:01:41.000000000","updated":"2021-12-24 04:26:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b1f38501f3f5a449aa71eae2eadb70f359c38729","_number":676383,"virtual_id_number":676383,"owner":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"date":"2019-08-18 15:16:25.000000000","_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":1,"date":"2019-08-14 11:09:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2019-08-23 16:23:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"date":"2019-08-14 11:18:51.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-23 17:33:52.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 11:09:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-14 11:18:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-08-23 16:23:11.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-08-23 17:33:52.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"9f6be4f4208e9b67df2d68c671e3d248bc0acd01","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-08-14 10:01:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d55f914ee5b388b1c24188d1d93b7fbff676c92e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 11:09:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/83bb0c13d17b4955b8016363a39fd8f9 : SUCCESS in 41m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ea031f8440f44a80bc5bf6be9731b40e : SUCCESS in 37m 07s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/6408bc3c494e4162a04d3b62314d87c7 : SUCCESS in 38m 51s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/6db8c510c20d416c80cbb7b0a1f3156e : SUCCESS in 40m 59s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/810e6fa67fae4bd0b6cb2aa23b8c9c3b : SUCCESS in 37m 06s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/932c90c05ca34458b092cf1c3081153e : FAILURE in 28m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/02aad39a31354d879f9321ac43eb5947 : SUCCESS in 57m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd72c462a6224e30984093a8350914b0 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7499fb1ec034ce48bed99544d790749 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01e5edba33240cfb3e3846f1de05ee4 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e46228124ba247a7a3f7f6ef12717b36 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac224820ab4e4e068c3a972f5cd64465 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0560f19d0a24b7d8514926b918cb03a : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a36312fdb6a47a18386d84cef1e2455 : SUCCESS in 3m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6b4c35b4885c492d83fe6fd832ec7fe6 : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f92ef77872c4d4736ce792d14dd8879bcc0779e7","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-15 14:30:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"424a24bf0c630215092512a0566723cf826983b3","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-15 14:30:24.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"d4139a68a1af99ce7f59f89420daedb77f3fc0c1","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-08-18 15:16:25.000000000","message":"Patch Set 1:\n\nHi Graham Hayes, is there any problem? I think it is necessary when there are multiple regions.","accounts_in_message":[],"_revision_number":1},{"id":"883ccb5f02b5eae1b038d0eccfe916b4dc3f264f","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-23 16:23:11.000000000","message":"Patch Set 1:\n\nI don\u0027t think we need the region name, as we are passing in the full endpoint URL. \n\nI don\u0027t have a multi region system setup to test right now, so it may take a while to get it up and running to test","accounts_in_message":[],"_revision_number":1},{"id":"22236725ac328ffef3393ca5403a62151862ccc0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-23 17:33:52.000000000","message":"Patch Set 1:\n\n\u003e I don\u0027t think we need the region name, as we are passing in the\n \u003e full endpoint URL.\n \u003e \n \u003e I don\u0027t have a multi region system setup to test right now, so it\n \u003e may take a while to get it up and running to test\n\nYea - with the endpoint this shouldn\u0027t be needed, but it\u0027s always good practice to pass the region.\n\nI wouldn\u0027t mind Graham testing this code before merging. In addition we should probably rip out the custom endpoint code in base.py and replace it with the regular Keystone client stuff.","accounts_in_message":[],"_revision_number":1},{"id":"b1f38501f3f5a449aa71eae2eadb70f359c38729","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:26:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e048702468143b3b58df267a95718593ab4a6919","revisions":{"e048702468143b3b58df267a95718593ab4a6919":{"kind":"REWORK","_number":1,"created":"2019-08-14 10:01:41.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/83/676383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/83/676383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/676383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/676383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/676383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/83/676383/1"}}},"commit":{"parents":[{"commit":"58260b4c176e6fe058d132b666391bbeacf8cd1d","subject":"Merge \"Fixing 404 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58260b4c176e6fe058d132b666391bbeacf8cd1d"}]}],"author":{"name":"zhanghao2","email":"zhanghao2@cmss.chinamobile.com","date":"2019-07-25 16:55:18.000000000","tz":-240},"committer":{"name":"zhanghao2","email":"zhanghao2@cmss.chinamobile.com","date":"2019-07-25 16:56:53.000000000","tz":-240},"subject":"Add region name when calling neutron api","message":"Add region name when calling neutron api\n\nThis patch can choose the region of the neutron endpoint to use.\n\nChange-Id: I4598cc184aefe9ab2570b8ef76a76aa5e65e1f6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e048702468143b3b58df267a95718593ab4a6919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e048702468143b3b58df267a95718593ab4a6919"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
