)]}'
{"id":"openstack%2Fdesignate~678266","triplet_id":"openstack%2Fdesignate~master~Id0152470274590242a41f6e110d9d63e3daa6f49","project":"openstack/designate","branch":"master","topic":"ipv6","hashtags":[],"change_id":"Id0152470274590242a41f6e110d9d63e3daa6f49","subject":"Add non voting ipv6 only jobs","status":"MERGED","created":"2019-08-23 16:12:28.000000000","updated":"2019-08-26 02:15:53.000000000","submitted":"2019-08-26 02:13:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678266-1566785627221-f069c2db","meta_rev_id":"32c49dd19ec807b99384b8efd8ffaff2be68085e","_number":678266,"virtual_id_number":678266,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-26 02:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-24 22:43:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-26 01:12:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 01:12:51.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-08-26 02:13:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04a025034ffd2ee3feaae9c42b310e49ad86c78f","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-08-23 16:12:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e22ab20e26b509f2ba7150be6c5254a8e274f874","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 17:29:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2ffc382b882343beb41ba9d56c4ec1f8 : SUCCESS in 42m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/80980edbec324518a4deefb447415e11 : SUCCESS in 40m 47s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/3e3121b1ea2848d38b8f448063bc2dc8 : SUCCESS in 40m 40s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/df2735e902e14f22ae570f12753ccb4d : SUCCESS in 48m 51s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/2b6fb0de601a4e818b769070dabbc921 : SUCCESS in 49m 05s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/4a6ee075843943349b48fbfade6a4866 : FAILURE in 30m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7a80b1e3dc25471da6270ae6ce2c1775 : SUCCESS in 1h 03m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/54f8f3c0bc314e628ab8cc32e8edaf16 : FAILURE in 36m 31s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d36219cbb31d48ea8fca9bb39b62a93b : FAILURE in 32m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b11a65f15df4473da6b9be7919dea347 : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48eadfc428714ed8ad37f838933969ca : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea3d181367a418d8e01d0a82f9c02d3 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/615a7e72f66849419de64bd1e7bafd57 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45114ba4a7f74bdeae614fdec9d1e13c : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cee2a28e9ae5422daaa6aa54aa6a55f7 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54286ed89b35462890011bb82b851617 : FAILURE in 3m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ee9159f38454949827bb24254ec0afc : SUCCESS in 35m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d905809ba7bc1cb4872fbef5eee22f08a63dc45a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-23 18:15:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f170c1e87bb85bc98bc21c87eda93207706fb9ed","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-23 18:16:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9d59dc4dd62f27f5f4e8bcb5fa19c59b4c8710b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 19:27:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a2d6edb907e446558656d5ccc7b7508b : SUCCESS in 39m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/116110d43ea74b8fae72512a548cc7db : SUCCESS in 41m 01s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/f0d6dab2688946b691cffdf59591950e : SUCCESS in 38m 30s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/929b32c8d22a443c9d74eec8d1b85d46 : SUCCESS in 49m 16s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/fc057817412249c9babac1c8abf281ea : SUCCESS in 38m 52s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/4cf0188fb6c241e7a1cdd2003b679577 : FAILURE in 27m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/38919192fb2e40acb111d593106ef5f2 : SUCCESS in 1h 07m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/43cc6e7f76664f08a7f0ba3ff4fbe30c : FAILURE in 29m 18s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0719c1de476f41f4a77619df2deae4d1 : FAILURE in 24m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4d9bfb1eccd429c823d010a2b27e727 : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4794936692b457496a2c98d1e0ea257 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa09155d62474649b69e00962e04954e : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2172ddc7d63d42cb8ca148b3f11d105d : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f10c6e2c6d646ba89fbe5231c23d84e : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00d0fe993b5a4ab88dfca3149ae59d8e : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a635622615954ababd0ab72af9c2c418 : FAILURE in 3m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/712d0110795441fd9a838236db815ae7 : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":2},{"id":"64814d3363d76ed7636122cd36def4bf92e6aed9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-23 19:55:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"241b774c6c67aa7decbcc5bdb020f69fed37516e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-23 19:57:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4409c42c2601fdb3a5b9db2bc4e375a15d8495c2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-23 19:58:46.000000000","message":"Patch Set 4: Code-Review+2\n\nCurrently pools.yaml does not work with [] surrounding the ipv6 address. We need to either use escaped values from devstack or strip the brackets out in code. \n\nHaving that said these are non-voting jobs, so lets get them merged.","accounts_in_message":[],"_revision_number":4},{"id":"844e98ff2ee52b96fadc2fb7510cb51b320b4612","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 21:00:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9673aa9da05f4948823dd7518e1c8364 : SUCCESS in 44m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/49afef9c3ba24337b903131a3c51e0ea : SUCCESS in 46m 37s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/500fa26ae6e74968954377a1f8eaad1f : SUCCESS in 48m 42s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/75ab2cf55d434eec84fdf619aab28736 : SUCCESS in 49m 41s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/a2453cc436a047b3803c50bd1adb7e6c : SUCCESS in 47m 02s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/bb25997f13a7493a8c9fcfad89e1bc2c : FAILURE in 29m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ae4b46d1487e4bf5ba8ac3923a26bc1b : SUCCESS in 55m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4db3ac7162444acdb76f7998604e8d41 : FAILURE in 31m 40s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7ce4dd1dc9a44630a48cd9acf3361abe : FAILURE in 32m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46e15705c6d1434a9b0b3691244e52e1 : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a214df1ade145438ea7bcd23e06f6de : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc75cf6d5371489f9fbe050b7608e4a2 : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9dac6ddc54de4517af2dc333f48a92eb : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ded22512c7594d2682274bf74e802566 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20ef06210d1242fd92ba6af1f6faf25b : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f18a9054c1db428f9b250c2ba0c79f31 : FAILURE in 4m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3e2c4d5c7ea14837abe15d81f2a417e8 : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":4},{"id":"8ff2a9113378bf86bb29a0d0f273f344a1e04e3a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-24 01:07:25.000000000","message":"Patch Set 4: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bcbce4bdccf855cc9edbc5f8ca9e4c5ba942c926","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-24 01:20:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d54152ef1255748fa937f533ac38ecaaf029c6d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 02:27:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b3b8a87a5d324f35a336da6c45a6cd8e : SUCCESS in 39m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a69298f7531440b8be46a6f9ddeb0a62 : SUCCESS in 36m 46s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/fcd3709f8d424fc4879c93fd2c2cb877 : SUCCESS in 39m 51s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/96dc1cc87da54a91bbd2756aa4abf141 : SUCCESS in 40m 00s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/3f8c39ea0ed34104a6a24fbeffd4f79e : SUCCESS in 35m 15s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/25a8ea0ec35343b5989fb126d124b6f5 : FAILURE in 26m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/83c1d4e6c8174e9c8e5b0c567bd7b14c : SUCCESS in 56m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1358e03fd41b4d1fb8dd0c6d8ca5b6a7 : FAILURE in 29m 28s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7cc6cb47c549459a9104e112196ddc1f : FAILURE in 27m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1de0c30907974705a4f7de1b2e1d406e : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6244e4422aa243acb9e90a4f4aeabfaf : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aa97402068f4df7b5cd3bb6d8fc77e8 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26d3c5ed9b1041969d8a73f115d5c79a : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/982cbae638d449f38b156eac2f4a1da4 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d8104cd93db4133bf5fee9d10887872 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e691c72c19fe48d6ad14af096f4a8576 : FAILURE in 4m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1f101886d4f04526a00d1eeef446b343 : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":5},{"id":"7b73d3f31d0fd5efb9f8b74390c426c2ada116c3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-24 22:43:55.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b786438d0744609067dda3942c2b24c517ac7e3c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 22:29:19.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b9b4c717904fa246217796a804996c46bfa1ce56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 23:32:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8b5f60ef6c7446ff8cf40faa73bcdf0a : SUCCESS in 1h 02m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8facb1df93e843678c79c48d74034f33 : SUCCESS in 50m 27s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/a05e1ab3eaae4ca49922665b10e495ec : SUCCESS in 44m 58s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/2603c5b9fe3e42e5ae4e2893a13a065a : SUCCESS in 36m 42s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/b1bb7302a2414776ba4e8c1251481e51 : SUCCESS in 50m 10s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/4cc7a8ee87954455b1405f72a2651378 : FAILURE in 27m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c021e3c062da4e799917f842f9c114f5 : SUCCESS in 57m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0c827723872747e8b605577aa23925fd : FAILURE in 31m 19s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5c138a6840ab4ed6b4e67f9f62cf8a7c : FAILURE in 27m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a98d930cca084840881ada661a13d527 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/270b12b9b8cc45ad88f55c8a3ef66ba1 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dc25851d6d24393813288d2499a108b : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02f65406a4034b0eb36b257df2f35bb7 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6bdcb1e9042478eba312cfb2cb1ce0a : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b2b06e49ad348b98ed556f2c71aa095 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d1d55f366704e569f3bd6203f176958 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e254b820575046259f1ba1290e339237 : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":5},{"id":"2f5753f4d657bc6f7ac35d6e649d82cfb1e7458f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 01:12:51.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c1f92748d359f64751e8e60abbb9f2daaaafe957","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 01:13:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e4bc78ac9defb2af224c1f462e600eb6c1a8eaa9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 02:13:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/99c00fa9aff44bc1b4ee1e36ac68e620 : SUCCESS in 38m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b797bc36f4e24fbaa685bb4dd0263c3b : SUCCESS in 33m 14s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/cfa7bc8b38c2413dac4e722466633669 : SUCCESS in 40m 33s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/8cbced947d44407d97a9599486a4825a : SUCCESS in 59m 41s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/f0dc24d189a04b7998dba55dca5b1f92 : SUCCESS in 36m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c0b4de121fbf48ae8073f07503be3d0d : SUCCESS in 56m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef5061be63db4c0da19b233a28912b76 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6f10cd9909d4fd1a5beb56af758edbf : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c57562ca2934accba5113b34b4add9c : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a0eb333f49040689032c47285cdb621 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7749e8f86d2449aacf15832254139be : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef77e2181ec42bca1b95c79a2b10f2b : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":5},{"id":"15df12611b8cf5a2fbc6d48885a697f885781498","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 02:13:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"fd62cf61a6618be881a61402e7e08586cdafe3b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 02:15:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f251d47b9867422fb93211dbeda80c7f : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c640baf7f1db389ed7a467b0e2266fcaa550c2f5","revisions":{"a3296f76e19e62d2451fab5ebcfee886e6ef1895":{"kind":"REWORK","_number":1,"created":"2019-08-23 16:12:28.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/66/678266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/678266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/678266/1"}}},"commit":{"parents":[{"commit":"58260b4c176e6fe058d132b666391bbeacf8cd1d","subject":"Merge \"Fixing 404 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58260b4c176e6fe058d132b666391bbeacf8cd1d"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 16:11:09.000000000","tz":60},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 16:11:09.000000000","tz":60},"subject":"Add non voting ipv6 only jobs","message":"Add non voting ipv6 only jobs\n\nStory: #2005477\nTask: #35887\n\nChange-Id: Id0152470274590242a41f6e110d9d63e3daa6f49\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3296f76e19e62d2451fab5ebcfee886e6ef1895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3296f76e19e62d2451fab5ebcfee886e6ef1895"}]},"branch":"refs/heads/master"},"c86bad90c77501df8aa9b284fcf18c52302982f2":{"kind":"REWORK","_number":2,"created":"2019-08-23 18:15:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/678266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/678266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/678266/2"}}},"commit":{"parents":[{"commit":"58260b4c176e6fe058d132b666391bbeacf8cd1d","subject":"Merge \"Fixing 404 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58260b4c176e6fe058d132b666391bbeacf8cd1d"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 16:11:09.000000000","tz":60},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-23 18:14:40.000000000","tz":-420},"subject":"Add non voting ipv6 only jobs","message":"Add non voting ipv6 only jobs\n\nStory: #2005477\nTask: #35887\n\nChange-Id: Id0152470274590242a41f6e110d9d63e3daa6f49\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c86bad90c77501df8aa9b284fcf18c52302982f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c86bad90c77501df8aa9b284fcf18c52302982f2"}]},"branch":"refs/heads/master"},"fc6d444ce7dd7da4c5cfe40239a73bd3ee07910b":{"kind":"REWORK","_number":3,"created":"2019-08-23 19:55:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/678266/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/678266/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/678266/3"}}},"commit":{"parents":[{"commit":"58260b4c176e6fe058d132b666391bbeacf8cd1d","subject":"Merge \"Fixing 404 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58260b4c176e6fe058d132b666391bbeacf8cd1d"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 16:11:09.000000000","tz":60},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-23 19:55:46.000000000","tz":-420},"subject":"Add non voting ipv6 only jobs","message":"Add non voting ipv6 only jobs\n\nStory: #2005477\nTask: #35887\n\nChange-Id: Id0152470274590242a41f6e110d9d63e3daa6f49\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc6d444ce7dd7da4c5cfe40239a73bd3ee07910b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc6d444ce7dd7da4c5cfe40239a73bd3ee07910b"}]},"branch":"refs/heads/master"},"f16e7333a3f47025b96af0873d6acb192296fef5":{"kind":"REWORK","_number":4,"created":"2019-08-23 19:57:02.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/678266/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/678266/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/678266/4"}}},"commit":{"parents":[{"commit":"58260b4c176e6fe058d132b666391bbeacf8cd1d","subject":"Merge \"Fixing 404 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58260b4c176e6fe058d132b666391bbeacf8cd1d"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 16:11:09.000000000","tz":60},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-23 19:56:56.000000000","tz":-420},"subject":"Add non voting ipv6 only jobs","message":"Add non voting ipv6 only jobs\n\nStory: #2005477\nTask: #35887\n\nChange-Id: Id0152470274590242a41f6e110d9d63e3daa6f49\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f16e7333a3f47025b96af0873d6acb192296fef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f16e7333a3f47025b96af0873d6acb192296fef5"}]},"branch":"refs/heads/master"},"c640baf7f1db389ed7a467b0e2266fcaa550c2f5":{"kind":"REWORK","_number":5,"created":"2019-08-24 01:20:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/66/678266/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/678266/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/678266/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/678266/5"}}},"commit":{"parents":[{"commit":"58260b4c176e6fe058d132b666391bbeacf8cd1d","subject":"Merge \"Fixing 404 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58260b4c176e6fe058d132b666391bbeacf8cd1d"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2019-08-23 16:11:09.000000000","tz":60},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-24 01:20:19.000000000","tz":-420},"subject":"Add non voting ipv6 only jobs","message":"Add non voting ipv6 only jobs\n\nStory: #2005477\nTask: #35887\n\nChange-Id: Id0152470274590242a41f6e110d9d63e3daa6f49\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c640baf7f1db389ed7a467b0e2266fcaa550c2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c640baf7f1db389ed7a467b0e2266fcaa550c2f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
