)]}'
{"id":"openstack%2Fdesignate~678432","triplet_id":"openstack%2Fdesignate~master~I536b92407bf6ca5bddf4c048909cd13d4e094d26","project":"openstack/designate","branch":"master","topic":"new_service_layer","hashtags":[],"change_id":"I536b92407bf6ca5bddf4c048909cd13d4e094d26","subject":"Refactored service layer","status":"MERGED","created":"2019-08-25 18:14:33.000000000","updated":"2019-09-25 18:13:46.000000000","submitted":"2019-09-25 18:11:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678432-1569435091844-d890bd4f","meta_rev_id":"99d05962ffb94d805e5e6768f5397b43196d3995","_number":678432,"virtual_id_number":678432,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 18:11:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2019-09-25 17:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-09-23 18:38:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":2,"date":"2019-09-21 09:06:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2019-09-25 17:03:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-14 12:16:50.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2019-09-21 09:06:12.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-09-23 18:38:24.000000000","updated_by":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"reviewer":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"state":"REVIEWER"},{"updated":"2019-09-25 17:03:57.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-09-25 18:11:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9011c1dcc836f0e1d9527baf8e51fd242251b518","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 18:14:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a688e8083bf66886cc69c2a91f644b1b1b802bd6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 18:41:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"70c7fed236ec4dd3e9e4f64925fdad47da2e9199","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 19:51:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e55bf5ee5521eafc9eb6ea43f4b555da328ed428","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 19:55:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"46b5db62aa2c85cc1fd23e9817cbd088e98f6dac","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 20:09:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"66028a13fa2cde3ccdf1feac8ccd72d393b930fe","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 20:28:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"56b4f4029535af89216a2b6cc3883d72447610e5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 21:07:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cbd1cde22f4385d6b657a7e6e43f0d9f5be2d332","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 21:15:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fd3fdf47670510cfb6b85c60e098d3fae6ce3756","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 22:07:31.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1e308e8766e6497296d4042f2f428e99 : SUCCESS in 42m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9ee033fa76e540099e56712b99704ce8 : SUCCESS in 32m 30s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/290f603fd0cf4beaaada2bb66f26817d : FAILURE in 24m 32s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/e0641ce0aeea4925bc972d66dad962a3 : SUCCESS in 38m 23s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/85dd7242d3fc4383994af665a5f3ca68 : SUCCESS in 31m 56s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/40267f48a43f452d8565c327e43d9cc2 : FAILURE in 18m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f2e8694fde6d466586510248458a14a4 : SUCCESS in 50m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34146751f5794d68bcfaf4b8f1440e26 : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c41b718cfe9f4466b80c352c98c70d6e : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/865c8673ce9546cf865a758f94e5ccdf : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8116ccde52614ae7b3f774ab819af70f : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8098b6fe5c694ba08b6a16df7f004d8e : FAILURE in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/431908fe65f94f6fadbfc0ef0d2a68e7 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68209d6763e416993573bfb3a389088 : SUCCESS in 5m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7e881981036540bc98e2a459a50f78b7 : SUCCESS in 45m 54s","accounts_in_message":[],"_revision_number":8},{"id":"bdb151f40b92554e1a74dedfa8e28ee9949e0ec9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 22:24:14.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d18c289691a28a4d2aa70419923361a190b1b56b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 22:27:55.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"737d07036b288cabd54e9d81f813226d35e744e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 23:44:54.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f2b31935c9be44198c1ec8680535f7ab : SUCCESS in 34m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/11124ce8f93740119086554235c6a8c2 : SUCCESS in 34m 20s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/134ac71728c44386a6620ed2a0bd05c4 : TIMED_OUT in 1h 14m 59s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/3c0c3f6d1d824b8f9f84d49b2eb64e51 : SUCCESS in 45m 00s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/1b1791694fd143fcae28c2532c4ec753 : SUCCESS in 36m 59s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/9f74eb13c26c48f3b6dea366787aef8b : FAILURE in 23m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7a86922a0278458f823628d031065479 : SUCCESS in 54m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee06246534de422fa7b7bf418d027a49 : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9117b874abbc494892d5bde42eb56821 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/648e15b27fd3455aaa7c206bae9aa3f4 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7c94880bcfa4351b765e4bcc6fab9b0 : SUCCESS in 12m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ad5ef322275494d8e57a0fa547c2048 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/510d3acab726415c8f40dbf01753b517 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e172b8cfe634d62bbcb450800c902c0 : SUCCESS in 4m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a02deef9154246eaae17d7a03d6acc90 : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":10},{"id":"1234418ebb46da2c35c5c2efb6d9fdd286ef3015","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-25 23:48:24.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"02155da4bb6117a866337935a565dd76604d3552","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 00:35:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"bdcafc31a5363d8a0b9963030b1bf1cbd128b6b2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 01:45:52.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"65907cb84d8af385bd4ac43b7fb1a64221107ed7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 02:41:07.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"673b95d9be85a1ba4371ff9e2c65662d61e6c25f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 03:20:47.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a6e634495b9581dcf06ef4cfa213c756f6863df6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 04:21:29.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bb86fb17fd274c9cba28339d07f750f2 : SUCCESS in 41m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7ccd9edd0c164b5289ad7db391283209 : SUCCESS in 38m 10s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/5f05eff3fd214be4b15a3485c0f07480 : SUCCESS in 57m 46s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/9c0aaaf0623348b693e20b004f2065ea : SUCCESS in 34m 06s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/0af9d74a4eae4c0c99f63572c4b6565f : SUCCESS in 35m 45s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/f657a514d9624efe9e50bbdbf3b8b8f1 : FAILURE in 29m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1ab966b3a5d44037811ef7106d85dca1 : SUCCESS in 59m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9a948ab2b57e438f889a6746fda8ac8c : FAILURE in 25m 06s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dcd0c669509c485c89562b71cd189954 : FAILURE in 30m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34fa616fa8764d66bae79e8441da46fb : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e44cba6d11274e0293fbc305cef72fb1 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb55a2d97b3649f093b7eb456ed15113 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/144e74e21bc34a4fbe84a52cadd5dd8e : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fd1d459ac63405aba5b98696668d8e4 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3b005b6631d4f048743c8f2f4cbbdaf : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884fef8886ac47d4ad3c6e94aa2d2581 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac188df1681e40af853d6e1e9486beba : SUCCESS in 6m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a16608c3e2344ae7b5c803308e9f432b : SUCCESS in 33m 47s","accounts_in_message":[],"_revision_number":15},{"id":"5197fed1fbb7536519382268a4923d34c06e09ad","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 04:31:16.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"184286f95f3216cca562b7299cc38d5bfe750381","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 04:55:40.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"137a1910a41057fafa2efe58edd5a6537128a2e0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 05:05:43.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"0c0127471d89aa070f792499208c4a2eb7c7a6c1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 05:17:15.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"06ff8c8e0316005c40805ee1ece8f2f5077efb77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 06:22:25.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b3c59d930b1d4356b60c434514f44216 : SUCCESS in 40m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/585a049419f3425d80b39a621108370d : SUCCESS in 34m 21s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/5f25c26d785649c59ff9ea5628507c6b : SUCCESS in 41m 20s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/ef96654aa26147a4bdd40bd56e9277a7 : SUCCESS in 38m 00s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/1f6bd4818ffa4746a26cbe8066a567ea : SUCCESS in 41m 43s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8724356e65f14cf1b5206f5d40424c3d : FAILURE in 24m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/63d9de1430fb470cb7deef5ede2cd34e : SUCCESS in 59m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/73fb9ad5a6624faba3588935764c1ae0 : FAILURE in 24m 24s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/70e3db3335cc45a9b192a49eb66f0d3e : FAILURE in 28m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57ca4da6c42c4e8fba8688babc1fbc5a : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93ea0bc1491644bbaed07259ccf50957 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0618b1fd865473f8ab3bbbe4f81deac : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70f1b2bc85d84a3382c91695c4cdee57 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2567cc16bb44ddfaf4b0ad4df9145ca : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71cfe39795c84ce281c343476ac4d41c : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f70b0e8a5cf047ad9417a6abf0a8da85 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f14dc7bd2f9643bbb6e8a6fb527e9ab9 : SUCCESS in 6m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fae73f5ad3a44f0aa7ffecc1564e8f17 : SUCCESS in 34m 32s","accounts_in_message":[],"_revision_number":19},{"id":"b223b8de59e0e1428bdadebae3afe8544236676c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-30 15:03:03.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"cc18829bf202ecc85cf0a7f6685354a585586e06","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-30 15:52:07.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"4022f220c4a3d0c2b0dce7071b17751ae8d1655a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 17:07:00.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/69a02932071c4e07a378699883685f1e : SUCCESS in 50m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a299e9ea8b30488b9dd9f183066cb5d9 : SUCCESS in 43m 57s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/d930b29ba58e45189a4bf55d57812140 : SUCCESS in 43m 22s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/fc5188c169664a758696d4996ba3eac6 : SUCCESS in 47m 35s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/ea6277e8384240d2bd43f0cbff975519 : SUCCESS in 39m 38s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/27cc1ce5bc48429dbd2a569c5d6d2e21 : FAILURE in 39m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c66bac9cbd2241cb8207371050c0919d : SUCCESS in 1h 08m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/220ed3c9169e42c1bd0d60db6465040b : SUCCESS in 50m 12s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4de8352cccfc474d8a5a1a52c75e4227 : SUCCESS in 50m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe463bf088334cb3923c9cbba5b8cd5d : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f33a7231e2144792a789792a13a6b707 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e1d293c95a4e32a593d9aa9ed5d5b9 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/117402292a2a4c0fa1e66fd67ce1dabd : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd6bd726a47742c6be235c3d64306334 : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e436b73c206646e59764f59b30bdb8d9 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3cc0bc294804f94a57e05a436f80bb1 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e3c67388e714222862c570a612106fe : SUCCESS in 7m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dcf45dff4d454c6f8bd7bb8a135bb8c0 : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":21},{"id":"62f7ec74b394c52be664b78203d431ecd819ee0e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-31 18:52:20.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"a78593754d05b4840f7a416ae28c037434d3c1b5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-31 19:14:19.000000000","message":"Topic changed from new_service to new_service_layer","accounts_in_message":[],"_revision_number":22},{"id":"9cdcf865b1dc07e1ffc059780c92028325cd675d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 19:54:24.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b2a9d163aff349278798927bb4fa2475 : FAILURE in 41m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f4eeb5bfb69e48489327d64a088ee8db : SUCCESS in 37m 05s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/c326ddeca3c04a2381fd730fe938f7d5 : SUCCESS in 42m 04s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/4b13987585694212ad371281ee686e83 : SUCCESS in 43m 01s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/9a138d6602314b29bfec84d13412e221 : SUCCESS in 39m 28s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/a1d65e803e4b4237ba520ffc80176aba : FAILURE in 30m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fe1d227b031e4c798af7ef533775b2d4 : SUCCESS in 59m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f0ea19cbeee947ec965efdc4b513702f : SUCCESS in 40m 19s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/52856c228a2d483e935096391a22a937 : SUCCESS in 40m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/598b43598f284cfca6dfec1dc875d893 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fca58e942244ee9b269208aa65599d9 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8504bd860d8b4218a14ffa1e8d016e55 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c13a27f6edd433e87f056a77bf5434f : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45b24618c7ab4cc390d4dca22584eddd : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb6aa42a68a9430394f6b8b5762bc172 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a94a82af5c24071aed3056a22fd594b : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c312d2f70fb4bdda6e56869d4caa533 : SUCCESS in 7m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/419750dba8304f09b6be2f1e5ba1ab4a : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":22},{"id":"a092d0ec78447572e2b1f6e74c5df66bc464e77a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-31 21:22:57.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"a19d96b2d1a76e15059659c4d2da647346a565b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 22:29:51.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/75b2fc0a5c3e4a288ad21e33f1e401e4 : SUCCESS in 39m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b26b9878c950495cbc87dccbb257c0c4 : SUCCESS in 40m 56s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/538f7a74e05d4c4fbae2b936f12722e2 : SUCCESS in 40m 01s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/df4e3a6a963b48cbb7087f9db36bfad6 : POST_FAILURE in 40m 23s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/acbcb4567fe247f7b904c8530bfeb3a9 : SUCCESS in 35m 59s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/943078bbae754aa196ae97271d4199ce : FAILURE in 31m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/86224ceda6c8405cb78f3c6b29dbfd07 : SUCCESS in 1h 00m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cc83d3422c8448139b3bfe61f87043fd : SUCCESS in 38m 11s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cd7cfab6030644f4b5a89263cb525743 : SUCCESS in 36m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc0f4e6d7d8646f48911801e2eaf9f54 : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6a347a5ef5b42dbbc5abb34e24db63f : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb773a3f17246aca00cf1cf27042f2a : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d788527442a48e6b0b24c434ae5f38c : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddd079e36cd846c0a30023444b6923c9 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57c194fbd30a4eb4a7712ff6aa47b66f : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38275759cc67424685d2c4352386b601 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0d15990f09c480592fc68dcd4c4e0ee : SUCCESS in 7m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/62463ff45ab44744a837aa6a2678d82c : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":23},{"id":"0d5ef8ebaa920f6339e7ea7c68bfae0f8a3fec8d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-31 22:35:36.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"6428e0e490d69fba47265aeadb5490256d5bb48b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 23:45:13.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a7210b144a904266ba66ffc2e594f820 : SUCCESS in 36m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/790447a8dc7d4830a009f2bef86703de : SUCCESS in 46m 07s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/a71d3ad8ef7d451cadd4936e45abeb38 : SUCCESS in 38m 25s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/1d87bb5850414cc2a73e77c10230fc61 : SUCCESS in 50m 45s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/809e9a625183488b93f1716136a727ee : SUCCESS in 34m 53s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/7898d3c393354765af061fb78a7557d0 : FAILURE in 29m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3a3cff694be347279d6a0f3fbb899a9a : SUCCESS in 59m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0978bf0e7776492985153f3597eb9ac3 : SUCCESS in 44m 47s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2b134f8a865d4a90b2ad1605f65729c4 : SUCCESS in 38m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/189908a7924f4138b9536c6a7d416ced : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10a64e972dc342c38c469ea3a5f63c6f : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fec0dcb85f54fe089cd4a8278cd1c06 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e80c68d02cf4dc9ba26a7648e9a7d0f : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fd612f6665e4899808119e577c7f59f : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ecf937f6ed64ecd8ce6b0f6e45c58a4 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a050c5bcde49c483d2ae934dbdc8b3 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b26ba4fce49d4fb6ac5ceac22a4be329 : SUCCESS in 5m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b14fcc2c582d4ba291a7548d6cb96273 : SUCCESS in 45m 00s","accounts_in_message":[],"_revision_number":23},{"id":"9fdb950a2f2dd73ab563cfba4615f31ca1e88dfb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 00:11:09.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"5861626b55e78b3271a16b6544330ea1af23fe45","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 00:26:30.000000000","message":"Topic changed from new_service to new_service_layer","accounts_in_message":[],"_revision_number":24},{"id":"1ca9fe4447c3929f62d095b28a4729126948277a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 01:14:00.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0e479059dbf406091e80b000d6094ad : SUCCESS in 39m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f1e07a2622e147c488dc87b9a0a1c16f : SUCCESS in 42m 18s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/ca0821c8c1a048e8ba9edd5d298abbbe : SUCCESS in 37m 19s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/0f3bd960282747a4b66357b22fa03c7f : SUCCESS in 48m 28s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/4f9dab84d01c4cb48a517b366bc59ca6 : SUCCESS in 36m 47s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/da44b4e5538f4179a2b24c134638395e : FAILURE in 29m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2ab87d597add4a62b67a2bcdb051a085 : SUCCESS in 57m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/366e6937bc5b405cb2cc51edb5db8d38 : SUCCESS in 41m 47s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bef60c9b8352469887b45dc82b6cc0e8 : SUCCESS in 37m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0220bf3ffe844549b2d19dfde63af5a : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1686678df9ad45f5a1b66558015d63eb : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98bb9df488e9439690414543fc015664 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c990fea82a34be5ab07fabf0c38cc2e : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/231d9d23adfe4d6aa0af779bfd1cd384 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f53290c74634d02be15846b5a8a5a29 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf9052e98b4d4257b60fcd58dc786b92 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3874afc15f154b93ad1f93c8fe809505 : SUCCESS in 6m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/46cd5cbc4465448ca99912df77cb7f29 : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":24},{"id":"7e537ea10a146e8250d4b6f936f4d9b042d0e899","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 16:57:42.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"ea6d7b97d45560f312be3c1501d23fbb47d46b56","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 17:43:13.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"bf52a391a466e55557030d08daa43b3b9b5bd3e6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 18:00:30.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"fda62ab2174b78d8e9e7a0aeacdbb9d7ae038efc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 18:03:32.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"a39bd670ce779b41d0f6d89a1fe35b0e73e7df7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 19:09:33.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3989bf01c18644e4b7c0f04d4c59bc2f : SUCCESS in 35m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b66d412f40dd4c6c94d5f545049db334 : SUCCESS in 44m 25s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/b35f13e4a1f2441caa458d3008a72b97 : SUCCESS in 35m 55s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/798eb4c6d9fc499e8751f53d6c5d911d : SUCCESS in 37m 47s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/d1765d1891d84bd68e1b8ef0e2476f5b : SUCCESS in 45m 07s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/391808e073764bacb3bf48561747d447 : FAILURE in 26m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/117fe7b8c70f47e6963b1002de8a9048 : SUCCESS in 59m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d6e1eadc3b64419baac26042aa94241e : SUCCESS in 40m 28s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bbf410ebcbf54e43adf2be57879bf497 : SUCCESS in 40m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95affbab672b4b14983a5a047b21849e : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b3639d3db5b4cca937d19a1f9340ee4 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc24c136d40e4b4ea62b1b417bbae21d : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cafc46d64f41416d86058c558b9dc3e8 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab121510545a4c71ae8b7630237d666a : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0f62304e4114e82b4c0f736b6eb2976 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d2010f43c1042cd90ad7c80599c933d : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99c3a04265eb4a5aba5c034625d9bc01 : SUCCESS in 6m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6f8cc9802e4841398e999680424056f8 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":28},{"id":"991eefbfa35f37a700b4ce63cc4a0aaec95871e5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 21:05:10.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"ff02cb71d41bfd45779fb37a2c0c83ad31f33071","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 21:28:10.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"aa89d42030d2dcc73e9ddea50642ee2aa7dd6a02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 22:32:30.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7bbbc34f81664dcbb75fc5b797198ce4 : SUCCESS in 37m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9c489ca8659140a4b32e9e5d47f6d9ff : SUCCESS in 36m 46s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/c7417255ce6c46c1b8d1bafb049f60bc : SUCCESS in 41m 37s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/885a0dc8647a4a9db3fcf23843fe25a1 : SUCCESS in 54m 49s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/e061503a7a7b4586a34979d18c3544c1 : SUCCESS in 37m 30s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/468464cc43e6445a91765d3aa657d990 : FAILURE in 28m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/90cd9863625b44a6929d6d39b4da2aba : SUCCESS in 58m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bb8698bbac63445eb5b40acaae6a4fba : SUCCESS in 39m 53s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3fab750ed89d4f8999c4d4b3da40288d : SUCCESS in 39m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d4bcc83502949998e48e0f1b914b552 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96d4f616074f4f4e9306e1c0518d9ad0 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/584d83c7f6264ad9b4ab6d9dddd614c2 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb55a12c4ad447febc983607a1db5a8b : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/769451203b39426ba90762d3f18bf0ce : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55459bea120749cc9bb211a7292f2231 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f495fc001bda4f03bc52113b330cab56 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2237b74d95b745d8a7ae40ffee4bf993 : SUCCESS in 6m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3aee8fc1c3434cebae2694ec968f663a : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":30},{"id":"a0b4ed7e6c8aa9345ec316fa25d9edcd70fd0b6c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-02 00:48:19.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"d7d2ba746238d4ed9fec927b4aa76e8056ce6442","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 01:51:17.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b93bcaaddfe54259b76ced9c1635720d : SUCCESS in 39m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/df81c3ee32834afab31cb2df80f47afb : SUCCESS in 41m 29s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/66df14086eee4ed884cd46ed78fb723d : SUCCESS in 41m 06s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/55a988d647c949edbebaa69d94a47cf8 : SUCCESS in 46m 24s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/7ab38f620ff440d1a10f3e6f8e38d627 : SUCCESS in 39m 33s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/91826fd71d744ba096837d753c698f95 : FAILURE in 31m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/330f44caeea047d18c10ae24e927d014 : SUCCESS in 1h 01m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7d781f45a8d845bf85141360ed0e9292 : SUCCESS in 41m 36s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6e6e0c1283224c59af6952b53bca1ea4 : SUCCESS in 45m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99f13e248a3240febff83c895d316248 : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58aa8dd106514e0bbac88cdb256f337c : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380bb314643244bf9179d4411c80ea24 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/284e8cd655c54253b20606cfa49040e3 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29ebf8f28d4846388f7310d0ebad6c81 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c483b96a8534201a2cad144da94f633 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/468ab371751c4990b02676caf5bc707d : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bc8e453735942f9854929f7fb8d7ae2 : SUCCESS in 6m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0612b5cced14b33931f4c72894aeb48 : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":31},{"id":"5a35f84ea5bd528cbf5636f69a30f82276e6428a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-02 02:31:08.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"102e62bc940f778b44ea7eaf32a15c2947707cc3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 03:36:40.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/43d8f9a0909d45d18e11b4f2b017cfa7 : SUCCESS in 39m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b78a00c0c4724b71b126ec9adf5b5483 : SUCCESS in 45m 00s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/4391cb36819649c5b6520bf59df9c5ee : SUCCESS in 40m 25s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/7eea7ec9ccf74ad49be9c22374f8c735 : SUCCESS in 41m 49s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/a410465bedeb4b8594ec117671488a24 : SUCCESS in 36m 01s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/1c295e0920aa452d8911ee083bd5ba44 : FAILURE in 29m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b7645690ae364e1a816187193f4c7ac0 : SUCCESS in 59m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4a495e6d5664412e98d0e36576e8b6f9 : SUCCESS in 37m 23s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6580c3ce6c4542b19687f2c65d219e59 : SUCCESS in 40m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2b726ab09c24b569e1f97930bffe429 : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/869f10d7bdc14662988399716e27f0a9 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b1cc7ddce444238c6190ea4b796f1b : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a182010e0b554861965bad5ca2115c32 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59e5966455bf418694edc3bfa6816eea : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2851b47d1df45e88c9d81ab3875ba74 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/067bb9cee12a4990a18400c2de4ec0b1 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30993b2bad154138af6b5d8996f87535 : SUCCESS in 6m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d59f5f2bcb9141fa8bc2085f5943fb2c : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":32},{"id":"447fb02c776fdb12863d64ae8fe013869d7b27ec","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-02 07:58:18.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"fa3164adf732aa586c32aa1aa3bed3b3f524453d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 09:08:24.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/76965799849940d69e29bb1da963d0f5 : SUCCESS in 39m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f7125948173b4ff1a153e324285cd8e2 : SUCCESS in 37m 58s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/32ca5657ac8048faa05803b4f60cd5a4 : SUCCESS in 40m 51s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/2b1b20f56c1d48d8b86d9d7b26a6b078 : SUCCESS in 42m 31s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/e200c16957d74143a7a76d8d724be436 : SUCCESS in 39m 26s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/323b4829b6f646d39f258cc3795a567b : FAILURE in 29m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fe086b11a52e4ea8b621e98cda369de1 : SUCCESS in 58m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/762ee0166b624544b4a1384feaca8fc1 : SUCCESS in 38m 59s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/916f67c6a13a41729d7899ec82cbeb14 : SUCCESS in 40m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44d6199ae7df4bb6adbf9a2a8b6c0936 : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44b86412a1634075b7cca51945c07a33 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac27d6b53a9c40bdbdbc9fb1c3288547 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e8bbdfce98f4682928c5d739978045a : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29bf5c848a2a4d39983d6fbadad570c1 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/455c5f41d3c940659b489a7129c3cd87 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f4acf5e34984d3790ad9f86d9ac7b1b : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e43e3500e454e0f8d75c2c68ecd6c6a : SUCCESS in 6m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b1d343d864343518ce77050a4041378 : SUCCESS in 34m 49s","accounts_in_message":[],"_revision_number":33},{"id":"802e3bc932c634c68cc5f9136f9023ea7237bd19","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-05 18:28:26.000000000","message":"Patch Set 34: Patch Set 33 was rebased","accounts_in_message":[],"_revision_number":34},{"id":"14c10a64d56a95fe429192fca75ad7f286461a7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 19:31:47.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a2a9d31dc26d45e3ab2d824adcd76ee1 : SUCCESS in 42m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2365fb8a5ca94e969b1ad1a0f4643cee : SUCCESS in 41m 32s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/58f4b937d0194594bd091d092a4edf63 : SUCCESS in 51m 28s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/58a0391546234afea8ba3fefd72a28d5 : SUCCESS in 56m 05s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/ba86d904fe384b8e8c165c9ea8bfba7b : SUCCESS in 42m 13s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8b735baf1ecd43daada06183778c1270 : FAILURE in 30m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bcb2b6ffc45e44a3af68f4430221c99c : SUCCESS in 59m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8c05f5a65e544b76afe9032061bdd75d : SUCCESS in 46m 03s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c49004a5d2cd484486c369dd87e49d00 : SUCCESS in 45m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d12adff94b724d50a58796fc4bf088d6 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b5618de26624cc68d5618eb3e5829ef : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/870d0f9b414b4fe68297515ddb724242 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7827b46bcfb0404091a0f8d27f416378 : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96aa5d8910b04343b6b2bf2e4b72d0fd : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc7e1a745e874928bf5ff81b63dbda0d : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a62d538e7564d6ab1f896a5a1036c05 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23a0348c478b4462bfab5d8116c7d806 : SUCCESS in 6m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb601f5ea813480eac5d24e131c20c05 : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":34},{"id":"b71140c8f80a206fb300c4cfa1b60f7bcea76351","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-09 22:41:57.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"5fdee1d9d6d4be448357ed54dbfa40456de0b97e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 23:52:25.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3eecf3d2b183468fa6fbb16cb7011fa9 : SUCCESS in 43m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0f2751dac34d48caa18c0a4657ece3d7 : SUCCESS in 42m 39s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/94e7d2c0f5ec470aa087e44f9d99a1d0 : SUCCESS in 40m 43s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/a4935c75cad74927903f2984cef20457 : SUCCESS in 44m 28s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/7115fafcaf9a4f2c982cff9bd859d8fa : SUCCESS in 35m 37s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/bd30af45db3b4eb8bf844d432f5dde41 : FAILURE in 44m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f754b85803ef4c29a6ea8bd83c2c296d : SUCCESS in 1h 01m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f5e1f878f00e478ca47f991fcf8795e4 : SUCCESS in 40m 26s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fcf7193f03324b268684202979edf4b1 : SUCCESS in 42m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/845034e09153471187957c4ed86d8b49 : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89dc5834bae84929a356d83c2c07d0b9 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7acfd9df3ba40dabb8819d3eacfbaa1 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d069058e516495689577977dd565858 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2658a572963a441ca30aea2394ba3d8d : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f536ffb27454a1ba2a626cd34a83825 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d0bf31094374d0bb7cc7a6138a22e0b : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2481f9318451485eac148f973bd8bbec : SUCCESS in 5m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0db38ae01c4f43938f735dc2cff94ccc : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":35},{"id":"e3121fc3ccd4dac8209632ba73fe2ee53ffe41bc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-12 17:35:56.000000000","message":"Patch Set 35:\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"70421267251629a48dfa3de946ee5127a9a491e7","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-09-12 22:17:26.000000000","message":"Patch Set 35: Code-Review+1","accounts_in_message":[],"_revision_number":35},{"id":"9b194fad6072bf1d1795dbc3739e2de57a1717eb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-13 20:35:38.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"52e4342d2feeff237ad3009975570d86bd1955b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 21:41:32.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5843aa3728ac41a7b3be42f23eabb3da : SUCCESS in 40m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d88153219954425db097089bda60ecdc : SUCCESS in 46m 12s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/537193e14007454d8ba7418a6008c1ad : SUCCESS in 48m 02s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/7fd128b980564c739db1d95f1075446a : SUCCESS in 52m 21s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/9aad83ecc01648ff91a24be09978cb7e : SUCCESS in 38m 25s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/601592eae6ed4df69aaba32ceed29ee4 : FAILURE in 29m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9001d4ecdcf34128bd6aa91ae71a20f4 : SUCCESS in 1h 00m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/57ec6c9cdf8040fd8cdfca220295ddf5 : SUCCESS in 40m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8a690a78b47e4441b0262a4666bc8432 : SUCCESS in 40m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ac8d60186584377bf74aaadd08d8b42 : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c80e7740e0640f28bdfcff3be4cd9fb : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744850a94df04b2aacd5990679651800 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/664ca771793a4b1286b0b385eb23ae43 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c029145236de4e268d01169e0db07f68 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/010f3664a25945b6bbe336c23e8e8832 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d4f7636074347a7a47ec5576d9dba71 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d672398265dc4f4a968cab71399d4aa1 : SUCCESS in 6m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4864511173284ebe943dcc1a3d51431b : SUCCESS in 39m 07s","accounts_in_message":[],"_revision_number":36},{"id":"8a61ada4e73f3fb335e261a8045f82cf1526d1f4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-18 03:01:07.000000000","message":"Patch Set 37: Patch Set 36 was rebased","accounts_in_message":[],"_revision_number":37},{"id":"4e58933e943acce5e376b0e5ab821f0e5acb2a5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 04:40:06.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dfdde7a4f0b542e18dc45335fd674a91 : SUCCESS in 43m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a36d146c2cdb40a3b6dd7341c14952c9 : SUCCESS in 37m 12s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/462e9e3c57eb46259431761e98f8c1de : SUCCESS in 46m 18s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/8225d696ffa74709bbf42cf39ffce3da : SUCCESS in 39m 08s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/ba8aa7e463194ea29fb772b5fb1eb950 : SUCCESS in 37m 52s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8396270a2ef04028becdb7990b6dfa36 : FAILURE in 31m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/563d83beba0448dcaef1a9450c478d59 : SUCCESS in 1h 04m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/95d9dda8eeec4b82a21caa796a3c27eb : SUCCESS in 39m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/08aaaf486df845d894ac3fbdf87d1153 : SUCCESS in 40m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/770fabb6932c4e88bef327da5db446cb : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e60e57e870b54ee7a7446ecbec029680 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae8184147324cfc9edd4f7deb7e4782 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4cb49879cd34482d8151cec4e91efa23 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f737e388150448d78300096b1ceb20fd : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a300c7ff99e4fc5916bede2f78dc06f : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde721d812384fc2b2a6e02e86119d85 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/371f5cde2d344a269928c7447bcd1b5d : SUCCESS in 6m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/690d597286ab4b21b6e6ddebd9079c0b : SUCCESS in 37m 26s","accounts_in_message":[],"_revision_number":37},{"id":"be8e6ec83fc7925299f2e219d919f96a7a131083","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-09-19 14:58:39.000000000","message":"Patch Set 37:\n\n(5 comments)","accounts_in_message":[],"_revision_number":37},{"id":"fed658dff5e4f55dcc8225e551ef9dd0e7a05b02","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-19 16:32:08.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"867e2986fd3c1f2c502dfc9afa68cc7ffacbe0b8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-19 16:32:40.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"b466d9a6d52e24d5863a731470de8dc1d95296f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:50:33.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6276d393d204d4cb84bddc3a283bdee : SUCCESS in 42m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/941db9c5994d4a319287bce71d134d4f : SUCCESS in 43m 43s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/0f292c0b59f14ac6b384143cf315b8e6 : SUCCESS in 43m 51s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/c68eb6c6df8b40c79af21b7aa30dcddd : SUCCESS in 41m 26s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/b901d2f7284648ddb9218985c4b162db : SUCCESS in 42m 37s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/cea4c1585ad8411db2ab61fb7c1ca09a : FAILURE in 30m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/86c10aba989f4ecca22e16ad31a7806e : SUCCESS in 1h 11m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2f147708c8d042caa899dd8d019d5653 : SUCCESS in 38m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/51928b85851343d18d825dd480d8e21c : SUCCESS in 49m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/119b1f640a80480db915256e6ef8054f : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4b672b4af9742c198b941bb4c3d9bdf : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/756773460e634800995ba49245be9688 : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6932024e77e141408cba0c4102815c0e : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a080229418994e788877bfa564090fcd : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b967be5177347baa9a63d450abb9fe4 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b60f80aa1b5464882b4c3dbb9243d54 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9340d66e547c4fd2a6e90ce26ab6d8d5 : SUCCESS in 7m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c31e7aa48db1482d9044d085c45fb4d4 : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":38},{"id":"190927f18ca03c0bd46380ac702d81b341ac0610","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-20 01:53:14.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"778e9e2a71dc1f52af92df2f283d315a99ca19fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 03:00:18.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a325b0869b34498fa6f2a9060cf0e8d7 : SUCCESS in 45m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d2c4b919f6bf44b2bca62327e0dc5f6a : SUCCESS in 37m 17s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/9725cabbee5f434fb60eeea71de4b866 : SUCCESS in 37m 33s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/069bb1f5ca464b1aae8abb97370f1439 : SUCCESS in 41m 12s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/de217a23e6d94d28b429c0104ea3c575 : SUCCESS in 43m 14s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/3deae646835a46b8a08bc43cbc3bbf02 : FAILURE in 30m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/db86fb5d3aa7492f870bffae4469644f : SUCCESS in 58m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e327cbe0e9dc4c889b430f24b0069143 : SUCCESS in 36m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1c49aab785884c259d23ad50c91c4bee : SUCCESS in 45m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a637e30421cf4e40ab4f056f363083cd : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ed7c78758c44014abc003bf05b4a1bd : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9637b5691a424a54aabf56ead2c9a32c : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ac0f6eb48904433ac57c28622320fb3 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d2c0eecd9a54060b2991cd8b9487cdb : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2347ae5ad5a4fa79eb837952a0bdaa6 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54e90c6d6b94deda8a8ea1a6c0b74bc : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54e96d5406da4d27a70bf98862d0e206 : SUCCESS in 5m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7993bd9e82d54416929ef99904637d47 : SUCCESS in 42m 39s","accounts_in_message":[],"_revision_number":39},{"id":"8ba4899da3ac8adc7781c605181f9c1a5ffa2143","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-20 22:10:59.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"1b94e2026d2a6ad40a0d249e6565f57d8f0cbc27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 23:14:15.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d192d75dd5e945ce8d117def6a3e84cc : SUCCESS in 39m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4f041f7a707d4feeb932cf2555dd4df1 : SUCCESS in 51m 24s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/28480fb1e20d414da4a5d7ae7633508e : SUCCESS in 39m 12s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/a5ec27121a374137b4bda0d269ce5224 : SUCCESS in 42m 01s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/2c44dd7635ef4b658b3958005b2d0cad : SUCCESS in 42m 50s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/9b03ce4fb26c411f9f28bbe0cc3c5658 : FAILURE in 27m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/44aa6fd34c5740f896a4b59f576c154b : SUCCESS in 59m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f9ba6323515b455cad2babc03ef4f247 : SUCCESS in 37m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b6b34f32b3724de9bce2a981301dfd70 : SUCCESS in 44m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01e239838fcf44679e809cf16db7b5a7 : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95ee858db71e4ce8aa8fbbe5604b75d8 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32bafcd057c84d44b59b42c915897652 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b4bb788a305459b844c73379c8e3831 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4713baa0dcc4594b9d3c67ff2b57ba6 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a89741c6bc3487fb32c8b59904dd98d : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca343b70d95e45f8aa24f37a38bfd745 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bb3091e49ee465cb7641e90a8793a2b : SUCCESS in 6m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2c03c251264b402eb7a73536811d09f3 : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":40},{"id":"0a809aba9d42c133e550288c31e173e749d4eccc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-09-21 09:06:12.000000000","message":"Patch Set 40: Code-Review+2","accounts_in_message":[],"_revision_number":40},{"id":"9f3744b14326cf6d9bc23013a452127481f70386","author":{"_account_id":30596,"name":"Scott Chaney","email":"schaney@blizzard.com","username":"schaney"},"date":"2019-09-23 18:38:24.000000000","message":"Patch Set 40: Code-Review+1","accounts_in_message":[],"_revision_number":40},{"id":"b8f86058596fc2dfd409302853acffe81739b8ed","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-09-25 17:03:57.000000000","message":"Patch Set 40: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":40},{"id":"0ed47b5fd95c0c028626a4d090ba0758cfae0e5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 17:04:09.000000000","message":"Patch Set 40: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":40},{"id":"d5fc3229c33a7751fe0bac27666c59e9b31becee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:11:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":40},{"id":"0ba6da3fc2135530a6b04890c6b3763c2c010322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:11:31.000000000","message":"Patch Set 40: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7ed9ba7bb79e4327a8358fde1833abbb : SUCCESS in 42m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c8439d14fbae4a0d9884d8a9cda9effe : SUCCESS in 49m 10s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/900b59fff0d241b3b716fa02dae17c8f : SUCCESS in 42m 28s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/6a19094227544b26820c34adaefab325 : SUCCESS in 45m 30s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/e2d378fba54c4d84912d1319c21c27be : SUCCESS in 41m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f8987cda1aca4ca69fc5c5b828581dc7 : SUCCESS in 1h 01m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6f66a3b903004f8cbb4d028f626b190e : SUCCESS in 37m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6e646327c3ce4f6ebe9ee25989717819 : SUCCESS in 39m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ae2461ca06d453ba3f4c34d93a01ccd : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88daba28668549c6a79a6e78f2d93a80 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9c4bde7fd9b4f7ea3dbbf96265c8fc7 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be9c8f4ea8474a2fa7c67e5e32790c95 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8dc2d81c3fc546e4ad3531eae9117565 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d62998a159f4c17a9fd17ccde29f048 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ce75788633b460189f5a71f940b35e2 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":40},{"id":"99d05962ffb94d805e5e6768f5397b43196d3995","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 18:13:46.000000000","message":"Patch Set 40:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ae2eee1a1b41069c3ba79cbc2e6b88 : SUCCESS in 1m 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4afb53a8d114ef1b2d6902df8a6fb90 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":40}],"current_revision_number":40,"current_revision":"a09064a5d15859703b97d61a1f014681a17799c6","revisions":{"f592151a2d5f2351a0d5861d14f6a2998f3a59f2":{"kind":"REWORK","_number":1,"created":"2019-08-25 18:14:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/1"}}},"commit":{"parents":[{"commit":"58260b4c176e6fe058d132b666391bbeacf8cd1d","subject":"Merge \"Fixing 404 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58260b4c176e6fe058d132b666391bbeacf8cd1d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:14:03.000000000","tz":-420},"subject":"[WIP][DNM] Super WIP testing new service layout","message":"[WIP][DNM] Super WIP testing new service layout\n\nTesting a new service layout based on code from\nnova.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f592151a2d5f2351a0d5861d14f6a2998f3a59f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f592151a2d5f2351a0d5861d14f6a2998f3a59f2"}]},"branch":"refs/heads/master"},"e0e00e8b6f1e566727dd929e08698abef97a60ec":{"kind":"REWORK","_number":2,"created":"2019-08-25 18:41:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/2"}}},"commit":{"parents":[{"commit":"58260b4c176e6fe058d132b666391bbeacf8cd1d","subject":"Merge \"Fixing 404 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58260b4c176e6fe058d132b666391bbeacf8cd1d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:41:13.000000000","tz":-420},"subject":"[WIP][DNM] Super WIP testing new service layout","message":"[WIP][DNM] Super WIP testing new service layout\n\nTesting a new service layout based on code from\nnova.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e0e00e8b6f1e566727dd929e08698abef97a60ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e0e00e8b6f1e566727dd929e08698abef97a60ec"}]},"branch":"refs/heads/master"},"439ab0785aa44b5462351db735cb70ae379ed999":{"kind":"REWORK","_number":3,"created":"2019-08-25 19:51:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/3"}}},"commit":{"parents":[{"commit":"58260b4c176e6fe058d132b666391bbeacf8cd1d","subject":"Merge \"Fixing 404 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58260b4c176e6fe058d132b666391bbeacf8cd1d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 19:51:40.000000000","tz":-420},"subject":"[WIP][DNM] Super WIP testing new service layout","message":"[WIP][DNM] Super WIP testing new service layout\n\nTesting a new service layout based on code from\nnova.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/439ab0785aa44b5462351db735cb70ae379ed999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/439ab0785aa44b5462351db735cb70ae379ed999"}]},"branch":"refs/heads/master"},"b9ab891faddb7f719faff54e4fab07809c861013":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-25 19:55:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/4"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 19:55:14.000000000","tz":0},"subject":"[WIP][DNM] Super WIP testing new service layout","message":"[WIP][DNM] Super WIP testing new service layout\n\nTesting a new service layout based on code from\nnova.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9ab891faddb7f719faff54e4fab07809c861013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9ab891faddb7f719faff54e4fab07809c861013"}]},"branch":"refs/heads/master"},"e7c9c7d4974050d3bd2d912bec02cec291261418":{"kind":"REWORK","_number":5,"created":"2019-08-25 20:09:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/5"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 20:09:20.000000000","tz":-420},"subject":"[WIP][DNM] Super WIP testing new service layout","message":"[WIP][DNM] Super WIP testing new service layout\n\nTesting a new service layout based on code from\nnova.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7c9c7d4974050d3bd2d912bec02cec291261418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7c9c7d4974050d3bd2d912bec02cec291261418"}]},"branch":"refs/heads/master"},"47798bd75f18416b35704b4ff46535867733673a":{"kind":"REWORK","_number":6,"created":"2019-08-25 20:28:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/6"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 20:28:28.000000000","tz":-420},"subject":"[WIP][DNM] Super WIP testing new service layout","message":"[WIP][DNM] Super WIP testing new service layout\n\nTesting a new service layout based on code from\nnova.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47798bd75f18416b35704b4ff46535867733673a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47798bd75f18416b35704b4ff46535867733673a"}]},"branch":"refs/heads/master"},"d6f28afdc4fc58a13c33ab258eeb692036d88c58":{"kind":"REWORK","_number":7,"created":"2019-08-25 21:07:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/7"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 21:05:55.000000000","tz":-420},"subject":"New Service layer","message":"New Service layer\n\nNew service layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. erros on shutdown.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6f28afdc4fc58a13c33ab258eeb692036d88c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6f28afdc4fc58a13c33ab258eeb692036d88c58"}]},"branch":"refs/heads/master"},"dc93d64609ae351039c0d614babb7bd8f17f46b9":{"kind":"REWORK","_number":8,"created":"2019-08-25 21:15:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/8"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 21:15:10.000000000","tz":-420},"subject":"New Service layer","message":"New Service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc93d64609ae351039c0d614babb7bd8f17f46b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc93d64609ae351039c0d614babb7bd8f17f46b9"}]},"branch":"refs/heads/master"},"da8a3bf2d18acfd67c39e45231511235f3af65f0":{"kind":"REWORK","_number":9,"created":"2019-08-25 22:24:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/9"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 22:23:52.000000000","tz":-420},"subject":"New Service layer","message":"New Service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da8a3bf2d18acfd67c39e45231511235f3af65f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da8a3bf2d18acfd67c39e45231511235f3af65f0"}]},"branch":"refs/heads/master"},"62dc817ba9ba56e6471ec8cb9719d7328e87fc60":{"kind":"REWORK","_number":10,"created":"2019-08-25 22:27:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/10"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 22:27:49.000000000","tz":-420},"subject":"New Service layer","message":"New Service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62dc817ba9ba56e6471ec8cb9719d7328e87fc60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62dc817ba9ba56e6471ec8cb9719d7328e87fc60"}]},"branch":"refs/heads/master"},"1ad6bac0c308d7da76a31b016bc979af4dcfb8d6":{"kind":"REWORK","_number":11,"created":"2019-08-25 23:48:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/11"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 23:15:14.000000000","tz":-420},"subject":"New Service layer","message":"New Service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ad6bac0c308d7da76a31b016bc979af4dcfb8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ad6bac0c308d7da76a31b016bc979af4dcfb8d6"}]},"branch":"refs/heads/master"},"06b760fdf580cf07e317caf40ef26d0d645a5a2a":{"kind":"REWORK","_number":12,"created":"2019-08-26 00:35:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/12"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 00:35:23.000000000","tz":-420},"subject":"New Service layer","message":"New Service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b760fdf580cf07e317caf40ef26d0d645a5a2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06b760fdf580cf07e317caf40ef26d0d645a5a2a"}]},"branch":"refs/heads/master"},"449fd113791f4474da3f1128f50e9ed3c8f05114":{"kind":"REWORK","_number":13,"created":"2019-08-26 01:45:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/13"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 01:45:40.000000000","tz":-420},"subject":"New Service layer","message":"New Service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/449fd113791f4474da3f1128f50e9ed3c8f05114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/449fd113791f4474da3f1128f50e9ed3c8f05114"}]},"branch":"refs/heads/master"},"b2ad85346cd46539ef8cda0bb471962229dc46a9":{"kind":"REWORK","_number":14,"created":"2019-08-26 02:41:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/14"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 02:40:57.000000000","tz":-420},"subject":"New Service layer","message":"New Service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b2ad85346cd46539ef8cda0bb471962229dc46a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b2ad85346cd46539ef8cda0bb471962229dc46a9"}]},"branch":"refs/heads/master"},"6c3926cd89e28443dcc06d139b66a42dac994fd4":{"kind":"REWORK","_number":15,"created":"2019-08-26 03:20:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/15"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 03:20:33.000000000","tz":-420},"subject":"New service layer","message":"New service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c3926cd89e28443dcc06d139b66a42dac994fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c3926cd89e28443dcc06d139b66a42dac994fd4"}]},"branch":"refs/heads/master"},"93e47ea8615b4373da0d3bf1593c6723d4ae6e4c":{"kind":"REWORK","_number":16,"created":"2019-08-26 04:31:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/16"}}},"commit":{"parents":[{"commit":"d160f36451cb3eb6fd1ad5f397531450005b3d6c","subject":"Fix missing _static folder breaking gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d160f36451cb3eb6fd1ad5f397531450005b3d6c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 04:31:02.000000000","tz":-420},"subject":"New service layer","message":"New service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93e47ea8615b4373da0d3bf1593c6723d4ae6e4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/93e47ea8615b4373da0d3bf1593c6723d4ae6e4c"}]},"branch":"refs/heads/master"},"4026cb578c3d0c46aedb602f0b9471546c039184":{"kind":"REWORK","_number":17,"created":"2019-08-26 04:55:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/17"}}},"commit":{"parents":[{"commit":"bb5b7761680debfffb45539b1518a2e0d98602a1","subject":"Merge \"Ensure coordination IDs are encoded\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bb5b7761680debfffb45539b1518a2e0d98602a1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 04:55:26.000000000","tz":-420},"subject":"New service layer","message":"New service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4026cb578c3d0c46aedb602f0b9471546c039184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4026cb578c3d0c46aedb602f0b9471546c039184"}]},"branch":"refs/heads/master"},"5a0c2d3d49b6988ebefa7eee25745250fd8c5ffa":{"kind":"REWORK","_number":18,"created":"2019-08-26 05:05:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/18"}}},"commit":{"parents":[{"commit":"bb5b7761680debfffb45539b1518a2e0d98602a1","subject":"Merge \"Ensure coordination IDs are encoded\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bb5b7761680debfffb45539b1518a2e0d98602a1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 05:05:34.000000000","tz":-420},"subject":"New service layer","message":"New service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a0c2d3d49b6988ebefa7eee25745250fd8c5ffa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a0c2d3d49b6988ebefa7eee25745250fd8c5ffa"}]},"branch":"refs/heads/master"},"7c2493a534ea9ad146f660e7097e3e617d1c873d":{"kind":"REWORK","_number":19,"created":"2019-08-26 05:17:15.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/19"}}},"commit":{"parents":[{"commit":"bb5b7761680debfffb45539b1518a2e0d98602a1","subject":"Merge \"Ensure coordination IDs are encoded\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bb5b7761680debfffb45539b1518a2e0d98602a1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 05:16:46.000000000","tz":-420},"subject":"New service layer","message":"New service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2493a534ea9ad146f660e7097e3e617d1c873d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2493a534ea9ad146f660e7097e3e617d1c873d"}]},"branch":"refs/heads/master"},"94406ce2ec6fbd54d98c5e7899b68f574f93cc6f":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-08-30 15:03:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/20"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-30 15:03:03.000000000","tz":0},"subject":"New service layer","message":"New service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\n* Added SSL support to api.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/94406ce2ec6fbd54d98c5e7899b68f574f93cc6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/94406ce2ec6fbd54d98c5e7899b68f574f93cc6f"}]},"branch":"refs/heads/master"},"f84fb472f13791e85a239abefb322bdd9c859b56":{"kind":"REWORK","_number":21,"created":"2019-08-30 15:52:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/21"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-30 15:47:15.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\nThere is unfortunately a bug in the eventlet wsgi\ndriver that prevents ssl support to work with\nPython 3.\nhttps://github.com/eventlet/eventlet/issues/308\n\n* Added SSL support to api.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f84fb472f13791e85a239abefb322bdd9c859b56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f84fb472f13791e85a239abefb322bdd9c859b56"}]},"branch":"refs/heads/master"},"cdebf57e33835b4fcd76b9720e423f3e54acc15d":{"kind":"REWORK","_number":22,"created":"2019-08-31 18:52:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/22"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-31 18:52:09.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\nThere is unfortunately a bug in the eventlet wsgi\ndriver that prevents ssl support to work with\nPython 3.\nhttps://github.com/eventlet/eventlet/issues/308\n\n* Added SSL support to api.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cdebf57e33835b4fcd76b9720e423f3e54acc15d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cdebf57e33835b4fcd76b9720e423f3e54acc15d"}]},"branch":"refs/heads/master"},"7906930b6466f54a5d725ad42af00f6ae2fcf545":{"kind":"REWORK","_number":23,"created":"2019-08-31 21:22:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/23"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-31 21:22:44.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\nThere is unfortunately a bug in the eventlet wsgi\ndriver that prevents ssl support to work with\nPython 3.\nhttps://github.com/eventlet/eventlet/issues/308\n\n* Added SSL support to api.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7906930b6466f54a5d725ad42af00f6ae2fcf545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7906930b6466f54a5d725ad42af00f6ae2fcf545"}]},"branch":"refs/heads/master"},"5e6a1d303905b1c206d148165c06faacbe2ba4c7":{"kind":"REWORK","_number":24,"created":"2019-09-01 00:11:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/24"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 00:10:56.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\nThere is unfortunately a bug in the eventlet wsgi\ndriver that prevents ssl support to work with\nPython 3.\nhttps://github.com/eventlet/eventlet/issues/308\n\n* Added SSL support to api.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e6a1d303905b1c206d148165c06faacbe2ba4c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e6a1d303905b1c206d148165c06faacbe2ba4c7"}]},"branch":"refs/heads/master"},"77bbc245bee98204814f32f50ff4c6ffa5accd1e":{"kind":"REWORK","_number":25,"created":"2019-09-01 16:57:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/25"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 16:57:31.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\nThere is unfortunately a bug in the eventlet wsgi\ndriver that prevents ssl support to work with\nPython 3.\nhttps://github.com/eventlet/eventlet/issues/308\n\n* Added SSL support to api.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77bbc245bee98204814f32f50ff4c6ffa5accd1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77bbc245bee98204814f32f50ff4c6ffa5accd1e"}]},"branch":"refs/heads/master"},"280b972491e8e304d6d011018ecc14e567dc66a4":{"kind":"REWORK","_number":26,"created":"2019-09-01 17:43:13.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/26"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 17:42:42.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\nThere is unfortunately a bug in the eventlet wsgi\ndriver that prevents ssl support to work with\nPython 3.\nhttps://github.com/eventlet/eventlet/issues/308\n\n* Added SSL support to api.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/280b972491e8e304d6d011018ecc14e567dc66a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/280b972491e8e304d6d011018ecc14e567dc66a4"}]},"branch":"refs/heads/master"},"e7d71588905e577653ed6ff3ba6fa717fa9f8678":{"kind":"REWORK","_number":27,"created":"2019-09-01 18:00:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/27"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 17:59:16.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\nThere is unfortunately a bug in the eventlet wsgi\ndriver that prevents ssl support to work with\nPython 3.\nhttps://github.com/eventlet/eventlet/issues/308\n\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Simplified Heartbeat implementation.\n* Removed deprecated options host, port etc.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7d71588905e577653ed6ff3ba6fa717fa9f8678"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7d71588905e577653ed6ff3ba6fa717fa9f8678"}]},"branch":"refs/heads/master"},"c36f52af5e406ace46e0b0b8d49fb360993b3932":{"kind":"REWORK","_number":28,"created":"2019-09-01 18:03:32.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/28"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 18:03:23.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer based on the nova implementation\nthat removes some abstraction, enables ssl support\nand makes it easier to test code.\n\nThere is unfortunately a bug in the eventlet wsgi\ndriver that prevents ssl support to work with\nPython 3.\nhttps://github.com/eventlet/eventlet/issues/308\n\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Simplified Heartbeat implementation.\n* Removed deprecated options host, port etc.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c36f52af5e406ace46e0b0b8d49fb360993b3932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c36f52af5e406ace46e0b0b8d49fb360993b3932"}]},"branch":"refs/heads/master"},"f15f588cd1a514454bba107a35396b02b62ff80e":{"kind":"REWORK","_number":29,"created":"2019-09-01 21:05:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/29"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 21:04:37.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by remvoing some\nabstraction and making the implementation more in line\nwith other OpenStack projects. We also now make use\nof the oslo_service.wsgi implementation.\n\nIn addition we also moved some of the heartbeat\ncode out of the service implementations. This is\nprimarily because we only need one instance of the\nheartbeat to run on the main process.\nIn the old design each fork would receive it\u0027s\nown instance of the heartbeat implementation.\n\n* Added Heartbeat to designate-sink.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f15f588cd1a514454bba107a35396b02b62ff80e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f15f588cd1a514454bba107a35396b02b62ff80e"}]},"branch":"refs/heads/master"},"864ece0ccd62bc74dba5ddd07f2b997f6840f9ea":{"kind":"REWORK","_number":30,"created":"2019-09-01 21:28:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/30"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 21:28:01.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by remvoing some\nabstraction and making the implementation more in line\nwith other OpenStack projects. We also now make use\nof the oslo_service.wsgi implementation.\n\nIn addition we also moved some of the heartbeat\ncode out of the service implementations. This is\nprimarily because we only need one instance of the\nheartbeat to run on the main process.\nIn the old design each fork would receive it\u0027s\nown instance of the heartbeat implementation.\n\n* Added Heartbeat to designate-sink.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/864ece0ccd62bc74dba5ddd07f2b997f6840f9ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/864ece0ccd62bc74dba5ddd07f2b997f6840f9ea"}]},"branch":"refs/heads/master"},"5b9f2012834067e7d9027c37e9bb93f1be1e3de5":{"kind":"REWORK","_number":31,"created":"2019-09-02 00:48:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/31"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-02 00:33:57.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by removing abstraction\nand making the implementation more in line with other\nOpenStack projects.\n\nMoved Heartbeat code out of Service class and\ninto the console scripts. We only need one instance\nof the Heartbeat Emitter.\n\nCleaned up the WSGI code by making use of the\nreusable oslo_service.wsgi code.\n\n* Added Heartbeat to designate-sink.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b9f2012834067e7d9027c37e9bb93f1be1e3de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b9f2012834067e7d9027c37e9bb93f1be1e3de5"}]},"branch":"refs/heads/master"},"98dc2aac385771433be7e54ae4d334717fd6aedd":{"kind":"REWORK","_number":32,"created":"2019-09-02 02:31:08.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/32"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-02 02:30:55.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by removing abstraction\nand making the implementation more in line with other\nOpenStack projects.\n\nMoved Heartbeat code out of Service class and\ninto the console scripts. We only need one instance\nof the Heartbeat Emitter.\n\nCleaned up the WSGI code by making use of the\nreusable oslo_service.wsgi code.\n\n* Added Heartbeat to designate-sink.\n* Api logs no longer showing up as eventlet.wsgi.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98dc2aac385771433be7e54ae4d334717fd6aedd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98dc2aac385771433be7e54ae4d334717fd6aedd"}]},"branch":"refs/heads/master"},"0ff1c86de92a03eef66714534f81b42d7256c2b2":{"kind":"REWORK","_number":33,"created":"2019-09-02 07:58:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/33"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-02 07:57:58.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by removing abstraction\nand making the implementation more in line with other\nOpenStack projects.\n\nMoved Heartbeat code out of Service class and\ninto the console scripts. We only need one instance\nof the Heartbeat Emitter.\n\nCleaned up the WSGI code by making use of the\nreusable oslo_service.wsgi code.\n\n* Added Heartbeat to designate-sink.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ff1c86de92a03eef66714534f81b42d7256c2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ff1c86de92a03eef66714534f81b42d7256c2b2"}]},"branch":"refs/heads/master"},"f376d5b93419c452d76c8516d37f78d30c3d39d9":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2019-09-05 18:28:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/34"}}},"commit":{"parents":[{"commit":"3f38b24ed4e010455d356a757d67e6a316890f10","subject":"Merge \"Updated install guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f38b24ed4e010455d356a757d67e6a316890f10"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-05 18:28:26.000000000","tz":0},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by removing abstraction\nand making the implementation more in line with other\nOpenStack projects.\n\nMoved Heartbeat code out of Service class and\ninto the console scripts. We only need one instance\nof the Heartbeat Emitter.\n\nCleaned up the WSGI code by making use of the\nreusable oslo_service.wsgi code.\n\n* Added Heartbeat to designate-sink.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f376d5b93419c452d76c8516d37f78d30c3d39d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f376d5b93419c452d76c8516d37f78d30c3d39d9"}]},"branch":"refs/heads/master"},"e183044a1baee8fe310ad2cbaf02c5ba4a1e5e83":{"kind":"REWORK","_number":35,"created":"2019-09-09 22:41:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/35"}}},"commit":{"parents":[{"commit":"3f38b24ed4e010455d356a757d67e6a316890f10","subject":"Merge \"Updated install guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f38b24ed4e010455d356a757d67e6a316890f10"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-09 22:41:39.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by removing abstraction\nand making the implementation more in line with other\nOpenStack projects.\n\nMoved Heartbeat code out of Service class and\ninto the console scripts. We only need one instance\nof the Heartbeat Emitter.\n\nCleaned up the WSGI code by making use of the\nreusable oslo_service.wsgi code.\n\n* Added Heartbeat to designate-sink.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e183044a1baee8fe310ad2cbaf02c5ba4a1e5e83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e183044a1baee8fe310ad2cbaf02c5ba4a1e5e83"}]},"branch":"refs/heads/master"},"6d8cd21580a5d5edaf0958851683baa09a175289":{"kind":"REWORK","_number":36,"created":"2019-09-13 20:35:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/36"}}},"commit":{"parents":[{"commit":"3f38b24ed4e010455d356a757d67e6a316890f10","subject":"Merge \"Updated install guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3f38b24ed4e010455d356a757d67e6a316890f10"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-13 20:34:19.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by removing abstraction\nand making the implementation more in line with other\nOpenStack projects.\n\nMoved Heartbeat code out of Service class and\ninto the console scripts. We only need one instance\nof the Heartbeat Emitter.\n\nCleaned up the WSGI code by making use of the\nreusable oslo_service.wsgi code.\n\n* Added Heartbeat to designate-sink.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d8cd21580a5d5edaf0958851683baa09a175289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d8cd21580a5d5edaf0958851683baa09a175289"}]},"branch":"refs/heads/master"},"6da97079513a69802e08fa192dd8907cda9b2a3d":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2019-09-18 03:01:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/37"}}},"commit":{"parents":[{"commit":"23f6a79aef1303d25f8eb1c9d5090fe7f4326d46","subject":"Merge \"Add missing RabbitMQ configuration to install guides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23f6a79aef1303d25f8eb1c9d5090fe7f4326d46"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-18 03:01:07.000000000","tz":0},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by removing abstraction\nand making the implementation more in line with other\nOpenStack projects.\n\nMoved Heartbeat code out of Service class and\ninto the console scripts. We only need one instance\nof the Heartbeat Emitter.\n\nCleaned up the WSGI code by making use of the\nreusable oslo_service.wsgi code.\n\n* Added Heartbeat to designate-sink.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6da97079513a69802e08fa192dd8907cda9b2a3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6da97079513a69802e08fa192dd8907cda9b2a3d"}]},"branch":"refs/heads/master"},"777f13cf5d942668e39fc7f5cefffe3eb2e583ba":{"kind":"REWORK","_number":38,"created":"2019-09-19 16:32:40.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/38"}}},"commit":{"parents":[{"commit":"23f6a79aef1303d25f8eb1c9d5090fe7f4326d46","subject":"Merge \"Add missing RabbitMQ configuration to install guides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23f6a79aef1303d25f8eb1c9d5090fe7f4326d46"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-19 16:32:23.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by removing abstraction\nand making the implementation more in line with other\nOpenStack projects.\n\nMoved Heartbeat code out of Service class and\ninto the console scripts. We only need one instance\nof the Heartbeat Emitter.\n\nCleaned up the WSGI code by making use of the\nreusable oslo_service.wsgi code.\n\n* Added Heartbeat to designate-sink.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/777f13cf5d942668e39fc7f5cefffe3eb2e583ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/777f13cf5d942668e39fc7f5cefffe3eb2e583ba"}]},"branch":"refs/heads/master"},"722cc2245336bcf71ae0e7123b1cd2902a854669":{"kind":"REWORK","_number":39,"created":"2019-09-20 01:53:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/39"}}},"commit":{"parents":[{"commit":"23f6a79aef1303d25f8eb1c9d5090fe7f4326d46","subject":"Merge \"Add missing RabbitMQ configuration to install guides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23f6a79aef1303d25f8eb1c9d5090fe7f4326d46"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-20 01:53:04.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by removing abstraction\nand making the implementation more in line with other\nOpenStack projects.\n\nMoved Heartbeat code out of Service class and\ninto the console scripts. We only need one instance\nof the Heartbeat Emitter.\n\nCleaned up the WSGI code by making use of the\nreusable oslo_service.wsgi code.\n\n* Added Heartbeat to designate-sink.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/722cc2245336bcf71ae0e7123b1cd2902a854669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/722cc2245336bcf71ae0e7123b1cd2902a854669"}]},"branch":"refs/heads/master"},"a09064a5d15859703b97d61a1f014681a17799c6":{"kind":"REWORK","_number":40,"created":"2019-09-20 22:10:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/32/678432/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/678432/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/678432/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/678432/40"}}},"commit":{"parents":[{"commit":"23f6a79aef1303d25f8eb1c9d5090fe7f4326d46","subject":"Merge \"Add missing RabbitMQ configuration to install guides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23f6a79aef1303d25f8eb1c9d5090fe7f4326d46"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-25 18:13:32.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-20 22:10:46.000000000","tz":-420},"subject":"Refactored service layer","message":"Refactored service layer\n\nService layer has been simplified by removing abstraction\nand making the implementation more in line with other\nOpenStack projects.\n\nMoved Heartbeat code out of Service class and\ninto the console scripts. We only need one instance\nof the Heartbeat Emitter.\n\nCleaned up the WSGI code by making use of the\nreusable oslo_service.wsgi code.\n\n* Added Heartbeat to designate-sink.\n* Cleaned up and refactored Service layers.\n* Fixed various bugs e.g. errors on shutdown.\n* Removed deprecated options host, port etc.\n* Simplified Heartbeat implementation.\n\nCloses-Bug: #1442141\nChange-Id: I536b92407bf6ca5bddf4c048909cd13d4e094d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a09064a5d15859703b97d61a1f014681a17799c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a09064a5d15859703b97d61a1f014681a17799c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
