)]}'
{"id":"openstack%2Fdesignate~678543","triplet_id":"openstack%2Fdesignate~master~If9a487b9fb20eadcfb9ba62267bff69476e863db","project":"openstack/designate","branch":"master","topic":"test","hashtags":[],"change_id":"If9a487b9fb20eadcfb9ba62267bff69476e863db","subject":"Testing","status":"ABANDONED","created":"2019-08-26 13:02:57.000000000","updated":"2019-08-28 21:00:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"620ab73bde2dc5b4104eac19caec5870a5fa5020","_number":678543,"virtual_id_number":678543,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-08-26 18:02:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 18:02:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc6a34c65753fc213ee81f779bec72b75ff19e6f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 13:02:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79ed56acf2e0e8ca4189a558500b7cf6b7679742","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 13:39:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"91138f8c6aa01572068b188c84b76c56a8eec331","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 14:35:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5d110dd1875540746ed2408d75f3f2111bc3f4c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 15:43:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bdcc65835e334fc3894d29d593b53e0f : FAILURE in 33m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a416b2d2834e40e99b3b22e261ef574d : FAILURE in 34m 16s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/8dfe7514ceb7484aacc2353e1224d8e3 : FAILURE in 37m 11s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/7f4a091f6bdc4522abdbf3e5277dd6fa : FAILURE in 33m 17s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/2b5d97040091444fa15123883008ae7f : FAILURE in 35m 51s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/c3662a3bca32436f83c80c9aab627eea : FAILURE in 26m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cfae868469f74af38e160c8412af5df6 : FAILURE in 1h 00m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/36ee502d7505479fa706cd7bcb70d82e : FAILURE in 31m 07s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/709c89420c7c4c449ec3732502afebb3 : FAILURE in 31m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e37f98e332f64562846ecb37d17f6426 : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f96b0e038cd54d8991ac2347cba1ef2b : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9212feff37d04b728b15317ef904bf8c : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/849b9b6348b246f7b092f6f1933bf502 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f172c5826b3e4c788d63d11a0b5305bd : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3dcd4c03a6af427eb017fce05dd8eaca : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c1d5a8c3f84dc881d5c711317e77d4 : SUCCESS in 5m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe68fb22c23c43b78830f2e3da2dfb1f : FAILURE in 31m 02s","accounts_in_message":[],"_revision_number":3},{"id":"14498a0bcfc14e77b96baebcfbf6ad0c7ec5ed5e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 16:17:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"93e9f3e8f267ab00ced9de91ea37b9480757c88a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-26 17:02:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9c7c3e01733ca448361a36be4f55dbaf4509caaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 18:02:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dc180ac090b94b8b8134760ff785bc6c : FAILURE in 35m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/401c60bd3a9c4a8a95a7afad80e2a69e : FAILURE in 42m 38s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/7dc3e5800b524f4eb6e7eff43917cc74 : FAILURE in 37m 09s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/cd6e5951ccbb4c5287b1212dee74effa : SUCCESS in 39m 54s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/63f6a1d81d504966b3bcf1bb567e9a52 : SUCCESS in 38m 49s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8155262b86ac478284dcb54187787d5e : FAILURE in 34m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/642d7ac6a23247eb87cebc8dc0bff81c : FAILURE in 55m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d8ee72e1b61b4b859b38159abb8e6f38 : FAILURE in 29m 38s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9529ce605e44487d95b80afe21db0a0c : FAILURE in 36m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e07d5ab9fec44df2956fe070e19aa664 : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd654294779d4b11b9a3e6dd385c17ec : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ec635b8eefa410db1c0046a150c7239 : SUCCESS in 7m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e23f5862864e4b9d8b6036d51b0ecb6a : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6b533d5cce94c62bca8d256fb75d67c : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2444ef184d0a4cee803264adbb0f4af5 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b748fb9c4b43dba692fc39796485df : SUCCESS in 7m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4a901110147c4672bb8681142d7ec5bf : FAILURE in 28m 55s","accounts_in_message":[],"_revision_number":5},{"id":"5399ab34b16bc31f58242a1b26fd1f00cc51a91f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-08-28 21:00:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2c15349e97d2f717e3a49f6d0224abf637328c76","revisions":{"f32d7d6d670730cb9f46a6143e33156431636424":{"kind":"REWORK","_number":1,"created":"2019-08-26 13:02:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/678543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/678543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/678543/1"}}},"commit":{"parents":[{"commit":"7c2493a534ea9ad146f660e7097e3e617d1c873d","subject":"New service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2493a534ea9ad146f660e7097e3e617d1c873d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 13:02:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 13:02:39.000000000","tz":-420},"subject":"Testing","message":"Testing\n\nChange-Id: If9a487b9fb20eadcfb9ba62267bff69476e863db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f32d7d6d670730cb9f46a6143e33156431636424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f32d7d6d670730cb9f46a6143e33156431636424"}]},"branch":"refs/heads/master"},"3e28fbf68cac4b3a7710fb719f84100f0ab8534b":{"kind":"REWORK","_number":2,"created":"2019-08-26 13:39:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/678543/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/678543/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/678543/2"}}},"commit":{"parents":[{"commit":"7c2493a534ea9ad146f660e7097e3e617d1c873d","subject":"New service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2493a534ea9ad146f660e7097e3e617d1c873d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 13:02:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 13:38:55.000000000","tz":-420},"subject":"Testing","message":"Testing\n\nChange-Id: If9a487b9fb20eadcfb9ba62267bff69476e863db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3e28fbf68cac4b3a7710fb719f84100f0ab8534b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3e28fbf68cac4b3a7710fb719f84100f0ab8534b"}]},"branch":"refs/heads/master"},"2f7ad7fd1a552a90e9dc2811b9dab4df72f66f90":{"kind":"REWORK","_number":3,"created":"2019-08-26 14:35:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/678543/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/678543/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/678543/3"}}},"commit":{"parents":[{"commit":"7c2493a534ea9ad146f660e7097e3e617d1c873d","subject":"New service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2493a534ea9ad146f660e7097e3e617d1c873d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 13:02:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 14:35:25.000000000","tz":-420},"subject":"Testing","message":"Testing\n\nChange-Id: If9a487b9fb20eadcfb9ba62267bff69476e863db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f7ad7fd1a552a90e9dc2811b9dab4df72f66f90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f7ad7fd1a552a90e9dc2811b9dab4df72f66f90"}]},"branch":"refs/heads/master"},"7149614722bad641e94899fc2f5384c808dd7ede":{"kind":"REWORK","_number":4,"created":"2019-08-26 16:17:13.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/678543/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/678543/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/678543/4"}}},"commit":{"parents":[{"commit":"7c2493a534ea9ad146f660e7097e3e617d1c873d","subject":"New service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2493a534ea9ad146f660e7097e3e617d1c873d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 13:02:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 16:16:57.000000000","tz":-420},"subject":"Testing","message":"Testing\n\nChange-Id: If9a487b9fb20eadcfb9ba62267bff69476e863db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7149614722bad641e94899fc2f5384c808dd7ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7149614722bad641e94899fc2f5384c808dd7ede"}]},"branch":"refs/heads/master"},"2c15349e97d2f717e3a49f6d0224abf637328c76":{"kind":"REWORK","_number":5,"created":"2019-08-26 17:02:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/678543/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/678543/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/678543/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/678543/5"}}},"commit":{"parents":[{"commit":"7c2493a534ea9ad146f660e7097e3e617d1c873d","subject":"New service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2493a534ea9ad146f660e7097e3e617d1c873d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 13:02:39.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-08-26 17:02:23.000000000","tz":-420},"subject":"Testing","message":"Testing\n\nChange-Id: If9a487b9fb20eadcfb9ba62267bff69476e863db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c15349e97d2f717e3a49f6d0224abf637328c76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c15349e97d2f717e3a49f6d0224abf637328c76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
