)]}'
{"id":"openstack%2Fdesignate~678952","triplet_id":"openstack%2Fdesignate~master~Ie454f8f9ebc8e36347ce33b22f2e6da0257fe4e0","project":"openstack/designate","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"Ie454f8f9ebc8e36347ce33b22f2e6da0257fe4e0","subject":"Add pdf-docs env to tox.ini","status":"MERGED","created":"2019-08-27 19:54:32.000000000","updated":"2019-09-09 12:32:38.000000000","submitted":"2019-09-09 12:31:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678952-1568032268551-72f0986f","meta_rev_id":"69498a0e77b834a06ff9c8057998db49c8d35bd1","_number":678952,"virtual_id_number":678952,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-07 13:08:33.000000000","_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-09-09 12:31:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2019-09-09 11:19:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":1,"date":"2019-09-07 09:27:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-08 03:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2019-09-09 11:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 19:48:40.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-06 21:16:15.000000000","updated_by":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"reviewer":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"state":"REVIEWER"},{"updated":"2019-09-07 09:27:22.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-09-08 03:43:24.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-09-09 11:19:12.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-09-09 12:31:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"708ea1293a2defea2242181a8f9e67c955de9a25","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-08-27 19:54:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fbf63fda9c6f15ac4db7d33ec5ba1d6dfdd5ade","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-08-27 19:55:15.000000000","message":"Patch Set 1: Code-Review-1\n\nStill work in progress","accounts_in_message":[],"_revision_number":1},{"id":"1a825b9f825785448a30ad3adfb48798dee4c659","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 21:09:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/89b6fc6bb03b4d5cb40f7abc1b444b5e : SUCCESS in 5m 17s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/693ffe49699e4eada23a0e42cd5153ae : SUCCESS in 43m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/af8423329b2c4bdfb96187685a8d5c2f : SUCCESS in 40m 13s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/9f7f3596f209418db4be26f37c350f9d : SUCCESS in 45m 45s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/06d91bbce1c842a180685b29a671cfb1 : SUCCESS in 46m 10s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/4f3b6ed6feee404f8fd2d47155313fff : SUCCESS in 44m 11s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/43d5cc842cdc4e6e8115eddf62414538 : FAILURE in 25m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a62661fab9eb461ab9107a2405fd9a75 : SUCCESS in 1h 09m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e96b09a2f1d9449784ee4d261f491a2d : FAILURE in 29m 14s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b533fbb5e0cb46d4b337508c61d312ee : FAILURE in 34m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/549be26e29694ce4b1245a537778cf2b : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/669bb5ec12204295af9209cade059038 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d40252d58f141d2a0c70af36cbd842d : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afa196079f9d4838a36b6ca07f72b697 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb33bd37d0864780ac7aabee6f89e85d : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bad50e267774db89369b7f665523457 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6296298618534dfd8c499efa5cbea4e0 : FAILURE in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/052042f037d24c8d8eb8467e9cf34244 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d6abbbbb1080d71a7adb77e1356aa9115924302c","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-08-28 22:24:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f12238b658d932aad74084ea37ef121702d98f84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 23:30:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b03cdfcdda9f441490cbff4df494a736 : SUCCESS in 3m 50s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8ed99ca9dcc4b528554706143e3275b : SUCCESS in 47m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1159764ce0614328a388c2bd5e43eb3b : SUCCESS in 35m 03s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/060dc835aa2e4849aaa9e8aeaba03b4c : SUCCESS in 44m 11s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/4f70dd23a4c44ea89564f5e360340324 : SUCCESS in 47m 09s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/3ab18eb1652b464da430087240bc6406 : SUCCESS in 44m 26s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/d8cd4fc09ef644679b3a705a975e9284 : FAILURE in 26m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4aa87ab7c9c1489e903a1d735e0d0eb1 : SUCCESS in 1h 00m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/83dae0375f2a408dabeb1e27c8846751 : FAILURE in 23m 57s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/362c65ae17a2416891a5c6f5e1db9c72 : FAILURE in 28m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7059d384ec14945954df7b9dfb8e87e : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/628ad9a83eb04038b29671e69fe07845 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30508df71b8747a8ac7f4cc406b57f3a : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d79e183a16047a0a0d4cb16750735c0 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ff022a987c446d997be038b1bc91773 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/451d406ad8424071a496455fda0ac77f : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc7a913b9ce84ed996e171fb03541399 : FAILURE in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5ee9128037d465f981fa71a99eb62b2 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ace83209776dcd3e24e43af9cb155b5d38e6a31a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-04 12:20:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1cafbe9904211a752844df3e77b9269b09bd0c6e","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-04 13:01:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"89e0c85be1c7f185b39c1faaa1e0a12b7d87d6c9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-04 13:28:37.000000000","message":"Patch Set 4:\n\nHow about using sphinx-feature-classification sphinx extension rather than using designate-specific sphinx extension?","accounts_in_message":[],"_revision_number":4},{"id":"0ed565bc8a922dbc8c2e2b99132d786ff3535f9f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-04 14:04:46.000000000","message":"Patch Set 4:\n\nThanks for the suggestion Akihiro. I\u0027ll have a look and push out another revision of this change.","accounts_in_message":[],"_revision_number":4},{"id":"587e0e83178a896a628544ab269f648e888b60c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 14:43:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4d567eac265545e595b36d04e7a1bb46 : SUCCESS in 4m 30s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3a100d7c0d8745ef9cf11a019e2f96c9 : SUCCESS in 44m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/09e0a09be9244cf49308b691dcfd4674 : SUCCESS in 40m 21s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/20b32651d7ff49168b6ccd83184933fb : SUCCESS in 40m 24s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/a160a002deec4fcebeab58dd956586f7 : SUCCESS in 42m 43s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/59ed0f43a633419a80cd8c3b56aecc43 : SUCCESS in 44m 14s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/251573635a884a839081c225a279f801 : FAILURE in 33m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0d57c489b3eb4b3dbe5358edad13b489 : SUCCESS in 1h 05m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/08ef4259b1fc44e0ab6b0636ff741765 : SUCCESS in 46m 35s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9a9c8a21c03944c6bdcfc3443df5ed96 : SUCCESS in 41m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e6cb5823fe5493c9130343fd1183899 : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/226a682656314c08bad905941fe647a9 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da99497a1f34f479a9aabdb0da26402 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cf83f08e6cd4f719bf5b766c099b949 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40e8773df09947248dd48fbf7f3bb4f5 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62163b7456a14bc99abbe7905ea63bfc : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b360daabc8453dba13129f78aca17b : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/434f643e998244709d007dde1868ecf2 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":4},{"id":"1035e7eca43db1d41ab1809d785f8b2cb058f5c4","author":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"date":"2019-09-04 19:24:07.000000000","message":"Patch Set 4: Code-Review+1\n\nSo, looks like you dug deep. Thanks Nick!","accounts_in_message":[],"_revision_number":4},{"id":"b9948c2115f3892d21d4b5ab3005cba4dcf38332","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-04 20:11:30.000000000","message":"Patch Set 4:\n\nThanks Ritesh!","accounts_in_message":[],"_revision_number":4},{"id":"0b9bc615fbdafecd22965269389949197348a27c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-05 06:53:12.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1de80162d02df08f355406534742d990128d090f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-05 08:46:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"810eaadf0d20a71b04ba7846c5a2dc736c8ba714","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 08:50:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d9b0887d71d84a2ba906955280f19974 : SUCCESS in 4m 24s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/336d37354ddd46b595800751a46db808 : SUCCESS in 41m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1737200506154f26b5464d5a197a9116 : SUCCESS in 40m 30s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/84fe5410af3c42d6b69df859c617f69f : SUCCESS in 45m 52s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/ac9058cec82f41a19ef27e593b6fb181 : SUCCESS in 54m 07s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/4c267d2045694d7e9e1483528fdca8ff : SUCCESS in 37m 41s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/7c7b21d004d849a0abd67ffe8d3587d3 : FAILURE in 28m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/147c1846d3ab411787d71e84b19de35e : SUCCESS in 1h 07m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f1da82e5d3ec4bf9b2c56c0aec03b108 : SUCCESS in 36m 51s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/71feec50f93a4f8786548c0cd2914704 : SUCCESS in 45m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/651407d867024960b817923d8c2861c4 : SUCCESS in 8m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/758128e5c08544aabda05f65726c3bf5 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb16b4968f5341a18f6673097691cf1d : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee6b54566f8a4089b4088ebe2debdf85 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3943fa4e551b443da184411cc1ad67b0 : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d686079845b4f64aebe81e8028f42c9 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee95e1e9f7b44b0e966ef3dad2d763c1 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc346f1331a6434fad3f09e35645f419 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":4},{"id":"51056dc602b2eac474e0678e176a1021fedf8eab","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-05 08:53:08.000000000","message":"Topic changed from build_pdf_docs to build-pdf-docs","accounts_in_message":[],"_revision_number":4},{"id":"01da39fe219bcef73eb946496d158123ebb92161","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-05 08:53:33.000000000","message":"Patch Set 4:\n\nI updated the topic to build-pdf-docs. We use this topic name consistently.","accounts_in_message":[],"_revision_number":4},{"id":"3fe81066b15a5f3398d11815057503221924c949","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-05 20:43:42.000000000","message":"Patch Set 4:\n\nAkihiro, the support matrix from sphinx-feature-classification has a significantly different feature set and I don\u0027t quite know how best to map the data onto it. We could merge this change set as is to get the pdf doc feature unblocked, and then thing about potentially modifying sphinx-feature-classification so that the data here can be represented?","accounts_in_message":[],"_revision_number":4},{"id":"237d280985e7ae0fa91eb457f96fbc4db4c26a2c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-06 01:30:28.000000000","message":"Patch Set 4:\n\nNicolas, thanks for investigating the detail. I don\u0027t think the support matrix topic is not a blocker on the PDF build. I just mentioned it might help in the longer term.","accounts_in_message":[],"_revision_number":4},{"id":"4cd3cbe65e418a10721447c64afee5d013cdb3e6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-06 01:32:54.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c5fc7832313408e4e1984ddf850c6bd15770f8b0","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-06 12:06:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"82e4924a52699594eabdf2f60940e89c5f0f2fed","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-06 12:07:27.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"af6eb9cc09424179d52f2014fb5c4eea511e072e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:21:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2f1b4aa7576140728a75da7f0f65b45c : SUCCESS in 4m 20s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dc15c537617642c0941f152c28b61314 : SUCCESS in 37m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ee6e5c5d5bb943098f3a73d42437cf76 : SUCCESS in 38m 57s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/0a625ba17c04410bbf0194d76ae5c53d : SUCCESS in 42m 27s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/b4b87e6e20664696b0f1b46e67e647aa : SUCCESS in 42m 55s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/289446ebdd214553ab34ddbc893511dc : SUCCESS in 41m 25s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/21c3b85c2b4548309a8e2e5877d9bcaa : FAILURE in 32m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/265cd21eba6741ada331e0ac90a7af19 : SUCCESS in 1h 06m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9ff4e273bab84a5b8567d75e09efdf88 : SUCCESS in 43m 38s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/abeca2211300448ca28d1b86be364804 : SUCCESS in 39m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a252b542b0447dd9b9e401055f9da25 : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99eef216056e4114af7373f2d9fc6813 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532093a762a34e60b394f3589e70d3d9 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2e8d8b2752c43dcac93d4db7bd0e1e1 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c82cf3fde25a44679bcee56f1cb0fb75 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd4ba7fc610346db8cfecafbf2494de9 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b89af43684be45bfa9d7cf0f2c24ea22 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4655e491fb44e16bd7ec34515c4b229 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":5},{"id":"fa19c9229b70e1c1b193dd3191ab05becbee5eda","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-06 13:31:07.000000000","message":"Patch Set 5:\n\nLooking at the generated PDF file, it looks like we need more work on TOC at least.","accounts_in_message":[],"_revision_number":5},{"id":"4b5e7574cbedda5f1c2ab9d214bf2bf9af1b4297","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-06 21:04:59.000000000","message":"Patch Set 5:\n\nThe latest change set should fix the table of contents issue. Akihiro, please have another look.","accounts_in_message":[],"_revision_number":5},{"id":"70726b6b92a4b72f4458915e055d9ef19bd17ca2","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-06 21:05:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ec5fd32dd6b7e09c684cd195616fb8e81764525e","author":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"date":"2019-09-06 21:16:15.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9d2be49a017f3073fe74722823147c63e1487774","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:24:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6b82bab6236e4665abe86c789e772ad3 : SUCCESS in 4m 03s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d58a99b0b02c4587b6244141dd623b6e : SUCCESS in 47m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7c9bbb2fd74342d2a8fa8661cad5a5a8 : SUCCESS in 37m 46s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/dc80f439af704b89b291f77e4d847325 : SUCCESS in 43m 03s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/579d5a68ca464b52b21f8138396d1c08 : SUCCESS in 41m 25s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/a65dadcaf39347d6b067a3514e23ac27 : SUCCESS in 45m 30s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/693e08ab41c1463da286d15849d869ef : FAILURE in 35m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ba2f1ff1bd7c4cc7858db6d2b4af4908 : SUCCESS in 1h 08m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/04fd82db5bed44ac8307b1bcbb5a59d8 : SUCCESS in 44m 08s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/41f89d0e37f940cebe67802e89bf9089 : SUCCESS in 42m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/525ab70d467c450d90d8480631bd1f51 : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/657476f3a3024790a604acf2c032f7f2 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ffa4189c3f4671a6d3fa0b8c6cbfab : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0eeddc23fb64fe29e24e10ab5efeffe : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78cf35325ef04fec944cf21a53edd6b9 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6bf20d90f814b67a15c66f315b5ea00 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eab8d2466854b4ea136a342efb1c11c : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/505e78c943584c48b68ae21a5d0cfbff : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":6},{"id":"2ca828fcdfdfd2b5953c32dece6b1ab41f19cc95","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-07 09:26:18.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nThanks for updating the TOC. It looks much nicer than before.\n\nI see several points from the documentation perspective (for example, \"Index\" is not easy to read), but I am okay as it is up to designate team.\n\nMy inline comment is about the author field. It is shown in the title page, so I believe it is worth fixed in this patch.","accounts_in_message":[],"_revision_number":6},{"id":"8069bcd1b11dd5bdbceb1548a7a001b9db8e4709","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-07 09:27:00.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"980c6e3ccd556b084dc694fdd1d3d8ec2e41de0b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-07 09:27:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fc0b861976a2f7fc288c533454d2312700647563","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-07 09:27:22.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"065c1d56c8ec428069a372ebff04f0020844bff5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 11:11:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/da5f2bc44ee146fb88ca21d8ecd24355 : SUCCESS in 3m 37s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5ad4c4a06dfd42ea89c11c8532e1e240 : SUCCESS in 42m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/705378c90d59441f9d458f14fe9e873a : SUCCESS in 35m 28s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/0d357e845b1147e8a464ac0efcaf0bc2 : SUCCESS in 38m 16s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/60cf524113d240e48c1c58b8a0f842e0 : SUCCESS in 37m 50s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/3217a28659a34412b2991fc21784af9f : SUCCESS in 35m 40s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/e8266e548734402cb9945467c2262d4a : FAILURE in 31m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e81d3c9d4bcb4608b6b399b00728d69e : SUCCESS in 49m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3b20a2d9f7aa4aab86ab6849f3b63c50 : SUCCESS in 36m 45s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c9bc547a5e64485595e4f44422d23773 : SUCCESS in 36m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91245c36ff344d4dac9d2a200a1faacb : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46667d8d77dc4dcd8e85ac76a3bc5617 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ed053a5d4684aedbebf903315d6304e : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd2b6d4121714aed8699540ad7e94acc : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e339b36b9994b62b8b5ff23957e4dcd : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/488543abbb214c039b830690f12b2cec : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a593a8febed42c4b16f6d1a6dfc903c : POST_FAILURE in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c0d87ce369c4854be5b2cc223b7ec8d : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":7},{"id":"3cd2a34d2759af709767f1554608af41831ec3e1","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-07 13:06:49.000000000","message":"Patch Set 7:\n\nThanks Akihiro!","accounts_in_message":[],"_revision_number":7},{"id":"b64d472aaa2f8cf4fbe5d575b168d2433e85e499","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-09-07 13:08:33.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"f4cc10d2310126ded9a606575e24db2957465c6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 14:16:07.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b54b51a1c870487b9490ee048f44fda6 : SUCCESS in 4m 17s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1a3d18f10abf4ed99856a52be7cfd202 : SUCCESS in 39m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5865fe8849304d5284fd1fdf0df45527 : SUCCESS in 36m 49s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/5e017962315b4b79b5fe5017cd9c3660 : POST_FAILURE in 40m 13s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/da449e097e9448c9a7030bab292ac0e5 : SUCCESS in 39m 47s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/d932bd59fe1242158274fc95423fb6de : SUCCESS in 38m 40s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/e31e584f176949679490efef2705ef49 : FAILURE in 27m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f0c6a13615d74ac8a1c16c49c41fb4db : POST_FAILURE in 1h 03m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/01809dfd58f64f2784be1064a43dd447 : POST_FAILURE in 35m 44s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/96a2fc629c32425da9697eb541c65559 : POST_FAILURE in 43m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2f149553cd24e9ca90c2f617f96d87d : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20f6d9f6c11e42389fa3fab9c0971f3b : POST_FAILURE in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c9da278e5ff46f092088030f6921ac9 : POST_FAILURE in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ddb6dfd75fc4a1c9bb9ee0afc437b08 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e1c791ad5134cd3aaeae6761d1b0891 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3548204cf0541c79c1a08a0e070d43c : POST_FAILURE in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e59d267bc5a4ec9a0bd2a89d8af2f6a : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d3a8ba98f834a7486757a552d172e8d : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":7},{"id":"ea99750b71b8e2190444d4428d6448893af67966","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-08 03:07:17.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"1fcf3846ad9f64071e2cd981ab409d8944d7318b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-08 03:43:24.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"52648345f9cc4bd88a63d0efc05137db9be05821","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 04:20:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/57d68a4caa4646d7a457ed2bf6976e29 : SUCCESS in 3m 31s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0fb530fa6e5d45bda84ac1a0b1851465 : SUCCESS in 40m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/df57cfbeb7294e459fda335e6bba44d9 : SUCCESS in 37m 58s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/1e862848322a4e8792a3ff46ae56d3d4 : SUCCESS in 38m 22s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/ce1a742c0d4a447d8e1eecdd4a89c7e9 : SUCCESS in 39m 40s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/ca0f8d4560c4437584af0bb63c5cc0b4 : SUCCESS in 38m 04s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/71dc4b6b9ec9464d8af1c21b125a72de : FAILURE in 28m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6a3020b5c6ac40bc9cc277369b690f49 : SUCCESS in 59m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b41586b34233498a88a8ac2f07ee22da : SUCCESS in 38m 59s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/867326d6e6c04d1c9d744a20dfc88b1a : SUCCESS in 37m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9436b8c529f541e9b41e2cae078dc723 : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f484fa684e684e63a4f110a55540acb4 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa5bd4b106bb4de8b5bb3db10c7d112e : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b16f5c9006b4ec7b09e4ef883446f15 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ba2adde55ae4789988ed7c81763497b : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/608d21840b4d4f23a9b7088f79c2eb89 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50b9d5b8898f452faf190b27231bd429 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6f86f627a1444d4bd1de990e921fcca : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":7},{"id":"95d9bdf8c806b1755b730f80948ece1234d4a2c6","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-09-09 11:19:12.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nThanks Nicolas!","accounts_in_message":[],"_revision_number":7},{"id":"f1556551a4b85b6704779b7178519e34b7b87e39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:19:24.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"beefd4a23b95ed6656bdb7674a1dc56df012f756","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 12:31:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"e98fd5ee06d63854708444b95fbb0646638170e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 12:31:08.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c2228d1b4d774e1baebaa0cd63ce2b9f : SUCCESS in 3m 49s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c1e2925a6f754a49a3f04a0d7a4962a3 : SUCCESS in 40m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/05f7f485f6f248cdae01cf54a8aa779f : SUCCESS in 41m 12s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/3f59b7ce293447c197555823bfd1a48b : SUCCESS in 45m 55s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/e5c24b8ed644438d82dce707ebc4fb0d : SUCCESS in 42m 47s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/7d9cc0213a4146bfb7b4bbd479f48dee : SUCCESS in 41m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a5b94be9b5074fe1aaddad38352a3597 : SUCCESS in 1h 03m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8d8aebea9694b75a61034890599e535 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd6f28bab64485e945b20b2e13be354 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4a3e54a6892408e92e0287f2e64a605 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf5f80fc373f49838900f1fe71b350c2 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/031434eb47cc455e8575991d991ec94b : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7959f139270146b48c870a886398d57e : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79d8ed79c79b4092932099c23a49a20d : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":7},{"id":"69498a0e77b834a06ff9c8057998db49c8d35bd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 12:32:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/4441dc6dc08a44c983ac9a4f30846487 : SUCCESS in 55s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/574597d3304a4ba489e53a6754d69ad7 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"437cbf05ad3c56faeccdf02e4b6864eb418292f7","revisions":{"352c458da3d678e893025e8b965d3293a99023c0":{"kind":"REWORK","_number":1,"created":"2019-08-27 19:54:32.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/52/678952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/678952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/678952/1"}}},"commit":{"parents":[{"commit":"a6f3b340bcf59039959604f60a83f6100f413bfb","subject":"Merge \"Drop creating extra endpoints in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6f3b340bcf59039959604f60a83f6100f413bfb"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-08-26 19:41:01.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-08-27 19:52:21.000000000","tz":-360},"subject":"Add pdf-docs env to tox.ini","message":"Add pdf-docs env to tox.ini\n\nEnable PDF documentation. Based on [1]\n\n[1] https://review.opendev.org/601659\n\nChange-Id: Ie454f8f9ebc8e36347ce33b22f2e6da0257fe4e0\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/352c458da3d678e893025e8b965d3293a99023c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/352c458da3d678e893025e8b965d3293a99023c0"}]},"branch":"refs/heads/master"},"2b58910e4dda0be3681654a1ee0531ca2195ddad":{"kind":"REWORK","_number":2,"created":"2019-08-28 22:24:23.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/52/678952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/678952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/678952/2"}}},"commit":{"parents":[{"commit":"a6f3b340bcf59039959604f60a83f6100f413bfb","subject":"Merge \"Drop creating extra endpoints in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6f3b340bcf59039959604f60a83f6100f413bfb"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-08-26 19:41:01.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-08-28 21:56:58.000000000","tz":-360},"subject":"Add pdf-docs env to tox.ini","message":"Add pdf-docs env to tox.ini\n\n- Enable PDF documentation. Based on [1].\n- Removed literalinclude of designate.conf [2]\n- Removed illegal character (a literal ESC)\n\n[1] https://review.opendev.org/601659\n[2] https://github.com/sphinx-doc/sphinx/issues/3099\n\nChange-Id: Ie454f8f9ebc8e36347ce33b22f2e6da0257fe4e0\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b58910e4dda0be3681654a1ee0531ca2195ddad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b58910e4dda0be3681654a1ee0531ca2195ddad"}]},"branch":"refs/heads/master"},"2dc7141fb0898bc3097b22deba3658ea03822df2":{"kind":"REWORK","_number":3,"created":"2019-09-04 12:20:45.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/52/678952/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/678952/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/678952/3"}}},"commit":{"parents":[{"commit":"a6f3b340bcf59039959604f60a83f6100f413bfb","subject":"Merge \"Drop creating extra endpoints in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6f3b340bcf59039959604f60a83f6100f413bfb"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-08-26 19:41:01.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-09-04 12:20:02.000000000","tz":-360},"subject":"Add pdf-docs env to tox.ini","message":"Add pdf-docs env to tox.ini\n\n- Enable PDF documentation. Based on [1].\n- Removed literalinclude of designate.conf [2]\n- Removed illegal character (a literal ESC)\n- Fix incorrect column count in backends summary section\n\n[1] https://review.opendev.org/601659\n[2] https://github.com/sphinx-doc/sphinx/issues/3099\n\nChange-Id: Ie454f8f9ebc8e36347ce33b22f2e6da0257fe4e0\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2dc7141fb0898bc3097b22deba3658ea03822df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2dc7141fb0898bc3097b22deba3658ea03822df2"}]},"branch":"refs/heads/master"},"d9e4bd821fd8830dd8a11f41a43cf760e1bd33bc":{"kind":"REWORK","_number":4,"created":"2019-09-04 13:01:20.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/52/678952/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/678952/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/678952/4"}}},"commit":{"parents":[{"commit":"a6f3b340bcf59039959604f60a83f6100f413bfb","subject":"Merge \"Drop creating extra endpoints in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6f3b340bcf59039959604f60a83f6100f413bfb"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-08-26 19:41:01.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-09-04 13:00:40.000000000","tz":-360},"subject":"Add pdf-docs env to tox.ini","message":"Add pdf-docs env to tox.ini\n\n- Enable PDF documentation. Based on [1].\n- Removed literalinclude of designate.conf [2]\n- Removed illegal character (a literal ESC)\n- Fix incorrect column count in backends summary section\n- Replace nodes.line() with nodes.paragraph()\n\n[1] https://review.opendev.org/601659\n[2] https://github.com/sphinx-doc/sphinx/issues/3099\n\nChange-Id: Ie454f8f9ebc8e36347ce33b22f2e6da0257fe4e0\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9e4bd821fd8830dd8a11f41a43cf760e1bd33bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9e4bd821fd8830dd8a11f41a43cf760e1bd33bc"}]},"branch":"refs/heads/master"},"6e67857fd9a7e3af5c6e8dd3e9ebd3340e80bdb2":{"kind":"REWORK","_number":5,"created":"2019-09-06 12:06:58.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/52/678952/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/678952/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/678952/5"}}},"commit":{"parents":[{"commit":"a6f3b340bcf59039959604f60a83f6100f413bfb","subject":"Merge \"Drop creating extra endpoints in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6f3b340bcf59039959604f60a83f6100f413bfb"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-08-26 19:41:01.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-09-06 12:06:47.000000000","tz":-360},"subject":"Add pdf-docs env to tox.ini","message":"Add pdf-docs env to tox.ini\n\n- Enable PDF documentation. Based on [1].\n- Removed literalinclude of designate.conf [2]\n- Removed illegal character (a literal ESC)\n- Fix incorrect column count in backends summary section\n- Replace nodes.line() with nodes.paragraph()\n\n[1] https://review.opendev.org/601659\n[2] https://github.com/sphinx-doc/sphinx/issues/3099\n\nChange-Id: Ie454f8f9ebc8e36347ce33b22f2e6da0257fe4e0\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e67857fd9a7e3af5c6e8dd3e9ebd3340e80bdb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e67857fd9a7e3af5c6e8dd3e9ebd3340e80bdb2"}]},"branch":"refs/heads/master"},"343aff901de8418ba3cef57b864ccd8441abcca0":{"kind":"REWORK","_number":6,"created":"2019-09-06 21:05:20.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/52/678952/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/678952/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/678952/6"}}},"commit":{"parents":[{"commit":"a6f3b340bcf59039959604f60a83f6100f413bfb","subject":"Merge \"Drop creating extra endpoints in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6f3b340bcf59039959604f60a83f6100f413bfb"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-08-26 19:41:01.000000000","tz":-360},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-09-06 21:03:53.000000000","tz":-360},"subject":"Add pdf-docs env to tox.ini","message":"Add pdf-docs env to tox.ini\n\n- Enable PDF documentation. Based on [1].\n- Enable table of contents generation for pdf\n- Removed literalinclude of designate.conf [2]\n- Removed illegal character (a literal ESC)\n- Fix incorrect column count in backends summary section\n- Replace nodes.line() with nodes.paragraph()\n\n[1] https://review.opendev.org/601659\n[2] https://github.com/sphinx-doc/sphinx/issues/3099\n\nChange-Id: Ie454f8f9ebc8e36347ce33b22f2e6da0257fe4e0\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/343aff901de8418ba3cef57b864ccd8441abcca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/343aff901de8418ba3cef57b864ccd8441abcca0"}]},"branch":"refs/heads/master"},"437cbf05ad3c56faeccdf02e4b6864eb418292f7":{"kind":"REWORK","_number":7,"created":"2019-09-07 09:27:00.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/52/678952/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/678952/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/678952/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/678952/7"}}},"commit":{"parents":[{"commit":"a6f3b340bcf59039959604f60a83f6100f413bfb","subject":"Merge \"Drop creating extra endpoints in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6f3b340bcf59039959604f60a83f6100f413bfb"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@suse.com","date":"2019-08-26 19:41:01.000000000","tz":-360},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-09-07 09:26:57.000000000","tz":0},"subject":"Add pdf-docs env to tox.ini","message":"Add pdf-docs env to tox.ini\n\n- Enable PDF documentation. Based on [1].\n- Enable table of contents generation for pdf\n- Removed literalinclude of designate.conf [2]\n- Removed illegal character (a literal ESC)\n- Fix incorrect column count in backends summary section\n- Replace nodes.line() with nodes.paragraph()\n\n[1] https://review.opendev.org/601659\n[2] https://github.com/sphinx-doc/sphinx/issues/3099\n\nChange-Id: Ie454f8f9ebc8e36347ce33b22f2e6da0257fe4e0\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/437cbf05ad3c56faeccdf02e4b6864eb418292f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/437cbf05ad3c56faeccdf02e4b6864eb418292f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
