)]}'
{"id":"openstack%2Fdesignate~679550","triplet_id":"openstack%2Fdesignate~master~I61b64f3adbf238235a6f0aadccece682c9f27431","project":"openstack/designate","branch":"master","topic":"fix_legacy_facade","hashtags":[],"change_id":"I61b64f3adbf238235a6f0aadccece682c9f27431","subject":"Testing sql","status":"ABANDONED","created":"2019-09-01 00:59:01.000000000","updated":"2019-09-01 04:17:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"93878847b0536aee7276b18cff6abbfa5905fe8d","_number":679550,"virtual_id_number":679550,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-09-01 04:07:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-01 04:07:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"411f7da17bff521af5c7b446d8ffa712c747fdf4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 00:59:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88eaf4fa10b3d9d354d60417f1c01537ce3be591","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 01:33:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"464c48eadffdf802213cbb18f72db5e4d6de68fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 02:22:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bfc796b7ce9e4e15bfe9ccfe44fc1859 : FAILURE in 27m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2652a94d7ae04a5ba23c142a69a7537a : FAILURE in 40m 01s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/307da05a786d45868fa8180471df5b7d : FAILURE in 25m 54s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/6ac74dedad264f038b7d1927025633a7 : FAILURE in 42m 28s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/05652e5a4c234946b98d444709002d2f : FAILURE in 40m 56s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/e1823e517d9e406a842be0380cf78503 : FAILURE in 26m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0bf3b6c5f2e14643b00450fdfa54ac04 : FAILURE in 46m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6dd0ca8d58e947deb50d5b05d8955456 : FAILURE in 40m 31s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/402804007cbd4af1a89b9d07818592ce : FAILURE in 29m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10cc237f1ddb41568b7635120237dd1f : FAILURE in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9e8376ac3c74b0d9cd6a5eed3cac680 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953b8d2ac04f46dab38830a4a8b984a2 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e9b8b48e36d42db94d72da78ef49723 : FAILURE in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4448227887814a1a9560657a9ac42ea0 : FAILURE in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6171e8f3ccd5441a91cbe0ad4ecba61b : FAILURE in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5e2cc312ea4616a1d7f8d2f9ac2ada : SUCCESS in 5m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3da66b02719f4fefbce29a2de91fdb2d : FAILURE in 32m 26s","accounts_in_message":[],"_revision_number":2},{"id":"4ec2a41a2a3c969cf7268a20c3b3e57914759f44","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 03:17:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d8de6d2e86bf27d28ab50f59d309ba5c2bf7fe4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 04:07:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1d80e67c754e493497ed207a17af0f5d : SUCCESS in 43m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2656ee9c14864c4caa522f3f068c23a7 : SUCCESS in 42m 40s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/5784edc5465640f08928b2d00805059f : SUCCESS in 37m 13s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/be394047285a46a49453b198513e9073 : SUCCESS in 44m 46s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/bc148a088d6a4ed1b8eac5ffe3ab6dbc : SUCCESS in 39m 36s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8dae8cf486694893a503f0bea61965c2 : FAILURE in 28m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e4186a6e4b05408e9c852e60f2fb00b2 : FAILURE in 47m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/27ab6a8cecf946c0b0189629af9f47a9 : SUCCESS in 40m 23s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d19d420d1ab647529d8be7e194a6cf05 : SUCCESS in 38m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/792e0d13cd3a46cc9599448175e07afc : FAILURE in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a874f37d75c401aa6e9227de02c55c8 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3db820d1824e0a91f7b18b7605e4c8 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/269be237db3345e8b686bf19d45343f9 : FAILURE in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fff2f4d4111a42b7999069c3dcced512 : FAILURE in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b92344afb88c4e189c7c6fd900266596 : FAILURE in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36a5dfa5d1b41a79e7fb0c715864fb6 : SUCCESS in 4m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f84826f83734f14a95f896fb6dc58ba : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":3},{"id":"42d23c6b5c32b08095c37df720749045b5511d2b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-01 04:17:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"751d0325e14eff13bbb132d20801f8fdab3d80bd","revisions":{"74a4c0f646d3653f43a8a1c8560caff879b9c800":{"kind":"REWORK","_number":1,"created":"2019-09-01 00:59:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/679550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/679550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/679550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/679550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/679550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/679550/1"}}},"commit":{"parents":[{"commit":"5e6a1d303905b1c206d148165c06faacbe2ba4c7","subject":"Refactored service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e6a1d303905b1c206d148165c06faacbe2ba4c7"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 00:58:44.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 00:58:44.000000000","tz":-420},"subject":"Testing sql","message":"Testing sql\n\nChange-Id: I61b64f3adbf238235a6f0aadccece682c9f27431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74a4c0f646d3653f43a8a1c8560caff879b9c800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74a4c0f646d3653f43a8a1c8560caff879b9c800"}]},"branch":"refs/heads/master"},"a7e4f237ed8ecbb78843f1826731d2d60f3da2a5":{"kind":"REWORK","_number":2,"created":"2019-09-01 01:33:00.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/679550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/679550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/679550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/679550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/679550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/679550/2"}}},"commit":{"parents":[{"commit":"5e6a1d303905b1c206d148165c06faacbe2ba4c7","subject":"Refactored service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e6a1d303905b1c206d148165c06faacbe2ba4c7"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 00:58:44.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 01:32:47.000000000","tz":-420},"subject":"Testing sql","message":"Testing sql\n\nChange-Id: I61b64f3adbf238235a6f0aadccece682c9f27431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a7e4f237ed8ecbb78843f1826731d2d60f3da2a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a7e4f237ed8ecbb78843f1826731d2d60f3da2a5"}]},"branch":"refs/heads/master"},"751d0325e14eff13bbb132d20801f8fdab3d80bd":{"kind":"REWORK","_number":3,"created":"2019-09-01 03:17:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/50/679550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/679550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/679550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/679550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/679550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/679550/3"}}},"commit":{"parents":[{"commit":"5e6a1d303905b1c206d148165c06faacbe2ba4c7","subject":"Refactored service layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e6a1d303905b1c206d148165c06faacbe2ba4c7"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 00:58:44.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-01 03:17:02.000000000","tz":-420},"subject":"Testing sql","message":"Testing sql\n\nChange-Id: I61b64f3adbf238235a6f0aadccece682c9f27431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/751d0325e14eff13bbb132d20801f8fdab3d80bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/751d0325e14eff13bbb132d20801f8fdab3d80bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
