)]}'
{"id":"openstack%2Fdesignate~679586","triplet_id":"openstack%2Fdesignate~master~I2d62fe1f530637c2d80df8f5d69a7e3159bda37d","project":"openstack/designate","branch":"master","topic":"updated-releasenotes-train","hashtags":[],"change_id":"I2d62fe1f530637c2d80df8f5d69a7e3159bda37d","subject":"Updated Release notes for Train","status":"MERGED","created":"2019-09-02 02:20:39.000000000","updated":"2019-09-03 03:15:28.000000000","submitted":"2019-09-03 03:14:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679586-1567480444291-5d26ed10","meta_rev_id":"6a8d6de64116dd5e1d2448b8ce25e1bdf5a4eefb","_number":679586,"virtual_id_number":679586,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-09-03 03:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2019-09-02 09:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-03 02:13:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-02 09:27:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-09-03 03:14:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e575fb1653ee5d9e723bd89d0dc4d856605b9379","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-02 02:20:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2963ac5fe4955da5d3c0db2f55bf265b1eccae7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-02 02:25:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a811a4ed0f86cf794d2e331818a2eed324141216","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 03:22:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/361c30dc62694756a061aa95863261b7 : SUCCESS in 55m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8dd1b06c404a9e8bf18210f9ffb725 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8cc3b6953b7443d9174f3c7ef4ca606 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bdf1f011b6d491ba67366cc992b77dd : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"3010ce1530efd1f12aa05198ce117b899187ce30","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-09-02 09:27:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fdf8141248cae3041f6bdee6fb8ef21c7d122fb1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-03 02:13:36.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b04ed12283eec3b85721705ca838b1d35c356e25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 02:13:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cb716265f70231432d4ef26ebf4aa0dfb4b225f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 03:14:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e2110ca8d9449e6e862018824eb60ceae2f6aeae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 03:14:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e3e76463675648419fb134814b0350b5 : SUCCESS in 57m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6c5ce77f77844a19007201ef1e6f7d3 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3831f371890440b979c9dbabbdb237b : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a834cd448294e61b1e9726dc2554b4e : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":2},{"id":"34785ca78c1f7f633d877ed000717d474984ca27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 03:15:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb5854a69bf04929977b8c9edaa93de7 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b04e4a45552022a8a60ac4e217a5c70662ad5369","revisions":{"24ff3c528b76236de22ab883c132e029060f3f43":{"kind":"REWORK","_number":1,"created":"2019-09-02 02:20:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/86/679586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/86/679586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/679586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/679586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/679586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/86/679586/1"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-02 02:20:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-02 02:20:00.000000000","tz":-420},"subject":"Updated Release notes for Train","message":"Updated Release notes for Train\n\nChange-Id: I2d62fe1f530637c2d80df8f5d69a7e3159bda37d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/24ff3c528b76236de22ab883c132e029060f3f43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/24ff3c528b76236de22ab883c132e029060f3f43"}]},"branch":"refs/heads/master"},"b04e4a45552022a8a60ac4e217a5c70662ad5369":{"kind":"REWORK","_number":2,"created":"2019-09-02 02:25:08.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/86/679586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/86/679586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/679586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/679586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/679586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/86/679586/2"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-02 02:20:00.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-02 02:24:59.000000000","tz":-420},"subject":"Updated Release notes for Train","message":"Updated Release notes for Train\n\nChange-Id: I2d62fe1f530637c2d80df8f5d69a7e3159bda37d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b04e4a45552022a8a60ac4e217a5c70662ad5369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b04e4a45552022a8a60ac4e217a5c70662ad5369"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
