)]}'
{"id":"openstack%2Fdesignate~679589","triplet_id":"openstack%2Fdesignate~master~If3fd2f808a65fd036e351b55975c48d59348933a","project":"openstack/designate","branch":"master","topic":"skip_grenade_on_rn","hashtags":[],"change_id":"If3fd2f808a65fd036e351b55975c48d59348933a","subject":"Skip grenade jobs on Release note changes","status":"MERGED","created":"2019-09-02 03:15:36.000000000","updated":"2019-09-04 19:01:23.000000000","submitted":"2019-09-04 18:56:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679589-1567623403177-233c4fa7","meta_rev_id":"9f2f28024ab39e6311fd2ff50ee4c762c8bdf9d8","_number":679589,"virtual_id_number":679589,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-09-04 18:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-09-04 15:28:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2019-09-04 17:49:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-09-04 17:49:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 15:28:07.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-09-04 17:49:39.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-09-04 18:56:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d05697c76e9c976791b828fa293e1f9dd9ee7115","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-02 03:15:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af76ad0d8fd36a76f6f8056d036eb73a2529b89c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 04:20:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a403b598e5c74c44815a9cb3aa74997c : SUCCESS in 40m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5b11f43835e943e3ba31c55afb57a3e9 : SUCCESS in 38m 20s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/e675dc311aa545e3b74aec2b20a8afab : SUCCESS in 39m 58s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/929aa519ef2046aebd1f2774d2ceb8ec : SUCCESS in 43m 54s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/41204bb48a93458296d0f9a7080324cd : SUCCESS in 39m 20s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/c7dbf5e2631b4398a1f3aa0173b594de : FAILURE in 35m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c0100679558348938b71695a2759f5ea : SUCCESS in 58m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d421601cce3d40e8955fdb147f800346 : SUCCESS in 46m 58s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bc0ef89d5bc34141ac58449b9003e8f6 : SUCCESS in 42m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5930f95e4e0740c59655ee592a8c3a2a : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fb295d4f44e4b018b32af250d3aa74a : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b55dd1109d184470a80835b47f022ead : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50a714a285ec459d9271f79aeda97923 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b83a0b84ef6b4649baa00171d14cd67a : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5771e7b45b354c2bbc55fb88d710d0e8 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/347ae6f9693e49afb48e03f28038c49c : SUCCESS in 4m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4e6f584b53384584af4580c3d172c779 : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b99c29ca025f535a7f186960fef1f76427458ef1","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-09-04 15:28:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4fb1e17ffe9e344631a1d0bc85b1f66711d5a636","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-09-04 17:49:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c59a26362f2b8f25eae63b183e822c0fb465e829","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 17:49:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d967d86c84a4bbba3b6a6842e1f6c9b39744fd83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 18:56:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ba729243df8743afb2631601a55c9923 : SUCCESS in 42m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/90b8936898e14dffb8072a3946bb2a92 : SUCCESS in 52m 20s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/e27b3bd5320047359ace5ed495c7f61d : SUCCESS in 48m 02s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/5471c0ffb49f4304b68dbc2b310e4be6 : SUCCESS in 43m 21s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/44e1502bfea14ba0bc4fdaeea84f9957 : SUCCESS in 49m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/088a440da04744d29786acd6432d82ae : SUCCESS in 1h 02m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a9e5582a20a4813bc965ad36eca56a5 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a346719777d438c8eaa860868d49e3b : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df2f87b9980f4630aaf34a178861e6a0 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57debd3a6b9048fca7c9867754b582c7 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06576be31fdf4e07a7a00085870861ea : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c4d54bb9674faa9ffc4b79a795edbd : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"93bf3f1ffa02fffddb94e84274466e1f99432fe8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 18:56:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e4556e064fa2e9847fcab825d909c7f80cfcb492","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 19:01:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04bb3c5d82244336bd086b712fc26b9e : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3afcde0448eda2e3ae336f9ffdbc90cf55f3364","revisions":{"a3afcde0448eda2e3ae336f9ffdbc90cf55f3364":{"kind":"REWORK","_number":1,"created":"2019-09-02 03:15:36.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/89/679589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/89/679589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/679589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/679589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/679589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/89/679589/1"}}},"commit":{"parents":[{"commit":"1ebc46950d43758c685689c35ede3363a108675f","subject":"Merge \"Fix ipv6 bugs with devstack and pdns4 driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ebc46950d43758c685689c35ede3363a108675f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-02 03:14:24.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-02 03:14:24.000000000","tz":-420},"subject":"Skip grenade jobs on Release note changes","message":"Skip grenade jobs on Release note changes\n\nChange-Id: If3fd2f808a65fd036e351b55975c48d59348933a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3afcde0448eda2e3ae336f9ffdbc90cf55f3364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3afcde0448eda2e3ae336f9ffdbc90cf55f3364"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
