)]}'
{"id":"openstack%2Fdesignate~681130","triplet_id":"openstack%2Fdesignate~master~Ia1bc4aea7a4d9daeb6aa9dcfbb7defc816f592eb","project":"openstack/designate","branch":"master","topic":"fix_axfr_with_attributes","hashtags":[],"change_id":"Ia1bc4aea7a4d9daeb6aa9dcfbb7defc816f592eb","subject":"Allow axfr for zones with different pool_id","status":"ABANDONED","created":"2019-09-09 21:45:55.000000000","updated":"2022-09-24 01:30:06.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9d7edd16e547e4d1810022b8fb1121eb14133d48","_number":681130,"virtual_id_number":681130,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"all":[{"value":1,"date":"2022-09-24 01:04:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 08:19:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-11 07:31:20.000000000","updated_by":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"reviewer":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"state":"REVIEWER"}],"messages":[{"id":"17fb48515ad12253b9503e1ce8e8ce165ae9268b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-09 21:45:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2418544ded921d4a64a7a8d3a7789e541d2c535","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 22:56:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/df20b87adc074ad3aec178d3d6330bbb : SUCCESS in 37m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/31c5941459964452a03e0c4c8ea42e6d : SUCCESS in 44m 25s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/2c970a679a5b41be9d39a8e7a2889671 : SUCCESS in 41m 30s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/fea814abb9444337a9b7343c99bd74cf : SUCCESS in 44m 57s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/bdbf265d69334b2b8f5637a7bb50e4cc : SUCCESS in 44m 14s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/ef744287e59d4e94bae84f5614a8d317 : FAILURE in 31m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/84c768e89f6a46188f87a7e047d7f4f4 : SUCCESS in 1h 03m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9ba26eb2715a4d8fabbefa4d353f54ea : SUCCESS in 1h 03m 18s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0493def3073c48f69a01d7e329b1b952 : SUCCESS in 40m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/897525c016ab4d1cb3c7043483f6aae0 : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ed6bb4f48a74797978eeb7628ba0a53 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f1ac81858854c4f8aea3125faa61391 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/663385b5857446f78ae942224c72db35 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d0a417187a646bb97ca2ca9d927beec : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b323101e877418e89231a8d22f21eae : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/340c360a2a074f68882972014b75dbbf : SUCCESS in 6m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a378d5cf8d67455b80c7c0c77c4cbcfa : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":1},{"id":"61db4dbf0d833a1b1140f0a15efd49d2b82649f9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-10 06:08:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"81c416bc1ddc10d6309ac04fe5af8ddf84dfdb10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 08:19:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d60787ae7f7946b0ba14a1a61b6cb3f8 : SUCCESS in 42m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3f0101efa4f54590822e712a6614674f : SUCCESS in 38m 40s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/c5962fb806de40228feac2e48bdfdf18 : SUCCESS in 45m 50s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/1b0af7bc47c94d2481820192dadffa9b : SUCCESS in 45m 12s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/e451fa4d25a14ec7a84ceddd331f88dd : SUCCESS in 40m 39s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/7508a9a8e4e64b14a47e888c6036c655 : FAILURE in 24m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5d24dae5e30140c59c914fa0f381221e : SUCCESS in 58m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ed918b24b0244179a695f388fc47c0b2 : SUCCESS in 44m 52s (non-voting)\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/00d089f7976d4112a80f9f8e8e4b9ca0 : SUCCESS in 43m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb9f4f96319c4ef18901f58df019f6a2 : SUCCESS in 7m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab508f83557741a9a1908e8b94b0a564 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b4f49cb93ac48df90ea5a0f04d83f8e : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72033ba5134f438bbbf9b2c58891a13e : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c09c5afb0294aa6b1880a677066e3fc : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e18476f390504b91b74b3448b56d3663 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a57811523a44709ae391fd94ece64a6 : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c8a1a5ae34cf4637ac5b9796036d166a : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":2},{"id":"fd15744cafcf332f7a3f97cfb513c362abdcc250","author":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"date":"2019-09-11 07:31:20.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nLooks good. Comment inline on what I understood from the change. Thanks Erik!","accounts_in_message":[],"_revision_number":2},{"id":"daac2265f2f4b10e19c50034d57f1e13b7cb1e8d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-11 07:39:34.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e Looks good. Comment inline on what I understood from the change.\n \u003e Thanks Erik!\n\nYep that is correct.","accounts_in_message":[],"_revision_number":2},{"id":"cc99913926cfde0714d3a1ab533470edd2f85112","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-23 02:29:28.000000000","message":"Abandoned\n\nTalking to Graham it sounds like this goes against the original design. We should probably look at documenting tsigkey usage patterns here instead.","accounts_in_message":[],"_revision_number":2},{"id":"2bf91257592ce90ae2e95e5c07e1dc5365bb5c2e","tag":"autogenerated:gerrit:restore","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 01:04:09.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"2d97867abdf37d90c8f353254f6711808996589c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 01:04:20.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9d7edd16e547e4d1810022b8fb1121eb14133d48","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 01:30:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"47a79fb53fea15b0c9dabcbd86376860ab227b97","revisions":{"da3d10d6f60e17b24f924d5b6abdf504766769c2":{"kind":"REWORK","_number":1,"created":"2019-09-09 21:45:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/30/681130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/30/681130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/681130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/681130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/681130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/30/681130/1"}}},"commit":{"parents":[{"commit":"9fb598b7d1582d6063913cb97f50863e0a0586c8","subject":"Merge \"Fix bug in recordset status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9fb598b7d1582d6063913cb97f50863e0a0586c8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-09 20:03:16.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-09 21:45:07.000000000","tz":-420},"subject":"Fix axfr for zones with attributes","message":"Fix axfr for zones with attributes\n\nThe current implementation limits AXFR to the\ndefault pool.\n\nChange-Id: Ia1bc4aea7a4d9daeb6aa9dcfbb7defc816f592eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da3d10d6f60e17b24f924d5b6abdf504766769c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/da3d10d6f60e17b24f924d5b6abdf504766769c2"}]},"branch":"refs/heads/master"},"ec0200c614ab9b54085cf435987a8dfbc15f3f4b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-10 06:08:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/30/681130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/30/681130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/681130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/681130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/681130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/30/681130/2"}}},"commit":{"parents":[{"commit":"9fb598b7d1582d6063913cb97f50863e0a0586c8","subject":"Merge \"Fix bug in recordset status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9fb598b7d1582d6063913cb97f50863e0a0586c8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-09 20:03:16.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-10 06:05:59.000000000","tz":-420},"subject":"Allow axfr for zones with different pool_id","message":"Allow axfr for zones with different pool_id\n\nThe current implementation limits AXFR to the\ndefault pool when the dns server isn\u0027t providing\na tsigkey.\n\nChange-Id: Ia1bc4aea7a4d9daeb6aa9dcfbb7defc816f592eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec0200c614ab9b54085cf435987a8dfbc15f3f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec0200c614ab9b54085cf435987a8dfbc15f3f4b"}]},"branch":"refs/heads/master"},"47a79fb53fea15b0c9dabcbd86376860ab227b97":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-24 01:04:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/30/681130/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/30/681130/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/681130/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/681130/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/681130/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/30/681130/3"}}},"commit":{"parents":[{"commit":"e3ac092a303e1f26a31797a6886a97d857f03bae","subject":"Merge \"Fix mdns test_handler setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3ac092a303e1f26a31797a6886a97d857f03bae"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-09 20:03:16.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-09-24 01:04:20.000000000","tz":0},"subject":"Allow axfr for zones with different pool_id","message":"Allow axfr for zones with different pool_id\n\nThe current implementation limits AXFR to the\ndefault pool when the dns server isn\u0027t providing\na tsigkey.\n\nChange-Id: Ia1bc4aea7a4d9daeb6aa9dcfbb7defc816f592eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47a79fb53fea15b0c9dabcbd86376860ab227b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47a79fb53fea15b0c9dabcbd86376860ab227b97"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
