)]}'
{"id":"openstack%2Fdesignate~683276","triplet_id":"openstack%2Fdesignate~master~Ie8d21dddb60ff853a96ec093f58d449672c484e8","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"Ie8d21dddb60ff853a96ec093f58d449672c484e8","subject":"update api-ref set x-auth-token required","status":"ABANDONED","created":"2019-09-20 03:35:13.000000000","updated":"2019-09-23 08:06:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5287d1b8863053708da2430a7ae7178ae4d69af6","_number":683276,"virtual_id_number":683276,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-09-20 04:26:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 04:26:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f762f5b92b1ea156974b64f2eb11f075921e6da5","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-09-20 03:35:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bce8d8ca6fff140bdae650e385a70df33a49532c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 04:26:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3d21229382534c0cae9d9121157404f5 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc2d244124b848819b9389bc2f201c8c : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0943770857f84d0dbdf2ca7296971924 : SUCCESS in 7m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c00bc04eecbd42409b17304edb30b37e : SUCCESS in 43m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5287d1b8863053708da2430a7ae7178ae4d69af6","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-09-23 08:06:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ff418aa5bc53ce89fc5b64c7172d8a5b06ad7f4","revisions":{"0ff418aa5bc53ce89fc5b64c7172d8a5b06ad7f4":{"kind":"REWORK","_number":1,"created":"2019-09-20 03:35:13.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/76/683276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/683276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/683276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/683276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/683276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/683276/1"}}},"commit":{"parents":[{"commit":"23f6a79aef1303d25f8eb1c9d5090fe7f4326d46","subject":"Merge \"Add missing RabbitMQ configuration to install guides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23f6a79aef1303d25f8eb1c9d5090fe7f4326d46"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-09-20 03:33:57.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-09-20 03:34:54.000000000","tz":480},"subject":"update api-ref set x-auth-token required","message":"update api-ref set x-auth-token required\n\nChange-Id: Ie8d21dddb60ff853a96ec093f58d449672c484e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ff418aa5bc53ce89fc5b64c7172d8a5b06ad7f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ff418aa5bc53ce89fc5b64c7172d8a5b06ad7f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
