)]}'
{"id":"openstack%2Fdesignate~684843","triplet_id":"openstack%2Fdesignate~master~Iba9ba655c47ee085a47ee98872437f07fdc3346c","project":"openstack/designate","branch":"master","topic":"wsgi_support","hashtags":[],"change_id":"Iba9ba655c47ee085a47ee98872437f07fdc3346c","subject":"Update devstack to use workers","status":"MERGED","created":"2019-09-25 20:55:22.000000000","updated":"2019-10-01 18:14:52.000000000","submitted":"2019-10-01 18:13:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684843-1569953594345-02882d20","meta_rev_id":"d999c5bcc02d9bfc559061cc4479a9c732c729ad","_number":684843,"virtual_id_number":684843,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 18:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2019-10-01 11:13:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-01 17:06:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 11:13:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-10-01 18:13:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee9e54c772e5638e12ce644454770bee996b551c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-25 20:55:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3121fe795e291b2425f792eb66cf6e8932d4fc65","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-25 21:08:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce8b3c139f122af19eeadbcd0696cc58891ad057","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-25 21:56:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aba88aa3b2f6e1e2217c360585a64b4972a74408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 23:04:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b33226adf7f848ccbff790bb9cbe9503 : SUCCESS in 39m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/836d0dcd017241a380d7a3619e6ce61d : SUCCESS in 36m 51s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/bbad94bb6e654179b3286983a2cf7787 : SUCCESS in 41m 51s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/c12d3ba9295b46989515aaba22db866c : SUCCESS in 43m 55s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/e7cfe9e7b6a94273b52c9491177dbc98 : FAILURE in 30m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2a900bb36dd443909141d42223524ce9 : SUCCESS in 1h 05m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/10bfbbef4df54720a9b182e6fd565e87 : SUCCESS in 41m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cbeceb6ecd744c09a7221d209583846b : SUCCESS in 38m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9f3464ada4f47609c03d687d1d0a8c6 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9894e0fe6e44fa9a6b1a0421405387b : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d969bb6e1cd9423997104c1c5a6aaf36 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5bda7b928bdc471d9b6f34b019a8b061 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6116b03092f94bb09ccd32b1be4f8678 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42f58d7a681e4246b7dec54ed5653b93 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/191555af8ab1497686fab96fa2ff0319 : SUCCESS in 6m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b63b0f927804ef6a17722427758db6e : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":3},{"id":"c5ef1ce674cab6b93e4ffb801ca7834515c42a12","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-28 03:29:56.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"72c2504ca262ac54d284cba3f0cfd8b1632c657f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 04:35:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/27ae75e1157c4eff836d27e3c3841959 : SUCCESS in 38m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5bf8d6b113114895b9edc2589c8774cb : SUCCESS in 38m 15s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/d2a102a0f4c140a7800ae1d3f5f300b9 : SUCCESS in 40m 45s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/7a253f8dd7b04e7daa73fb03891c12aa : SUCCESS in 37m 30s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/37bc57c5bc5c423b92ad92995d5a94c6 : FAILURE in 28m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/607111d317434423bb978a23255fdead : SUCCESS in 1h 00m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3f2f3a4b04a84afab44c6e357e475206 : SUCCESS in 39m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/41c4f7f148634ea792e6df2fae766533 : SUCCESS in 36m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac28fc6b7a8e465a9a1fbf77cfd4d7b8 : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee7b5da3de13478c883a532a3cdbdb93 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e2d67a1f9a49fda42959cca22b1eff : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e36ebbdb233d48cdbe19abd2f4fd15ac : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f62d590c40bb4a5986a49a94ee2e560c : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41488fbb88c6414d9a8771f44e5e6170 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78fe0f52533435e945022052613dcb5 : SUCCESS in 7m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/852badc41384422c8c1eded87e79e0eb : SUCCESS in 36m 17s","accounts_in_message":[],"_revision_number":4},{"id":"2c30f13a422132ff237339554b38b91a0fdf0157","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-10-01 11:13:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3c77490d0cfbbffa00cfc490f0d400fd36a84320","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-01 17:06:28.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0bf96bdcd501cef1dd67cc00c6ee8094ed90df57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 17:06:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c480bff93e601653cc4107a16dbb2860a1f259b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 18:13:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/349e3d6f7de94bf1bf1bdc0851a94911 : SUCCESS in 43m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/305eeac1700d47be85471a77146aa611 : SUCCESS in 38m 02s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/656b61f9e2514bbe9fff820ffc9206e5 : SUCCESS in 46m 58s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/525920f92c064c84b88361234142c702 : SUCCESS in 37m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a57b2bee552447aba27edffc562e967d : SUCCESS in 1h 00m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b87db5c31f484cfcb0de8fa20460df0c : SUCCESS in 43m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/31b4720163a545869c51c20401ed2b96 : SUCCESS in 42m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efa676fd7e944ae9a9f63b8005aeb6fc : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e608697de84d129605833341ce03e0 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d149dcdb5e72450b920a2dd1f2a737de : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f17b7c5423f94ae4840346cc2a847fdb : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/398cbcbc097c4970abe3635eefaa593e : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/290f13113fa6423e8d151d326e200d30 : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":4},{"id":"a3ad444760fa6e55efa3aa15b200adb20f72c6f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 18:13:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2463246e4547f49f65b1b5d4d9da8ee00e3eae60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 18:14:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6745a0de141646c7baa728041f2ce6ea : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ac52daa60e1496bbdacf62513b5fb790351deee1","revisions":{"d2847e4a3fd4aab79ccc911d59399611b61b1ccd":{"kind":"REWORK","_number":1,"created":"2019-09-25 20:55:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/684843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/684843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/684843/1"}}},"commit":{"parents":[{"commit":"63f216755a61f76452d900264948d55ce94b75a4","subject":"Merge \"Adjusted default heartbeat interval and renamed file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63f216755a61f76452d900264948d55ce94b75a4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-25 20:52:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-25 20:52:57.000000000","tz":-420},"subject":"Update devstack to use workers","message":"Update devstack to use workers\n\nUse API_WORKERS for all services, to make sure we\nare testing with at least 2 workers for every service.\n\n* Use API_WORKERS for all services.\n* Removed unecessary path for disabling the worker service.\n* Use enable_host_header is tests.\n\nChange-Id: Iba9ba655c47ee085a47ee98872437f07fdc3346c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2847e4a3fd4aab79ccc911d59399611b61b1ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2847e4a3fd4aab79ccc911d59399611b61b1ccd"}]},"branch":"refs/heads/master"},"ae435421bf28f69813d880708503e33623693172":{"kind":"REWORK","_number":2,"created":"2019-09-25 21:08:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/684843/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/684843/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/684843/2"}}},"commit":{"parents":[{"commit":"63f216755a61f76452d900264948d55ce94b75a4","subject":"Merge \"Adjusted default heartbeat interval and renamed file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63f216755a61f76452d900264948d55ce94b75a4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-25 20:52:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-25 21:08:17.000000000","tz":-420},"subject":"Update devstack to use workers","message":"Update devstack to use workers\n\nUse API_WORKERS for all services, to make sure we\nare testing with at least 2 workers for every service.\n\n* Use API_WORKERS for all services.\n* Removed deprecated and removed settings for the worker.\n* Use enable_host_header is tests.\n\nChange-Id: Iba9ba655c47ee085a47ee98872437f07fdc3346c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae435421bf28f69813d880708503e33623693172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae435421bf28f69813d880708503e33623693172"}]},"branch":"refs/heads/master"},"f91e5e11493be73267dfd523fea4ba55e619aa79":{"kind":"REWORK","_number":3,"created":"2019-09-25 21:56:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/684843/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/684843/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/684843/3"}}},"commit":{"parents":[{"commit":"63f216755a61f76452d900264948d55ce94b75a4","subject":"Merge \"Adjusted default heartbeat interval and renamed file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63f216755a61f76452d900264948d55ce94b75a4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-25 20:52:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-25 21:56:26.000000000","tz":-420},"subject":"Update devstack to use workers","message":"Update devstack to use workers\n\nUse API_WORKERS for all services, to make sure we\nare testing with at least 2 workers for every service.\n\n* Use API_WORKERS for all services.\n* Removed deprecated and removed settings for the worker.\n* Use enable_host_header is tests.\n\nChange-Id: Iba9ba655c47ee085a47ee98872437f07fdc3346c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f91e5e11493be73267dfd523fea4ba55e619aa79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f91e5e11493be73267dfd523fea4ba55e619aa79"}]},"branch":"refs/heads/master"},"ac52daa60e1496bbdacf62513b5fb790351deee1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-28 03:29:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/684843/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/684843/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/684843/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/684843/4"}}},"commit":{"parents":[{"commit":"d078b0abf375a93fd56580c380746171d1ec1d39","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d078b0abf375a93fd56580c380746171d1ec1d39"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-25 20:52:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-09-28 03:29:40.000000000","tz":-420},"subject":"Update devstack to use workers","message":"Update devstack to use workers\n\nUse API_WORKERS for all services, to make sure we\nare testing with at least 2 workers for every service.\n\n* Use API_WORKERS for all services.\n* Removed deprecated and removed settings for the worker.\n* Use enable_host_header is tests.\n\nChange-Id: Iba9ba655c47ee085a47ee98872437f07fdc3346c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac52daa60e1496bbdacf62513b5fb790351deee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac52daa60e1496bbdacf62513b5fb790351deee1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
