)]}'
{"id":"openstack%2Fdesignate~686023","triplet_id":"openstack%2Fdesignate~master~If27b9473a84da412baf1da83b583be6c19376db7","project":"openstack/designate","branch":"master","topic":"685887","hashtags":[],"change_id":"If27b9473a84da412baf1da83b583be6c19376db7","subject":"WIP: Change --dry_run to --dry-run","status":"ABANDONED","created":"2019-10-01 18:20:37.000000000","updated":"2019-10-11 09:34:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"57d3d4b5ab1e57e82e4fab0af015ffc0b19a5a79","_number":686023,"virtual_id_number":686023,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-01 19:21:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 19:21:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c28594211925e909112ad131064fd75f342db290","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-10-01 18:20:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce27bdb7578ca829e701a8e840e35fe7cd72c25b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 19:21:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4d5fa781d7b64fe3a55da36a50fe9599 : SUCCESS in 48m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0df2b44ca350446f9a70f06f1604b42b : SUCCESS in 34m 27s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/3d283965fb084374a8991e3a4d93c6e4 : SUCCESS in 42m 05s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/fe2b8a07c84849e0b13dbd07262c2cb3 : SUCCESS in 36m 43s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/4299647f85e54a4b86f198ae50f44590 : FAILURE in 28m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eb8c5621ad1645d8ad5be5f6103c41e3 : SUCCESS in 55m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5cae7c9a94c44dfeac54334a1b31c346 : SUCCESS in 34m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ab880cc1ee944441bf1d23a81522da7f : SUCCESS in 47m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d5a09e6e7524431b069d6c8d1c592aa : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/198371bce84f40dbb991567298aced25 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3086552b04574d5fb87eecba47cc4212 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aad41d2e6fd243c39bf0df1d09e47491 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0380e1cdc82e403b8bda91346d8bf50b : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d876574fd94f44929113f66855b54724 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87ff3fccccaf49c499e9c312a21ed347 : SUCCESS in 8m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/19ea8e7331f4455e8d816cb40a5d2855 : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e81ee6cac20e5bf403687d2294944b5a6fabd38b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-10-11 09:34:14.000000000","message":"Abandoned\n\nIntegrated into https://review.opendev.org/#/c/685887","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ccb5dded294ca40811948941e99da0f488b9dfca","revisions":{"ccb5dded294ca40811948941e99da0f488b9dfca":{"kind":"REWORK","_number":1,"created":"2019-10-01 18:20:37.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/23/686023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/686023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/686023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/686023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/686023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/686023/1"}}},"commit":{"parents":[{"commit":"8cadd7437bc1f696d7a124816296aeddb0186a0f","subject":"designate-manage pool update --dry-run not should pass argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cadd7437bc1f696d7a124816296aeddb0186a0f"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-10-01 18:19:55.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-10-01 18:19:55.000000000","tz":0},"subject":"WIP: Change --dry_run to --dry-run","message":"WIP: Change --dry_run to --dry-run\n\nChange-Id: If27b9473a84da412baf1da83b583be6c19376db7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ccb5dded294ca40811948941e99da0f488b9dfca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ccb5dded294ca40811948941e99da0f488b9dfca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
