)]}'
{"id":"openstack%2Fdesignate~687819","triplet_id":"openstack%2Fdesignate~stable%2Fstein~Ia3e07c42dc548207aa7ff9efcc35a215c9660e54","project":"openstack/designate","branch":"stable/stein","hashtags":[],"change_id":"Ia3e07c42dc548207aa7ff9efcc35a215c9660e54","subject":"fix get the name servers for a zone not validate zone_id","status":"MERGED","created":"2019-10-10 08:15:52.000000000","updated":"2019-12-16 17:09:52.000000000","submitted":"2019-12-16 17:07:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687819-1576516053956-91269b4c","meta_rev_id":"6dfe3a6c5a1f4342402699dcad7076c47f80c0bb","_number":687819,"virtual_id_number":687819,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-16 17:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-12-16 15:55:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2019-10-11 06:46:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-11 06:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2019-12-16 15:55:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-11 06:45:16.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-10-11 06:46:50.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-12-16 15:55:48.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-12-16 17:07:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95cfe424041e7d001af968dbf9738684f2a1d33a","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-10-10 08:15:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1918b9d9710fcf2d3112c88633e5accb4f2c3ed1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 09:18:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3355701bbdc64462b14292ed7417347f : SUCCESS in 40m 14s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/223bfa7e77e14976b18ca39e6bbc7866 : SUCCESS in 38m 19s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/5cf1d63ec21b43cca95cbf84fa4cea00 : SUCCESS in 38m 45s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/22c7093cdcea4b5282c6b3fbfeade785 : SUCCESS in 35m 36s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/6affb6653b954f98961d4ff7f332a661 : SUCCESS in 35m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/02d21d18ee2c4b318b232db0949c32e9 : SUCCESS in 56m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78ab9b52593441ca9fc542dfa8c1affd : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4aff482e542422cb6e8c75898f889aa : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/525f35df24954ac497b9139c3229a7c6 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f417b43307254e1e87c3b6ae4f0a2640 : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ada1655b383a4fd58c3f5ac90467b2a5 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13a0f0d09fde4343abbfec26247ca935 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea57abb5f2c949d48e383739cda98217 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d15ad04abee405f81e7ffba4c3a1a3b : SUCCESS in 3m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a5de09d799584759a0bcc7d90eb769c5 : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":1},{"id":"806ae15b743b8442d5116e60d13fa94ce8c87045","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-11 06:45:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2be5e21707f13a93b09743e8e2dccdf6470e8f45","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-10-11 06:46:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0807c2cd43ef3e24ff9fde7682e4bc1460242f55","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-12-16 15:55:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a0ca25f403e78c2e0805649bc2ac22e8ad4578e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 15:56:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"989aa9760492bf32ced5d7472bf34725500c0901","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 17:07:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cbd428cd7c7eb5e56be359e55028ab5ef3c6d756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 17:07:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/998435faef2841bdbce0ad99d8bda104 : SUCCESS in 43m 29s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/163d6b7f6d1943b89af6928fad76118f : SUCCESS in 48m 29s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/bfb3dcf70f1c4409aec748b5a46b44df : SUCCESS in 41m 27s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/20a6d46ca86e45138317e1267c3a7e98 : SUCCESS in 39m 18s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/0a582ff90e644fd69481cc001763ccb6 : SUCCESS in 39m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/67f0a97510e04e3ca2e27d224a5f52d6 : SUCCESS in 1h 03m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37414641f4854b0399b0dc29df9a1e18 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c04df98c272044b1a441087e5bc4c423 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9687b7476944f938f254b423d0f73ef : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c62c4b5f422b4e25ad1212edb6e09351 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3181e7a8a7e143e39b9753d45a888a7c : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a5ee3d3db0a42739fa7a924218f73bb : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9380fd8881b47c0a55d5b375d766e3e : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6dfe3a6c5a1f4342402699dcad7076c47f80c0bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 17:09:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8adcfb16c91b4af1b1116143ac5d4339 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"357a41c4281181cd0cffa04b377cdfc3020eaf1f","revisions":{"357a41c4281181cd0cffa04b377cdfc3020eaf1f":{"kind":"REWORK","_number":1,"created":"2019-10-10 08:15:52.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/19/687819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/19/687819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/687819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/687819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/19/687819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/19/687819/1"}}},"commit":{"parents":[{"commit":"3382867b38cf3f72ec494794bceaa5523b43d9c4","subject":"Fixing 404 link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3382867b38cf3f72ec494794bceaa5523b43d9c4"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-09-20 06:46:36.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2019-10-10 08:15:52.000000000","tz":0},"subject":"fix get the name servers for a zone not validate zone_id","message":"fix get the name servers for a zone not validate zone_id\n\nnow the api of get /v2/zones/{zone_id}/nameservers not validate zone_id,\nif set zone_id contain spaces, will trigger DBError and return 500.\n\nChange-Id: Ia3e07c42dc548207aa7ff9efcc35a215c9660e54\n(cherry picked from commit 0ba52bfb62cb64992a06b64c44f3e6a9dcadc193)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/357a41c4281181cd0cffa04b377cdfc3020eaf1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/357a41c4281181cd0cffa04b377cdfc3020eaf1f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
