)]}'
{"id":"openstack%2Fdesignate~688706","triplet_id":"openstack%2Fdesignate~master~I7ebc7991bd4f59579195d337f1787fd8abefbe90","project":"openstack/designate","branch":"master","topic":"fix-dashboard","hashtags":[],"change_id":"I7ebc7991bd4f59579195d337f1787fd8abefbe90","subject":"Don\u0027t install v1 dashboard panel","status":"MERGED","created":"2019-10-15 13:10:25.000000000","updated":"2019-10-15 17:22:20.000000000","submitted":"2019-10-15 17:20:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688706-1571160027801-14372de2","meta_rev_id":"572d4337580eeb289270a65210488f8381ae1bd2","_number":688706,"virtual_id_number":688706,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-15 17:20:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2019-10-15 16:14:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2019-10-15 16:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2019-10-15 16:15:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 16:14:27.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2019-10-15 16:15:09.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-10-15 17:20:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b3e146900ad138b8c6abd8f197b241adc411b40","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-10-15 13:10:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c6f937e7fd0cf8a80cd1868d18f7cada51847ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 14:19:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b81f334b8c1d46c39ae36b887f9ece10 : SUCCESS in 46m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ab205225df5b49348ea9caaad7ca558b : SUCCESS in 39m 25s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/716ea8683048494abab7639f631e296b : SUCCESS in 38m 01s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/00b08c86d76347778de1290fc351d5f4 : SUCCESS in 53m 46s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/64c363b05a20431395fa132c6927cc87 : FAILURE in 31m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7e3c336bce08472891f2278e9774f23c : SUCCESS in 1h 02m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1a25a8bb1cb8409f889d7290c740af45 : SUCCESS in 42m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a200dc96838d4243acfe2067cfc3dc05 : SUCCESS in 49m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c531a74f49e401db6ef320d19a131d9 : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e11c4af9b50f4a80baa93eb9a0964a3b : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae9cd4a344a64fcaa2704af5575d7eb1 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e7bc305ad4f4729850e9eb88a451688 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4ba06ef87cf43f9aebe47568aebc1c8 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30e4cb15d5b04426b33a7ffcfd47605c : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7ae016034540cf8a743e9d16c22b18 : SUCCESS in 9m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b5b546b494ef456db5e782733730a71b : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":1},{"id":"7a2d6f2d71d911c2ccd67e889e13eecca5635ce6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-10-15 16:14:27.000000000","message":"Patch Set 1: Code-Review+1\n\nGood catch!","accounts_in_message":[],"_revision_number":1},{"id":"8631211db1c03a8b1c7ff87a4a3528b410972996","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-10-15 16:15:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlooks good - non code change","accounts_in_message":[],"_revision_number":1},{"id":"9918e5f1c188d58557b096c699aac762d808e263","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:15:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df21f15fc049eaaeaa1c800eba50be575290f5fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:20:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f84386916974e778ea0d1943fa5f6239ae261d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:20:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fa2d52facf964aef8c6d4ca2f93b050a : SUCCESS in 39m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0902c91fdf274a619c470d8ec4907d1a : SUCCESS in 52m 46s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/f98d0bdd3a2f487d9caa4bbd58b3b520 : SUCCESS in 48m 30s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/13e9381406d241eda58a5244221b9597 : SUCCESS in 39m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0112a9784a27498899819dde45bf1042 : SUCCESS in 1h 01m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f285b6e7121e4e04b0b1cab86004274b : SUCCESS in 39m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9568ff44a62447afb6a2359d30c6ecbe : SUCCESS in 43m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/045104380c744c3ebb171d162bb607b4 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abbe940a4f045e6b5072f2244963c9b : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5aaa1f2337747118606544bafb9be83 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4150aae4f1f14d7c87cb67eb194cc973 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cad56faecc0d44b4bfaf9398c3509f9c : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1e8dc94c3742c183a4154138d0e530 : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":1},{"id":"35204fb829312408bd166183cea8062409cf1e7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:22:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07464b4fe0943af8367fbbea88b0964 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9e6549ad30b2b06ef6ca409a3a89e1c68ad6a47","revisions":{"b9e6549ad30b2b06ef6ca409a3a89e1c68ad6a47":{"kind":"REWORK","_number":1,"created":"2019-10-15 13:10:25.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/06/688706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/688706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/688706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/688706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/688706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/688706/1"}}},"commit":{"parents":[{"commit":"dd2bb0486e57df80abc1ac6ad55ddaeed1398953","subject":"Merge \"fix ListObjectMixin._obj_from_primitive bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd2bb0486e57df80abc1ac6ad55ddaeed1398953"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-10-15 13:08:16.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-10-15 13:10:11.000000000","tz":0},"subject":"Don\u0027t install v1 dashboard panel","message":"Don\u0027t install v1 dashboard panel\n\nThe panel _1720_project_dns_panel.py has been removed in [0], stop\ntrying to install it in our devstack plugin.\n\n[0] Iabf7d396ea6feb0cd7de1c5408a81a7a3ce1efbf\n\nChange-Id: I7ebc7991bd4f59579195d337f1787fd8abefbe90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9e6549ad30b2b06ef6ca409a3a89e1c68ad6a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9e6549ad30b2b06ef6ca409a3a89e1c68ad6a47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
